)]}'
{"id":"openstack%2Fglance~923259","triplet_id":"openstack%2Fglance~stable%2F2024.1~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2059809","hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"MERGED","created":"2024-07-02 14:09:48.000000000","updated":"2024-07-04 13:42:36.000000000","submitted":"2024-07-04 13:41:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923259-bug/2059809","meta_rev_id":"5240eda9737a63c124b4036ec21e97f79deafb36","_number":923259,"virtual_id_number":923259,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 13:41:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 09:49:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 09:49:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:15:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"faba6e77d49255493364d46c053ae8d02ba724cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d1fe9b983027db5ef02e514cdd3d07c1439adf2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:15:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7481c59f14db4f9a8c6ffa8f35aeee4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/692fd004dfd94c30b904ef118340d437 : FAILURE in 3m 37s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/37c7c6e4a66f4598821f3a45d52b7e1f : FAILURE in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8965d629b82d47716c86edeaca88b6a358554cdf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:20:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e304ec145d4a4323abae729ba8544839\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec138424a1ed4bf995b8cbd2c7b7f6c2 : SUCCESS in 56m 30s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/77428d6f86304172b855caaf3adf57d9 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/61943eb0c683408aa66f7faf41542a10 : SUCCESS in 1h 52m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/21337f3372a84153a9abd6e657939206 : SUCCESS in 55m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5334e18cc524e8995ef96abe785fa0c : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a50b0a592a1a49ac9c676c90d116b31a : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/93080deb531249d7aaef733fef91eabc : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdfb4a1150764396a2a1f64f39f2aab3 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/daf30560370a422682b88629d199f153 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ead154cb1ef4e75b7b5a2fae3e6abfd : SUCCESS in 9m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/772eb70faaf54e8ebf3cfc5d26f531a6 : SUCCESS in 7m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0aa0a4aa3d784e95ba2e2c79a26a9f08 : SUCCESS in 7m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e66c102db8274cce87cd2130b894dbb4 : TIMED_OUT in 2h 04m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6afa62db9a0246a6ae901279daa086a7 : SUCCESS in 1h 52m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2870dfc157ea42dd86218d7907459a8d : SUCCESS in 1h 50m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/42c69b9fa2ca416da3d921e9e104565d : SUCCESS in 1h 47m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/379c10a93a494259a3806f674bcc4b54 : SUCCESS in 2h 13m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7ffe5ea1646443b590fe4165e33835d7 : SUCCESS in 56m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3665073e6c364106956d6c7d3e2e07df : SUCCESS in 2h 04m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/084fa455feb143d1ac3b94ae50d94cd2 : SUCCESS in 1h 59m 18s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f55b4cc3ecfd43fca0ae470b2b12c299 : SUCCESS in 1h 06m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bbdaaaab12ce47e7bf59add89509f021 : SUCCESS in 34m 32s","accounts_in_message":[],"_revision_number":1},{"id":"9a59a5f7ccd7fa6c05ada7958d600fb09fae2e8b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 09:49:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"59a3e5736cb8a281d89dd0264a09091418a38926","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 09:50:17.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ea7ba4b222b49927e8b55e99bf56b9693edd9e2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:41:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9cd7102968d142a897778bdbf6f3edd5\n\n- grenade https://zuul.opendev.org/t/openstack/build/7e6003f54d3348bfbc915e15b7350683 : SUCCESS in 1h 06m 51s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/dc773ee129394d0283e08d5d8b9b4e61 : SUCCESS in 57m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/939745b721f24180aac5a77c7cb8fbdd : SUCCESS in 1h 20m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bef90427777943979c774bf681d2828c : SUCCESS in 53m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccc7ea171e6b4ba08ff8cf9efce52cb7 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/643f552f7fad4be29ee12b41fafdfae1 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c717fa31aab472aad543dd50b04a053 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c47972232294f139d6b0bb6f0961e31 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65ea60b2328f4a4f851a3c677af7eae3 : SUCCESS in 8m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/86d7a4414ef64aeebd0d4992cf6a16e0 : SUCCESS in 9m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b45827235bc44e7845d7a7e7c480f67 : SUCCESS in 1h 35m 11s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bdba228d4b60440ba862fafacc297164 : SUCCESS in 1h 45m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03dd0dda5fdb43e2ad3f2c5d19c8bc7d : SUCCESS in 2h 02m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12491200580a4e97a299b8842c3a7ecc : SUCCESS in 1h 00m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6fcb76b55db04acfb72b9313517bd45c : SUCCESS in 1h 49m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/49554e1139354fe3b22129b1d5838db1 : SUCCESS in 38m 35s","accounts_in_message":[],"_revision_number":1},{"id":"d8a34a6d051ceb1ea0ce49a944b5c893b01b04ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:41:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5240eda9737a63c124b4036ec21e97f79deafb36","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:42:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2fa48bf0885543f0bab8cac93c4c68ce\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9dddd40f233c40a683b9f45d5b6983a8 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3","revisions":{"35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/59/923259/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/923259/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/923259/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/923259/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/923259/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/923259/1"}}},"commit":{"parents":[{"commit":"eb3eb4ea269f608a7b534bb205f917309420cd51","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3eb4ea269f608a7b534bb205f917309420cd51"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:27:23.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
