)]}'
{"id":"openstack%2Fglance~923260","triplet_id":"openstack%2Fglance~stable%2F2024.1~Iaf86b525397d41bd116999cabe0954a0a7efac65","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 20:43:13.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iaf86b525397d41bd116999cabe0954a0a7efac65","subject":"Extend format_inspector for QCOW safety","status":"MERGED","created":"2024-07-02 14:09:48.000000000","updated":"2024-07-04 20:45:39.000000000","submitted":"2024-07-04 20:43:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923260-bug/2059809","meta_rev_id":"46a53b09e194bc1dc107541481ac3d0e726f0be1","_number":923260,"virtual_id_number":923260,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 20:43:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 09:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 09:49:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:19:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:24:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 15:56:31.000000000","updated_by":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2024-07-04 16:00:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-04 20:43:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b598c2009255108dd4729c20e31506613f634948","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4913ab19a7d6807ada1487978b194a9006f2c43d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:19:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e7a9c0f94f4dbcb1dbb7dce7ef0a0d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f720575dae2e4219a0c81e186e0c3702 : FAILURE in 3m 48s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/067f25901e534fbd83b63ca8506736f0 : FAILURE in 3m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"362a3853df6c972f217933d7ece0a0253f7a62cf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:24:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dae2af47e4674db98763bd51752b1f0f\n\n- grenade https://zuul.opendev.org/t/openstack/build/edce1587961a45989a47bd78bf28d536 : SUCCESS in 59m 25s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e99be4e5e0814610abb97bb0c2cf71b8 : SUCCESS in 1h 13m 52s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/32f6c000ddd0420c8da0081a94cf5d8f : SUCCESS in 1h 38m 47s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/80ed6689a2a048a7b526fb0105b5d3ab : SUCCESS in 1h 00m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec617174f6614f6f8b5ed528ded126d0 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aea9f48dd370470f86adfdce44f94bf7 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/962d3b3ac47f46adab6228abe1506ec7 : SUCCESS in 5m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2770914773e74d2baf2e76cfc4383ab4 : SUCCESS in 5m 43s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d8586bb94e9b40ca987b36f92a73eb95 : RETRY_LIMIT in 2m 13s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/095b689c732a422b90761bc951f59aab : SUCCESS in 9m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6076aa4a9024ebababae58b2250d581 : SUCCESS in 11m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a74b5d5c885b43938cd1b1aec91418f9 : SUCCESS in 9m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2786b9825f8848ca969e7a1851b905b4 : TIMED_OUT in 2h 04m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b213f024c9334e01af30b862fed36dc3 : SUCCESS in 1h 52m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/feab0f0ec25449fc9f71d0a6e8ca71b8 : TIMED_OUT in 2h 04m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/48d22605e7cd4353b7adbf14db62c4a2 : SUCCESS in 1h 55m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0de707903dc4a6bbcd363bb04db7a25 : SUCCESS in 2h 05m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/978a3343df40442e98104a7b2f2336f2 : SUCCESS in 1h 08m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4daf3ecdc10b42218ca40a1b498d8982 : SUCCESS in 1h 57m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/20d08aa8b3504b7d8c8dee64ae013978 : TIMED_OUT in 2h 03m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bc7939af3fa44f6185eda8a01072898b : SUCCESS in 1h 04m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7bb738e715f84ec1ba1ff991b6e1a63f : SUCCESS in 40m 48s","accounts_in_message":[],"_revision_number":1},{"id":"61e52c481944ab586c35d60483ba31164e69cdbb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 09:49:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"776af0fbd3cb52e6555e4d3aa6943669ca6b1dc4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:00:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/df213e3481b7409b9bd284d32c8a5756\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c392a6fc132d434e9d51d3a124502b39 : SUCCESS in 8m 41s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4ac5ad524f474c4e852e6a8ed1962677 : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eaf8e05f60d8b142e3b61c357f3d08c97266e1ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:10:32.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eeb5f92d21924128927c02bd4d521b62\n\n- grenade https://zuul.opendev.org/t/openstack/build/30601da083604802b6faba370b03459b : SUCCESS in 1h 04m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/27f480833bca4d85a1135e77b6c9a6d7 : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ad7b1e7314c94220ae8879646d2f77a7 : SUCCESS in 1h 19m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/72fcc310ded345f98b3b4defbcd3afdf : SUCCESS in 1h 05m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d34f08f180b4136bc51558ed13c905b : SUCCESS in 6m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b4309cc24904cf286974ccea36cd9d9 : SUCCESS in 7m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/11e3cfad634a4048a022e9a83579c396 : SUCCESS in 7m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d13784c19d4ca69a33ab28f0e61284 : SUCCESS in 10m 03s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/eff1bf3557c345e3885cae63d2c74a69 : RETRY_LIMIT in 1m 53s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ec73c007cb543ed848382bd4062c8dd : SUCCESS in 9m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/250d7f128d8849dca6993d9e5b8c77fc : POST_FAILURE in 8m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c8bc090ac90846e08344caabdc89b4c2 : SUCCESS in 8m 43s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9dbf3a26c40742c7921370f2ac5cefdf : SUCCESS in 1h 48m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72551343a0cb4e6f9fafd53c9edfd885 : TIMED_OUT in 2h 04m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f4539ba00b1b403ebcc8f911cdd40177 : SUCCESS in 2h 01m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/57630af24c034244867f22b001117e70 : SUCCESS in 1h 55m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dc51dce860a24debb7a5b20194f1da83 : SUCCESS in 2h 08m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b9685b990544a37b478662981a72fb4 : SUCCESS in 59m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8954f47c9cf44cb8bf4348ea177115a7 : SUCCESS in 2h 09m 44s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/09eec38c1df9480ca2426515dfc65edb : SUCCESS in 1h 55m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5e101192392f4f609322653a6948193a : SUCCESS in 1h 05m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da3b350aefcf4b128911374ef1b982df : SUCCESS in 36m 47s","accounts_in_message":[],"_revision_number":1},{"id":"75c692e0af2dab9e5b1e21a2ca84156ac4331105","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:43:03.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9f08b96bf3e4ee0b0e1f7c62f267192\n\n- grenade https://zuul.opendev.org/t/openstack/build/669b31b7eef44c329f33833ab846df80 : SUCCESS in 1h 09m 29s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d39cd909229249a799eb75d80507c3fa : SUCCESS in 1h 03m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99aa3c4cf1b848ea85df028c6faa99d3 : SUCCESS in 1h 17m 44s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fa8c886c55aa4d9bb3b403636b3f7341 : SUCCESS in 1h 04m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f70b0d56e594d8097c5bd12f2a3c1c8 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fc68ca86aeba4817a73706c084192579 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2217747df2c04e3e9d22d662c8c14d2a : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35409c1e8fc340b7a3bdfbfda4c64c69 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0a7d7b2fa2854277b1eb83818a863fc1 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1efc1d05517c4c8ca5b2f67493b13681 : SUCCESS in 8m 25s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8818d05898a542d3956707f66169d921 : SUCCESS in 1h 45m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/adeb15e278694d13a31566ddd25cd5a7 : SUCCESS in 1h 52m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f9048a4a66374ef6867098a9e9e425a5 : FAILURE in 1h 36m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e4d173a4d878422892c1b47a2f8825d1 : SUCCESS in 59m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/48ffa64b6ef4467f98ea41fcfa2cf564 : SUCCESS in 2h 04m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/de1a536c7ad94d389e01dd827546ecf0 : SUCCESS in 35m 56s","accounts_in_message":[],"_revision_number":1},{"id":"143525a03fb0af75e39f9b9b98ed5ef9285cc508","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32622,"name":"frickler.admin","username":"frickler.admin"},"date":"2024-07-04 15:56:31.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_22348\u003e with the following votes:\n\n* Verified-2 by \u003cGERRIT_ACCOUNT_22348\u003e\n","accounts_in_message":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"_revision_number":1},{"id":"8fe19099d78c38fdbd50844806b0516b96b81aa1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:00:54.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"621c5356df529b0d1edf4f0f284e45776c6a5877","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 20:43:13.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f6a4644ed82c476fb5ab22556dde2f4d\n\n- grenade https://zuul.opendev.org/t/openstack/build/acd07173047b493bb25e240bded70421 : SUCCESS in 1h 06m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/011c7129c14a406f9a49613c5eaa45a2 : SUCCESS in 1h 02m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/87ec9a6dec154145a42761b8ec33cb2f : FAILURE in 1h 14m 11s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/539acac920d54b79b2411d5407479ac5 : SUCCESS in 1h 04m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/156a978772744d77a4b4b4b5e73a1136 : SUCCESS in 4m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/27cc3c29a6aa4a58ac250728cd19f572 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/48104846d10643319444807c964f7b4e : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/159fd905fc014cc0b4c31f0c6c990c16 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f594ca4ab6f4d17a6b968180ef71979 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ac604c21cf7c438296a9de3addf839f8 : SUCCESS in 8m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de6523d65f354395ac690b77c31282ec : SUCCESS in 1h 44m 21s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/becffd2f9c6a48fa8910ae7f09b6d828 : SUCCESS in 1h 55m 26s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/edafa4fdb66e4d058d3d228c8555053c : SUCCESS in 1h 48m 39s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/11fcf8b303fe45ee8c2dd3233124a04c : SUCCESS in 1h 00m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c70e5187ae514cdf85b755e4f59e2021 : SUCCESS in 2h 00m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e365851515d64882946512cd15cdaf61 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":1},{"id":"bca83048adf8147c75d6de1a07b13af8b3445c3d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 20:43:13.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"46a53b09e194bc1dc107541481ac3d0e726f0be1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 20:45:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa8c6515a39740728f1f40fa48dca079\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7b0b5b7a3b1147758777e606e5cd60ab : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7a6c344c8b9d60c89e324e11c2edac8e2ad4a229","revisions":{"7a6c344c8b9d60c89e324e11c2edac8e2ad4a229":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/60/923260/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/60/923260/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/923260/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/923260/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/60/923260/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/60/923260/1"}}},"commit":{"parents":[{"commit":"35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35c7e85d40ae00b29ae13cb5b9dc2743baf34bf3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:27:23.000000000","tz":0},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6c344c8b9d60c89e324e11c2edac8e2ad4a229"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a6c344c8b9d60c89e324e11c2edac8e2ad4a229"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
