)]}'
{"id":"openstack%2Fglance~923263","triplet_id":"openstack%2Fglance~stable%2F2024.1~Ieea7b7eb0f380571bd4937cded920776e05f7ec4","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 16:05:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ieea7b7eb0f380571bd4937cded920776e05f7ec4","subject":"Add QED format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:09:48.000000000","updated":"2024-07-04 21:20:15.000000000","submitted":"2024-07-04 21:19:10.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923263-bug/2059809","meta_rev_id":"8f16eb027fe4b2cb0fcc93344de31b2f4398f5d1","_number":923263,"virtual_id_number":923263,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 21:19:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 16:05:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 16:05:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:35:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:42:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 15:58:17.000000000","updated_by":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"reviewer":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"state":"CC"}],"messages":[{"id":"ba3036550c60a36cba981fe7d4db998d7444c2f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fe35cfc421674f692ef8b16ec7b2e35e2fc6cb98","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:35:59.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dea7531e0d9743bea8d32617de312a4b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ebe10777ad85426ebd21383adab8c424 : SUCCESS in 12m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b573874035594082a15570271c9c931c : FAILURE in 3m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ea23672f6bb0f22125d68eef140a79435b30148","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:42:28.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed8d96c7847d43af8dcf2f042540a6c6\n\n- grenade https://zuul.opendev.org/t/openstack/build/9ffb37728e3e422c8bc06b0bf460e0fa : SUCCESS in 58m 44s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f83052de56ad4c6a965d932608c899d5 : SUCCESS in 1h 10m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/528494f7943941fd8be09218e07eaff0 : SUCCESS in 1h 49m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b7e5089b758b44f68da16b33b2591f4b : SUCCESS in 1h 03m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02e874f72a6c4861b7731cc73ab4e688 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ff2d82d864148598ae11f4ab08d5192 : SUCCESS in 6m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9512aa8d32ab48ed9558a3fba026cc99 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f554c835da024da99266deb87d65214a : SUCCESS in 5m 11s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/690e8c88ef77487786a331215bfc18d2 : RETRY_LIMIT in 1m 59s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4430fc089d0d4fd997fdfa149c8c66f4 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa89b5e0802043b3ada7258c3260cd92 : SUCCESS in 7m 35s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca9bb1c17048473683734368f6d03a31 : SUCCESS in 8m 09s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cce55e228c9d4be5ad24e6a91f51230f : SUCCESS in 1h 58m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7b3efa5dc36d498e85c0c1a19c1e13f4 : SUCCESS in 1h 34m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e4242f1853f048769c22fb056422b7a8 : FAILURE in 1h 39m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73f06cfb27d24425aebd5aa992518676 : FAILURE in 1h 46m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fedf18e98cf94ad795946fb32a62b845 : FAILURE in 1h 55m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a2c8f92e2e0b495a92ef7e443151e122 : SUCCESS in 1h 00m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3bb6f2acba9d45919711728afc869258 : FAILURE in 2h 04m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b97b4781f1f7497299d13a2c88d9d79e : FAILURE in 1h 42m 58s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/16f97a8b5f564b9b91bf329be7f90250 : SUCCESS in 1h 07m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e50fdcf4c09d40e2a688f3850df1f66c : SUCCESS in 33m 53s","accounts_in_message":[],"_revision_number":1},{"id":"2e288d0554372627ad9b0b5ce26c7d8e5bc57500","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-04 15:58:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5cfcb9b0f6bda858f9b2dd8081910f9b374e14bd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 16:05:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"37f2864f598f6af6a042cac04c570cfc2ed72339","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:10:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0acb16053c4243d49770e7fd04ffb886\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/ee24c05f3e9742668acee029a5a02a8f : SUCCESS in 6m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6ae1ba3f129a40fda7da08f683277043 : FAILURE in 2m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"40644b77d27fbe31353fe2584f8f5858ccbd0d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:59:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1773f2110bb9cfb68e6049b2708cd780c120b94a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:15:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8819441b2cc940ef92a82a41f1452f9c\n\n- grenade https://zuul.opendev.org/t/openstack/build/230c30eadd1b4db9b50b24eb34eecc41 : SUCCESS in 1h 11m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/663891f86aaf43da9e773085d83ac996 : SUCCESS in 1h 04m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d937c87e95db4af1b95f50f8b9d61a56 : SUCCESS in 1h 31m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f10c7b13b3fb4c5b99d72de984782980 : SUCCESS in 1h 02m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa5b06caa1ce4268a679240cd629755f : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/88266dadd9364efa91db3b07a6bc100c : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d81c6040a0c48ac8dacf62b3ea13c22 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/921d6d94dbc74522ada30a00e2c68e95 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e694278f8f7340a881cd8571cdbad409 : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3e67b5103a9844aeba7555c54b9caee0 : SUCCESS in 8m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/41e9fb20247f4b628761dc935d790edc : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b2f57fbd61ff4d529afe1bda29228357 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/37b711e32a5c40e7b1fd1597c3ecf1cc : TIMED_OUT in 2h 04m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c0732882fde14531879856a6c285c675 : SUCCESS in 1h 43m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c6b830cc43df4146a74909bdac79ab9c : SUCCESS in 2h 01m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5d64b2425a444dfb9cdbda70f20ddb58 : SUCCESS in 1h 48m 51s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29bef72410994fcfbb0088e9100cc58f : SUCCESS in 2h 11m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b478b4e51e944c77a0d656b07d40d400 : SUCCESS in 58m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/572987a845ae47e295079b76535fc07c : SUCCESS in 2h 08m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b81e9b2482f84c989b0da78443bdac57 : SUCCESS in 1h 48m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/32ca8e5dda6d49608bca2c2e53fa8afb : SUCCESS in 1h 07m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99990a1a79f1455e8a9554c7f33778a4 : SUCCESS in 38m 06s","accounts_in_message":[],"_revision_number":1},{"id":"1d9244240447e48aa77276222556d4d4c8a445a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:19:10.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/26798a4839d745e88094a891d9ce5e78\n\n- grenade https://zuul.opendev.org/t/openstack/build/ae5ce58a809c4bed92690d3363fa7d5c : SUCCESS in 1h 04m 27s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/567ac93006864f958f3cb0cdf328f29e : SUCCESS in 1h 02m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7939549bd5ff49ff90e340d3d0384b2c : SUCCESS in 1h 36m 23s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8177e828f4684cc5a13f442e14013fad : SUCCESS in 1h 00m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de63a81cdad34892bf0a7dd4916c2795 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ef70ddd7c7a44e6698bf1ce2202270ba : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c547ee39a66243cfb1c557e6c9713a0d : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2526533d8b5a4751b6e6acdb1bbd76e0 : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cc72e1a0e56b4adc8ac5018af69fa04a : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f3fa680b308a4e64b5d3954606299a09 : SUCCESS in 8m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0401e16fd65b437b8d2e8badf469d4fd : SUCCESS in 2h 30m 02s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cd9f6cf1bb1040d18498f88de660e8c4 : SUCCESS in 2h 11m 29s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/57ac45f3b5394da9b337d42010a711f3 : SUCCESS in 1h 59m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5a22466aac764db1ad08690d081e9a62 : SUCCESS in 1h 00m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dd08f40509bc406d99de3afea7863296 : SUCCESS in 1h 51m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/63dfae0a9e67430cb2ede879478664bb : SUCCESS in 36m 34s","accounts_in_message":[],"_revision_number":1},{"id":"7cd97a88008d55bc55d9219c10dee05c43a1423b","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:19:10.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f16eb027fe4b2cb0fcc93344de31b2f4398f5d1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:20:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98b1d1402bc541dead41ff466e03fbf7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/822aca4c053d4223851bb4eaf12c0b73 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83acfc462066866cc94f1b6fa7f559e0849ac7dc","revisions":{"83acfc462066866cc94f1b6fa7f559e0849ac7dc":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/63/923263/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/63/923263/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/923263/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/923263/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/63/923263/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/63/923263/1"}}},"commit":{"parents":[{"commit":"f482a92f5acb53938466b47302449f6871a754a0","subject":"Reject unsafe qcow and vmdk files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f482a92f5acb53938466b47302449f6871a754a0"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-27 16:33:55.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:27:23.000000000","tz":0},"subject":"Add QED format detection to format_inspector","message":"Add QED format detection to format_inspector\n\nThis merely recognizes this format and always marks it as unsafe\nbecause no service supports it. This prevents someone from uploading\none that we will ask qemu-img to inspect.\n\nChange-Id: Ieea7b7eb0f380571bd4937cded920776e05f7ec4\n(cherry picked from commit 4096c5aff1d046d5c28d0e4a69b3c9574e9e8cc8)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83acfc462066866cc94f1b6fa7f559e0849ac7dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83acfc462066866cc94f1b6fa7f559e0849ac7dc"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
