)]}'
{"id":"openstack%2Fglance~923264","triplet_id":"openstack%2Fglance~stable%2F2024.1~If0a4251465507be035ffaf9d855299611637cfa9","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 00:14:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If0a4251465507be035ffaf9d855299611637cfa9","subject":"Add file format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:09:48.000000000","updated":"2024-07-05 00:15:47.000000000","submitted":"2024-07-05 00:14:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923264-bug/2059809","meta_rev_id":"7a188bc65e4f788a20317d8bbb84483164405b23","_number":923264,"virtual_id_number":923264,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 00:14:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 16:06:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 16:06:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:40:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 18:50:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 22:01:37.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REMOVED"},{"updated":"2024-07-04 22:03:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-05 00:14:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0c9f85d7e45e23595f5714a9ec7f3593c08e0c21","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b64bb0b2a14d7ba122119944530b995b1d01fad7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:40:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bb3bda5eec2749af861218ee8857503e\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/d685748df4a44350ad353eb57b0fdb9e : SUCCESS in 17m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6944ecac6f654918a4a71b7441543480 : FAILURE in 4m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"61e7313999290692dc7c91d08fd0be2bcac9480c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 18:50:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/61e9f930b0044bd998290a74b968b63a\n\n- grenade https://zuul.opendev.org/t/openstack/build/33be0812bfeb4cfa9a0d14dce9a2402b : SUCCESS in 1h 12m 08s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ce7a2f148ff247a38807406434cd9e10 : SUCCESS in 1h 03m 27s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1027bbebba3a4fbd8017fd8a9fb5efb1 : SUCCESS in 1h 32m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/06de43772c2144b4857a860e6531215d : SUCCESS in 1h 03m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af25f54085fc42ffa292153ec852a55d : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db5bef926d03413eab065f00232db20c : SUCCESS in 5m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a9b3c0a031794af9901b554e66982d78 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b22b25d4d8144bdba991bc2e9cfd312 : SUCCESS in 5m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b8d757265cc34238b70adb52dd78f2eb : RETRY_LIMIT in 1m 52s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/14d7ee2789a64bc6a2a5c9c155ce444b : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/828c60fa7736455094320559eed31149 : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6b8d51eeb1f648b3b9b3111bf129362c : SUCCESS in 7m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/093b851528474eab8e8c26b671ed9f4d : SUCCESS in 2h 01m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/816946f60968498181e79b3aa4c7040b : SUCCESS in 2h 01m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0f2bcb18f19445f2ac3619f103acb5fc : FAILURE in 1h 54m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1b758a1fe2744963b3d77d87e6a42384 : FAILURE in 1h 41m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/abe8c0d2f24e4782a4d93616f5773e17 : FAILURE in 1h 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea7ad1af5a374e2ba2f1dbb9df558656 : SUCCESS in 58m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5204be7bb72b45a3a0c601183b50fdb6 : FAILURE in 2h 01m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e81cfc4f0c4046f9b9e6bb103e36f961 : FAILURE in 1h 53m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4e742bac520a4c7eb92feb409f556a95 : SUCCESS in 1h 13m 17s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9f05a9e0f73c48bfa0431b560d26ae2a : SUCCESS in 31m 33s","accounts_in_message":[],"_revision_number":1},{"id":"e9ff67db92c81f5383e048ed0dd3817035442fae","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 16:06:19.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"20c4f97750c8565c6555cedb1be79ed979ddd56e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 16:59:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e058e7403d9a9b52b04062a5ccfe85d96586e00e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 18:15:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e0e3d1d0bc28428aa0b6055ceaac54a1\n\n- grenade https://zuul.opendev.org/t/openstack/build/ef84879a706b44b29cca9fb9d2e824ef : SUCCESS in 1h 03m 54s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d3ea9313dec34d2a9a08bff78259a869 : SUCCESS in 1h 03m 00s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6aa3cba9d91a4a2c8b35a486c3fcf1ca : SUCCESS in 1h 53m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2bdbeb61156d48fea2cb0dc91f90700c : SUCCESS in 53m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cdc3c37d3d54683859af1ccccf64873 : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18a7d7d7bc654c39ba3e51809c1a469a : SUCCESS in 5m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a8f1471a4b9146e5a98d37c0a9a6179f : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/072d9de85d404943b7d6b8fd754c7548 : SUCCESS in 6m 16s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/950464edc05a46eeae0af90dbf011291 : RETRY_LIMIT in 1m 58s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/03860b0be4304496ad49ada4130a9552 : SUCCESS in 10m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da3e587164b646039978700fd0e168b1 : SUCCESS in 9m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fbfa8a59c7b8470e8b8e1e7b55298146 : SUCCESS in 9m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f261987a057c4005a22b5f504ce63804 : SUCCESS in 1h 59m 36s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3c6417b9060f43d18913e03a45860282 : SUCCESS in 1h 50m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b14638bc819b469c8421f9b991cc2cca : SUCCESS in 1h 46m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/aa8c9561d8184882b003390728052a47 : SUCCESS in 1h 59m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8be10189a500459dbcc7aa341e617d48 : SUCCESS in 2h 03m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02411bacc2e34427abcdd2c8315d2247 : SUCCESS in 1h 10m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cebf6d6960ec4f4cbb4cb5c3bb2a1f8e : SUCCESS in 2h 05m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/427e230ac6844b17857e0f157e2f7247 : SUCCESS in 2h 02m 14s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/922ffc15850042628bc164aaacff2b31 : SUCCESS in 1h 05m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b0d3b118c9e4beaab0601eb96691d89 : SUCCESS in 48m 08s","accounts_in_message":[],"_revision_number":1},{"id":"ba0cd2e12d5eb2a6b7a46616544a4248bafe214c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 21:19:12.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/668a0ec1657448e39ef3dd0d3a05f2c9\n\n- grenade https://zuul.opendev.org/t/openstack/build/fa0e1df8630b433380e6e05c3661d4cb : SUCCESS in 49m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0a4af6d151df4bd1bf9893458dc98c6e : SUCCESS in 1h 01m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e2f204f81da844259bb7924dae702ccc : SUCCESS in 1h 53m 58s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5dd271c514d94ee3b84c4efc499da673 : SUCCESS in 1h 05m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68c58525f9d746e0b3c44f7a98312565 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9bd4a427cd54eb0ae3988aef1f6e97f : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c3a7c30eebe241e5b44c637c2fee5800 : SUCCESS in 6m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ab08247c4f3451aab08ade33eeada02 : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/228994b63ddc4d8ea9923b8b60eeea97 : SUCCESS in 8m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4b1a4e6c8ab6470e86e4fef866842896 : SUCCESS in 8m 29s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/32ee9e7499e0421ba033d6b990854ed7 : SUCCESS in 1h 40m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/830810bd44a34973ab95212fe8d07e31 : SUCCESS in 1h 40m 24s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4a695d70251b499e927089d2295f5cb8 : SUCCESS in 2h 11m 47s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/da11bd397c0e4995bdd549043b598a89 : SUCCESS in 54m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1ecfba0c6bf84a5eb755b630e57e2420 : SUCCESS in 2h 07m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2ff362fcfa41489097078f63422fcd81 : FAILURE in 29m 48s","accounts_in_message":[],"_revision_number":1},{"id":"630ed0defeea1787a72eb0a4f370d3abe09599f5","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-07-04 22:01:37.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_22348\u003e with the following votes:\n\n* Verified-2 by \u003cGERRIT_ACCOUNT_22348\u003e\n","accounts_in_message":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"_revision_number":1},{"id":"ede2810011e316065d12548a0dcb3f4159d40779","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 22:03:09.000000000","message":"Patch Set 1:\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6b5e45353b0fda92958d86b2c8f5564df407285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 00:14:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/30fd4405c9014d32a6e1fd20540032a2\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c54088fe10b4b6ba4f364a79477b2c2 : SUCCESS in 52m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c407aa8d800744aea9f136f2d1f9b013 : SUCCESS in 59m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/53944699359642c394ec4717f6fdb639 : SUCCESS in 1h 27m 08s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/272fe134892c47e0994c5843d08844d8 : SUCCESS in 1h 08m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb60164053e4d68a65da612b7b401d4 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f371e3a4d0bc415491f21878459eb74a : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c0aa0ce6fadf4b449ff6cfaceb4ed9ce : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ab170efdeb147ddb2635f88e05e734d : SUCCESS in 6m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b73649a7e60e44d387b1a9db90ce45bb : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f275bd5db0d4d44835ac6a99145fc6f : SUCCESS in 7m 46s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/144de5d514f64739b813be280d6be376 : SUCCESS in 2h 11m 30s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3f13bf64006412298f09e8a9926b814 : SUCCESS in 1h 47m 02s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7facd6ed2e0a40ad9e75a09f493c051f : SUCCESS in 1h 43m 15s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/49296c4d82884ccaa49f4ced6f6a22b2 : SUCCESS in 46m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9260ecfa3f324416b4336be60e1e8507 : SUCCESS in 1h 25m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/688ab52e7f7b48a6ad1b3add4fd71a95 : SUCCESS in 24m 42s","accounts_in_message":[],"_revision_number":1},{"id":"5150232cb792a16598e9bb5d6d436dcee3521458","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 00:14:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7a188bc65e4f788a20317d8bbb84483164405b23","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 00:15:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0f95a8ef9d7948f4a45c330286bf8c1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70741d7c2ebb4cac8bede22b6dd11ff8 : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"26e5b8551a1be966c560763436a13e2ecda63b69","revisions":{"26e5b8551a1be966c560763436a13e2ecda63b69":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/64/923264/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/64/923264/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/923264/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/923264/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/64/923264/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/64/923264/1"}}},"commit":{"parents":[{"commit":"83acfc462066866cc94f1b6fa7f559e0849ac7dc","subject":"Add QED format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/83acfc462066866cc94f1b6fa7f559e0849ac7dc"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-18 15:25:24.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:27:23.000000000","tz":0},"subject":"Add file format detection to format_inspector","message":"Add file format detection to format_inspector\n\nChange-Id: If0a4251465507be035ffaf9d855299611637cfa9\n(cherry picked from commit 79271eaa5c742a1741321198c43807857fb6ed94)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26e5b8551a1be966c560763436a13e2ecda63b69"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26e5b8551a1be966c560763436a13e2ecda63b69"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
