)]}'
{"id":"openstack%2Fglance~923265","triplet_id":"openstack%2Fglance~stable%2F2024.1~I447e7e51315472f8fa6013d4c4852f54c1e0c43d","project":"openstack/glance","branch":"stable/2024.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 15:52:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I447e7e51315472f8fa6013d4c4852f54c1e0c43d","subject":"Add safety check and detection support to FI tool","status":"MERGED","created":"2024-07-02 14:09:48.000000000","updated":"2024-07-05 15:53:11.000000000","submitted":"2024-07-05 15:52:26.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923265-bug/2059809","meta_rev_id":"72b62de100b1b258cb04a8f71d07b51e3066930a","_number":923265,"virtual_id_number":923265,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-07-05 15:52:26.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-02 18:16:48.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":2,"date":"2024-07-05 15:48:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":1,"date":"2024-07-05 15:52:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:09:48.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:43:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 18:16:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 15:52:26.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"f747dc04b80f1d7de6fbb8fef5cb50be8359fd24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:09:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c89343a0bec989f700dd03592c95b8311e53254d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:43:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b3c8651c782e46a898fc4f6bc7d7b289\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fb52fece91cb4100ae3601d3d163ed57 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0602c984f7cd41358a63d9dda985a24c : SUCCESS in 11m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a023b7c420116e4502454b4a7e72e9e3f016412e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 18:16:48.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70c9656f0d0146e9af233f1006205ba5\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/b88596d478954d17a8902e4ea3b4202c : SUCCESS in 58m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab9e4a63807b47c299a6c13061390af3 : SUCCESS in 56m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f308bf2adb24781943f67cbc460f392 : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2407e7a60d8a4a3c95e3d81928147c2b : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/270b1030da9c41a3bf56e5cf4df9f938 : SUCCESS in 7m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76ae8610d6214998bff2b667778da567 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/77f250d43f8f487dbc33c320811623c6 : RETRY_LIMIT in 2m 06s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2aa44f19107f4b1281ec9e57adf6a49e : SUCCESS in 8m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6bf2ca6558b943429fc50b399f3ba240 : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/16c7a41fe7714953abd1b980440d5c90 : SUCCESS in 8m 07s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c0aaa2feaf9d45eebfd498163a887f1d : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"27f4b1876c3a2a59dd25fefdfb89ad1f3658c718","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 09:54:00.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f870958092a1862d0d178bcfdc6090e26b2c5fbf","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 16:48:55.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"be018673b6a21abaac5d2dd9a9337290738c3547","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 15:48:53.000000000","message":"Patch Set 1: -Workflow Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ecb282710095fc87d62467575e1c8f0464e159fb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 15:52:12.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"1c32c8eec21f7bb0d1310d208a7cbef817ada066","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-07-05 15:52:26.000000000","message":"Patch Set 1: Verified+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"59713b5eb99894968cd18f5feccfbee9721e5983","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-07-05 15:52:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"72b62de100b1b258cb04a8f71d07b51e3066930a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 15:53:11.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7057f581004446d286aa57112564eb1d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df70c87e60e546f0b945a185464465c3 : FAILURE in 25s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1e5c4444548e674c7718682959a59cf331b44947","revisions":{"1e5c4444548e674c7718682959a59cf331b44947":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:09:48.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/65/923265/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/65/923265/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/923265/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/923265/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/65/923265/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/65/923265/1"}}},"commit":{"parents":[{"commit":"26e5b8551a1be966c560763436a13e2ecda63b69","subject":"Add file format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/26e5b8551a1be966c560763436a13e2ecda63b69"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-26 15:41:02.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:27:23.000000000","tz":0},"subject":"Add safety check and detection support to FI tool","message":"Add safety check and detection support to FI tool\n\nThis adds a safety check and detection mechanism to the\ntools/test_format_inspector.py utility for verifying those features\noutside of glance.\n\nChange-Id: I447e7e51315472f8fa6013d4c4852f54c1e0c43d\n(cherry picked from commit b27040b87e43ff4acfb6870ef0d13d54e5ca5caa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e5c4444548e674c7718682959a59cf331b44947"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e5c4444548e674c7718682959a59cf331b44947"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
