)]}'
{"id":"openstack%2Fglance~923266","triplet_id":"openstack%2Fglance~stable%2F2023.2~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"stable/2023.2","topic":"bug/2059809","hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"MERGED","created":"2024-07-02 14:10:25.000000000","updated":"2024-07-05 07:46:07.000000000","submitted":"2024-07-05 07:44:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923266-bug/2059809","meta_rev_id":"a10b162bbc11f1f1137a7639b314ef56972f61ed","_number":923266,"virtual_id_number":923266,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 07:44:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 03:57:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 03:57:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:27:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:05:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"47529ec6d19e32a2c2b1aaf17e42581fc318bc0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ae0f5e43245621f8cccd522897987a3a8e85fc1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:27:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbda94f964244e40a40e2c3c73d68c0a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e14318bca45e433abcfdc2ef456e9986 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e680bd8de45b4845866c2ac269bd734c : FAILURE in 4m 03s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/60d8b0afd29f41b9b4a7e551563d6da9 : FAILURE in 4m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a643252083e5448287571cf919a78272 : FAILURE in 3m 12s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"41562bf7264b886d4260c711fe2978a707f406e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:05:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f56f7b83e0d84ad0ac32ffd5cd40bf3f\n\n- grenade https://zuul.opendev.org/t/openstack/build/b0d6177cfad84aa5adcdd318ac90d1cf : SUCCESS in 55m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4156b38e26a84e9d92c298b9d0c9dd8e : SUCCESS in 1h 16m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/105caed33dd949cb86db596b2e60af24 : SUCCESS in 1h 05m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2b8c8f13ad5f4d06a0a4ad76f6024607 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2245fe2fc0b948ab81273c1eba98539f : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ffebc0083d214eb6ba4ecb53760cb31f : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b29b26ec53b04707aca577741c562580 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/28ac1341435a4ea9a16626de3f436dbb : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d76a1cb0bd304fd19d14c4acca931989 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/400829c46684497cac11673918ce63e2 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/673fd38df37a4960ab66148c8b1df4f9 : SUCCESS in 8m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6b323a1ba6934168a379fcb0fbf00e22 : SUCCESS in 7m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e5d3a1b849044d4f85d2c90f0a3781c9 : SUCCESS in 1h 53m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba5e64209e3244a8a2da5f23ccc23f4e : SUCCESS in 1h 32m 29s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c39855efd9b4213989f650d0fed17ad : SUCCESS in 1h 53m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67e006d347c641e8a3c1721d15b5703b : SUCCESS in 1h 33m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/33c70b8f6e1d431ea6d27c3c59d3a819 : SUCCESS in 1h 42m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a79bd2793a13441abdbd0effe9cfae1b : SUCCESS in 52m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4a08ec41e3014827833d5c2872fa0a2d : SUCCESS in 1h 58m 14s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/065317b8cd164ed6bb6021dff670b4ba : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"f63818b057e67d598a5deda9c45ed876dc304e53","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 03:57:15.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4137134fa2d14ab1b3b4b07ccf9812a4c4a7041","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 03:58:05.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eca2e73e94b775cbf4d44ce85d1642233242161e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:44:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/810b91fad4d3436f9b436bafa7eca899\n\n- grenade https://zuul.opendev.org/t/openstack/build/4de3027f274547b6a4549400cc00dee1 : SUCCESS in 1h 19m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2e7b1078e13647bcae7652a7154f518c : SUCCESS in 1h 22m 35s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0923ce8347d64821b0b6aca14574685a : SUCCESS in 58m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b82cb3c6ba3b41f08b1d284548ba7b15 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1e2deda92884424814c1f77b017b5ba : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07433b3eb42b4136bb273040cbbf66b5 : SUCCESS in 7m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a876d0916aba4cd698ffa8f0bf2be951 : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d60b6d51202c4d709d65cfa4a74e5727 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e95fca9ca1e84037b477430d2ef4eec3 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83f38f1a432344fa9fb21be9b2bf8b2b : SUCCESS in 7m 37s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/937c4cca4dbf4f13bcead72b6a6c0e0a : SUCCESS in 1h 48m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f3b2e9d9320f464da1d54f464b916dfc : SUCCESS in 1h 47m 13s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/45de421d62aa494c95b1f4bd7e48b9cf : SUCCESS in 1h 52m 33s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6dbb54d05981482c9b1a69cf6987cdea : SUCCESS in 46m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/240060151fa7497187604d0b2c514d54 : SUCCESS in 1h 52m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9f80be2f2654095b3ec0267b34f4f66 : SUCCESS in 47m 51s","accounts_in_message":[],"_revision_number":1},{"id":"6c0adefa435f5e6057fc67c0efcbd0e3453e8a4f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:44:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a10b162bbc11f1f1137a7639b314ef56972f61ed","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:46:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bffe821774c44135a0277309fb8949dc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad0824c1b82646eea290b28b2a23d38d : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"87a98f0d25a3b3948d3d49ffad354934589442a3","revisions":{"87a98f0d25a3b3948d3d49ffad354934589442a3":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/66/923266/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/923266/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/923266/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/923266/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/923266/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/923266/1"}}},"commit":{"parents":[{"commit":"5e2acc91a4fcc834e3faea3ddb2d0ea13e2af606","subject":"Support Stream Optimized VMDKs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5e2acc91a4fcc834e3faea3ddb2d0ea13e2af606"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:29:26.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87a98f0d25a3b3948d3d49ffad354934589442a3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87a98f0d25a3b3948d3d49ffad354934589442a3"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
