)]}'
{"id":"openstack%2Fglance~923267","triplet_id":"openstack%2Fglance~stable%2F2023.2~Iaf86b525397d41bd116999cabe0954a0a7efac65","project":"openstack/glance","branch":"stable/2023.2","topic":"bug/2059809","hashtags":[],"change_id":"Iaf86b525397d41bd116999cabe0954a0a7efac65","subject":"Extend format_inspector for QCOW safety","status":"MERGED","created":"2024-07-02 14:10:25.000000000","updated":"2024-07-05 07:59:55.000000000","submitted":"2024-07-05 07:58:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923267-bug/2059809","meta_rev_id":"3bdeddb0c2ac12ff0c05bfb486df0d687bc4ff43","_number":923267,"virtual_id_number":923267,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 07:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 03:57:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 03:57:31.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:28:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"05f9ba6f1ae41b8da31df9d68d6f1d9d5d00d9e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c0ed12083c9329c2b25623d5bbaf3cb6c02dc19c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:28:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d03987494eb044e6a26354586b52615f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/12c820699f6442dc83d01c81d2e7f19f : FAILURE in 3m 40s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0190338d3bc74e6dbcc6fa06f6a918c1 : FAILURE in 3m 35s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/9510a2defd064117b136352aa8b93080 : FAILURE in 3m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/71b72cd1273044729e7800a4c1f8a637 : FAILURE in 3m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1fe3dfab9e577b1fd1a308607742ec2840ce2054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:28:53.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33b8919e1e90488bb71914fbce32956e\n\n- grenade https://zuul.opendev.org/t/openstack/build/ee782facd8a542e1b2bf4baed64fee7c : SUCCESS in 1h 08m 51s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a184816f3e9e4356a45a244ff88cd66b : SUCCESS in 1h 58m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8556777b48ce45ab845c48656e6be019 : SUCCESS in 55m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b9cf8b1d0874a5daafeb975dbf03afd : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bcf4b4c2decd445d82cee1f974067f4f : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d559d0f83e549c5ab0b9b954afb56a6 : SUCCESS in 5m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45d1fdbc52074e3ea2c1d199730ec0b5 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b59030f013f44742be79d1915b1885d5 : SUCCESS in 5m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e76994251b2441db28cc8855c4e1ccf : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/06074f6c652746a6b9bb6bbed1a5755c : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7deffa2dcbf34112bc919d8c7e9452f9 : SUCCESS in 9m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/40c18c51bca249cfacdf9251c07140e4 : SUCCESS in 9m 02s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/10570c9778944e85b0ee89d9e6da2395 : SUCCESS in 1h 53m 23s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/781f6fe77fd44ee881254f6d2da19758 : SUCCESS in 1h 51m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ef363930be62455d914add1e1d893b69 : SUCCESS in 1h 45m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d8c45719a6f344c6ab5c62b103ba02f1 : SUCCESS in 1h 31m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2434585f06b24d699bf8cc18eda2c2ac : SUCCESS in 1h 41m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d7a41393b424d6a9f0901b8ebef0917 : SUCCESS in 51m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/20ded0bfd1f74f1a9677d52d5b68afbe : SUCCESS in 1h 53m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c78739a7171041e2bfe9d2cd950882f7 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"0d710ffb4de098e5a2e887b0550b6ace1af07270","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 03:57:31.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67f774a84d536e0566af01bfad3eb306a820917e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 03:58:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ecb8019f7f9ed983fa7dfca33461ebde0eec8234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:58:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46a2b1f4a58249d9b59c9108b07514f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/a0c2fae6a0a8402d96427fd061ea64fd : SUCCESS in 57m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cbd69910e32a4526a93a2126d07aca98 : SUCCESS in 1h 28m 01s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1c42fa364cc444c49f6e4ffdb914c8d0 : SUCCESS in 1h 04m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/696a4fabbb6a4e88a6b06bcba33ca7cf : SUCCESS in 4m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7ee994bd6bb14182bb08b1002cbd48c3 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/658e366f16fa4ad68ee4a6370b5db4d1 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e43c37865754b379f1ae9a6cda75ebd : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/de816cd774994f62827f70b22148960a : SUCCESS in 5m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ccf290e5d7bc4f53a0c3e329c85fd287 : SUCCESS in 9m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/def6e724900b4dae82e3be297c03ace0 : SUCCESS in 7m 30s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f77f2b670f3b4586831b15a99a1bcdf3 : SUCCESS in 1h 48m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db24f68a212f46af98d314c9ba488ace : SUCCESS in 1h 32m 09s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ffdf0e9c7891443987c99bbaf3ee009e : SUCCESS in 2h 03m 54s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/378e14a3fc5941278a141af67cc52a69 : SUCCESS in 59m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/301978072a7a4df987fe5d291bfb5394 : SUCCESS in 1h 51m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/62114d8779b0454090a50cf3ea270725 : SUCCESS in 26m 09s","accounts_in_message":[],"_revision_number":1},{"id":"1f25f330fe1f8201806bf4af8038066e6e0a46e2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:58:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3bdeddb0c2ac12ff0c05bfb486df0d687bc4ff43","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 07:59:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/325b78da579849d6aa979770532e464a\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b2d4a421a5a494bb2b568e119ae1e1e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"733d771a73bffaf1501c7a1d91829646daf1372b","revisions":{"733d771a73bffaf1501c7a1d91829646daf1372b":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/67/923267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/67/923267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/923267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/923267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/67/923267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/67/923267/1"}}},"commit":{"parents":[{"commit":"87a98f0d25a3b3948d3d49ffad354934589442a3","subject":"Reject qcow files with data-file attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/87a98f0d25a3b3948d3d49ffad354934589442a3"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 17:29:10.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:29:26.000000000","tz":0},"subject":"Extend format_inspector for QCOW safety","message":"Extend format_inspector for QCOW safety\n\nThis adds two properties to the QcowInspector that makes it able to\nindicate whether the file specifies a backing_file or data_file in the\nheader. Both conditions are considered unsafe for our usage. To\nease checking of this condition, a classmethod is added that takes\na local filename and digests just enough of the file to assert that\nboth conditions are false.\n\nChange-Id: Iaf86b525397d41bd116999cabe0954a0a7efac65\nRelated-Bug: #2059809\n(cherry picked from commit ae536bb394793c9a7a219cb498e03d5c81dbbbb7)\n(cherry picked from commit 2eba54e0821106097dfeceb424e53943fd090483)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733d771a73bffaf1501c7a1d91829646daf1372b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/733d771a73bffaf1501c7a1d91829646daf1372b"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
