)]}'
{"id":"openstack%2Fglance~923270","triplet_id":"openstack%2Fglance~stable%2F2023.2~Ieea7b7eb0f380571bd4937cded920776e05f7ec4","project":"openstack/glance","branch":"stable/2023.2","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 03:58:06.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ieea7b7eb0f380571bd4937cded920776e05f7ec4","subject":"Add QED format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:10:25.000000000","updated":"2024-07-05 08:26:39.000000000","submitted":"2024-07-05 08:26:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923270-bug/2059809","meta_rev_id":"7bf76052e507034b4f8a7420f0cc39428de42822","_number":923270,"virtual_id_number":923270,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 08:26:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 03:58:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 03:58:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:25.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:32:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:06:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6a2887ec5f4a844f760c06c4049f08deea9ad6d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8e08ba74f3baf88682292da8666db7e0a5559141","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:32:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a698ceb1fe94788aa0a12883e93f244\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9088e2bf2a3d46f68b5dec097fe17fce : FAILURE in 3m 51s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec8212c695124f25abf30c73ef1e2316 : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/bd521dec56954e298ca31a1334a6632a : FAILURE in 3m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7fc59943ddf14c36bb2b81ca856c3888 : FAILURE in 3m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cdb547ece26e110497c6909c66bdaea84c0b6b4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:06:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/26479ea1a8ce4cc7a305f15f033b4261\n\n- grenade https://zuul.opendev.org/t/openstack/build/56fbedd08563437ca70960a4cf15a5f7 : FAILURE in 52m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b11f7d3e4eb94680ab941ac32f381f2d : SUCCESS in 1h 26m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c3311458f044651b4f585d417c018de : SUCCESS in 54m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9cf5020d1d494b4a950144655fac39c8 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbb98476da734c7380c0e371d4f86dd6 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a1e1d9784c64e6f8dbf89a0bbf70569 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/89d9723fa760470dbcce144ab0f9a328 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/36ab717f206943a1893dc0fce5b8733d : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea6d777e32824a6f8080078421eece27 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a0177dcbeb254eba99c57a08044f70a8 : SUCCESS in 8m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36c1aa668a1b45aa884188d51e8e1a8a : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4fa454482ea743c6a64a9238221518d1 : SUCCESS in 8m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/793b597b7a40431080e76448aa82cafa : SUCCESS in 2h 01m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/29f56796b25a41d5b3bbdcd029b6e10c : SUCCESS in 1h 52m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0efcf2a4e8148349be99430f5d34bf2 : FAILURE in 1h 40m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/62654f49319d483781de68dfdb6f0713 : FAILURE in 1h 28m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3e4e260726b74405867a37dbecbb6aac : FAILURE in 1h 46m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9ba705fb095c4d1592391892336ff095 : SUCCESS in 1h 09m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1daad416a0db469db321ca187dd0f0bf : FAILURE in 2h 03m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8454829d76974413b93c70935bcf81d9 : SUCCESS in 36m 31s","accounts_in_message":[],"_revision_number":1},{"id":"46d261fbccef9aedd615568df3ef88040153cb63","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 03:58:06.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f652cb91a9419a9edd0322552a2d2d8a64abd3c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:05:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ebe9be6b9691456cad9d71e5ee15538a\n\n- grenade https://zuul.opendev.org/t/openstack/build/45a53956379b453c9b6188fa368666db : SUCCESS in 55m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3765d99a2b9c4763a597f93d2dd191b0 : SUCCESS in 1h 34m 27s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8a43dbce85fa4edb92fac562f0f6f483 : SUCCESS in 54m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbf6d1856ad4573ae2dc0d211dbf5b5 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cee2ae3097dd4cbc83e66f933b5bb29a : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/66c541e485194e99a30c6ad811153c35 : SUCCESS in 5m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/81628723615a48d6956ca02eb22b21d3 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e71ce2eb94c74d03aca2e53b98abac59 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6d926004852405292883983e6de4e3a : SUCCESS in 7m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ccbfa5aed0154a148a0d200d125d0d12 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e03010b51a74c769f7c82d5109a5f2a : SUCCESS in 9m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bfc67f46daa34388b3824281bebecbe9 : SUCCESS in 10m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cab8763e12c341f5873c50d0f6df5945 : SUCCESS in 2h 00m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0f2a0b299a604fe7a9c8dbad17fffd03 : SUCCESS in 1h 46m 56s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7d2963ac6e564b3b9645a4b2bfaf0912 : SUCCESS in 1h 43m 20s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4e6b215bac2c4acfbfc884dc23c54f84 : SUCCESS in 1h 14m 05s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2b9c5284b3b34f5e9cd18a2a333e22c2 : SUCCESS in 2h 00m 58s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6627e6eacf25438c87596b664fe155d9 : SUCCESS in 57m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/08c9243eaf284d24afe89e318aed8c79 : SUCCESS in 1h 54m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0534bd6c8aaa42e5a00b949fd1f727e4 : SUCCESS in 28m 53s","accounts_in_message":[],"_revision_number":1},{"id":"21309a18807d299da04939318555b60e92816193","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:17:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"8ed9943029f36f5a23c7f4e86828a1585a9e1219","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:26:38.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/583ae6e71ea449d0be5c7b517cd47194\n\n- grenade https://zuul.opendev.org/t/openstack/build/4a67146a47964ba4b13c958eefb0d2fb : SUCCESS in 1h 03m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/521d1fb258494f948d75f4cde5ef5d57 : SUCCESS in 1h 27m 06s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/760cf120f8684c0cb38bf25a8e09100c : SUCCESS in 1h 05m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a69491c9d45e490f912c4b431624f832 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6f5a9fbf24164e16aa6feb0e6cbfba55 : SUCCESS in 6m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e752fc82f5834cceb585f89d10655342 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d807e08082f3475e96e949e8e2864eaf : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f500a9b3b27f42d99c3827716a5a020b : SUCCESS in 5m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7067af25ff94d17b1a955bc45887c14 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6feed43d1347417f9e0bd304186d0460 : SUCCESS in 10m 04s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/08f83f6b2d8c40df9313c38f8bc235cc : SUCCESS in 1h 56m 18s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4f7dd2ec305047748e64bc643d7961d6 : SUCCESS in 1h 39m 00s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1a36aceec9964417a4aaf209688a1d2f : SUCCESS in 1h 37m 36s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93c9fea0404e487e971a9d88ea122a2e : SUCCESS in 1h 10m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2360cab74797462c822733caf52e16a9 : SUCCESS in 1h 58m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f168f5a3d4514afeb0d06566fc028531 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":1},{"id":"7bf76052e507034b4f8a7420f0cc39428de42822","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:26:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4bc13f954522d81199d9fcbfd4dccdbc0d9d7ef3","revisions":{"4bc13f954522d81199d9fcbfd4dccdbc0d9d7ef3":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:25.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/70/923270/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/70/923270/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/923270/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/923270/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/70/923270/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/70/923270/1"}}},"commit":{"parents":[{"commit":"c23730304a107927a5677c8f1887c84d241b8afd","subject":"Reject unsafe qcow and vmdk files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c23730304a107927a5677c8f1887c84d241b8afd"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-27 16:33:55.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:29:26.000000000","tz":0},"subject":"Add QED format detection to format_inspector","message":"Add QED format detection to format_inspector\n\nThis merely recognizes this format and always marks it as unsafe\nbecause no service supports it. This prevents someone from uploading\none that we will ask qemu-img to inspect.\n\nChange-Id: Ieea7b7eb0f380571bd4937cded920776e05f7ec4\n(cherry picked from commit 4096c5aff1d046d5c28d0e4a69b3c9574e9e8cc8)\n(cherry picked from commit ba98022b98ef5b9c98e1d7d20c88e1ca4b23fa80)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bc13f954522d81199d9fcbfd4dccdbc0d9d7ef3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4bc13f954522d81199d9fcbfd4dccdbc0d9d7ef3"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
