)]}'
{"id":"openstack%2Fglance~923277","triplet_id":"openstack%2Fglance~stable%2F2023.1~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"stable/2023.1","topic":"bug/2059809","hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"MERGED","created":"2024-07-02 14:10:45.000000000","updated":"2024-07-05 10:22:08.000000000","submitted":"2024-07-05 10:20:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923277-bug/2059809","meta_rev_id":"b08b8f19e6bd774d54c41726b3775a789d696c30","_number":923277,"virtual_id_number":923277,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 10:20:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 08:32:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 08:32:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 19:13:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"3bc91b195473f6d71e3055ed65835b243495a370","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9e8c91c55a1c780b2e7f8881724f02c6ae6b0239","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:51:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b74a0390bde9417cba2f8ab59dd391cd\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/72cac5de2ac84ee38daecd23edf933c7 : SUCCESS in 9m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b08b10fa812ccf6f57df5c510f23f3e72ac0deb5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:13:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d7a6df771f44996877ca4941414bb42\n\n- grenade https://zuul.opendev.org/t/openstack/build/960141016dd849df934403ecc04660ec : SUCCESS in 52m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/18428d80e55a42bd94787137b64e7cf7 : SUCCESS in 55m 09s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6b22f8b947b42c882faf586d71c018a : SUCCESS in 1h 24m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a4451da9f98344eaa7b33f935439f282 : SUCCESS in 1h 00m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91757dde8eb9433496109c0d06237b24 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9394e0d7a2084b9fb7e647fbabe5ec43 : SUCCESS in 6m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65031e386a8a4d189a9c784593fe0b70 : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/20894cafbf57448492854db127076479 : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c66c7fbaf99448eeb822c5c39cfb2018 : SUCCESS in 7m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f922078c96f2489da22adcb179834484 : SUCCESS in 1h 41m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/839284165d0c4f5faed4310cb1b7cf8f : SUCCESS in 1h 07m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/605137426a3646a584f14c6b2bbf98ce : SUCCESS in 1h 23m 52s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/25cf26f9733148b990fce73f2d188ddc : SUCCESS in 1h 41m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d41ec09405774d1da1551ad7bb4b49b3 : SUCCESS in 1h 33m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22902a980e2d40498688cda5e28d05dd : SUCCESS in 50m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/709f1b171e2a4ae18cd70a791db0af36 : SUCCESS in 1h 52m 02s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/75756a10df7a4cffb62495731f5a7f7a : SUCCESS in 37m 08s","accounts_in_message":[],"_revision_number":1},{"id":"8a35fda4e9d949e69ae11957efa1a8df1e1ea90e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 08:32:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0840d5e60e9e1c0d45eb3f93df3d9d5717c8eeae","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:33:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cd10ca5d4da5b585aa75380d0c3229bb494f90a2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:20:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d6fb54b4ef0243debd4fbeae936fefdc\n\n- grenade https://zuul.opendev.org/t/openstack/build/96204a03c1b74ef090808a06ef283f83 : SUCCESS in 1h 01m 47s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ae5b43259dfb4f50a81f743a617b38df : SUCCESS in 1h 13m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cf5d077027be4424884dd6b3ed415c49 : SUCCESS in 1h 15m 58s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3836a10e98b2447f81318eb92cb3bbc9 : SUCCESS in 55m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a82eb96736643829e026cd24b340d26 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8fccc3591584448199f1f78413e29148 : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26dc33681c3845519ae4a0e7d282d0a5 : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4b9c52fad2714060b46783f331057857 : SUCCESS in 9m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ef1430cb87d2468c8de54625416864ab : SUCCESS in 8m 11s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dfc6895416484676a1b93e8f9593e10a : FAILURE in 1h 01m 12s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d4f4c10ac8d478297c3e321f20ffd5d : SUCCESS in 1h 39m 23s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e17aaaa619364a239fa43aad203641bf : SUCCESS in 1h 43m 11s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4480b7498aec4101b60a0f3baea6842b : SUCCESS in 48m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/57fd92ba666046af847bad6c0c1b73ac : FAILURE in 1h 34m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f1bf17a3046a4b95afa46c26bb5d5b5d : SUCCESS in 29m 52s","accounts_in_message":[],"_revision_number":1},{"id":"5823c442040bd98f5cd54a543d305060a49be6a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:20:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b08b8f19e6bd774d54c41726b3775a789d696c30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:22:08.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/835dc041a86c43a39f532e0b8ad345c2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/497d24e0dff44c4193e7eecbb95e226e : SUCCESS in 55s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2","revisions":{"dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/77/923277/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/77/923277/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/923277/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/923277/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/77/923277/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/77/923277/1"}}},"commit":{"parents":[{"commit":"ef22a7425a5fe4f7006f9434a33dd092e2594830","subject":"Support Stream Optimized VMDKs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef22a7425a5fe4f7006f9434a33dd092e2594830"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:31:14.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
