)]}'
{"id":"openstack%2Fglance~923281","triplet_id":"openstack%2Fglance~stable%2F2023.1~Ieea7b7eb0f380571bd4937cded920776e05f7ec4","project":"openstack/glance","branch":"stable/2023.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 04:01:50.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"Ieea7b7eb0f380571bd4937cded920776e05f7ec4","subject":"Add QED format detection to format_inspector","status":"MERGED","created":"2024-07-02 14:10:45.000000000","updated":"2024-07-05 10:43:48.000000000","submitted":"2024-07-05 10:42:23.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923281-bug/2059809","meta_rev_id":"415f549cd5f0f16e36f46cc05e06b075ae9a4706","_number":923281,"virtual_id_number":923281,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 10:42:23.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 08:33:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 08:33:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:39:28.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 20:57:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"9f677d3b6a0a534d607a662142ed4f6dee48c273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d3053beea8237531c056befa938b60a83792c318","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:39:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2614299ed3fd4d35a507b86332233aa0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/25d7dfb79293462486fd38c70d17540a : FAILURE in 3m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"158828fb5278b3bb341a433eb1e0a376973b81aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 20:57:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/11ba779f2495478887da11ca5d66c31a\n\n- grenade https://zuul.opendev.org/t/openstack/build/44327a91d08542d2820c418ad5d05062 : SUCCESS in 1h 00m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/515c2f9892f64645ba819518c65faeab : SUCCESS in 59m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddb499297e9740f68898d727ba626f0e : SUCCESS in 1h 18m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/96789d4eef904ed58e7e45dbc78ddc1b : SUCCESS in 55m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/70d2ff8937c14587a7c073b5e72b8c4b : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0866c7fd0f3842829c774d073d58aa57 : SUCCESS in 6m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c265f09c959a4c68a71a896a4b91f636 : SUCCESS in 5m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/245f9d7c56a14fa9bf3d791203b470ff : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43cf77bff4c246619d08dfd848ff7c07 : SUCCESS in 10m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b94598d572264465b1ccfb9b5640ae77 : SUCCESS in 1h 41m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/233aba49b76a4398acff5d9319a54598 : SUCCESS in 1h 24m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fa5ccd5902bb4966abe8e191c06025d4 : FAILURE in 1h 44m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/055dd5e1cbad466dbf299b941bad4e96 : FAILURE in 1h 24m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a1325de9217d45428d1bd876402f5dda : FAILURE in 1h 37m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/44231c3b1fe0493386d89befa041a5dc : SUCCESS in 47m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2bf43e27ad6b45ab97ff7ca7708ed144 : FAILURE in 2h 07m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2a4ac2e1e8f4f7292613f8cbdd1246b : SUCCESS in 38m 59s","accounts_in_message":[],"_revision_number":1},{"id":"a8ab98f5a1f7a4df7fdd91e20031d9ba4eb58a4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 04:01:50.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"617e801647ef6d3c9a269781a0ddd277d5da6ce4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 04:13:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc8398922f764391a46511fd8314fc97\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c3cd81c5e9044d1d90f1ea638a9c8af0 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a52ec686c3dba480834e736f1ca0810553bd7f5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:03:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1914671f0dec4e44a921aa1b0e066b15\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5d5214527ec4a8bb84c09e0c865ffd9 : SUCCESS in 1h 04m 58s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f4e09a8131884687a8a59261f2c005ca : SUCCESS in 52m 39s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b29be45809874c41a416b0feb0036a17 : SUCCESS in 1h 18m 40s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9260473628ad420a9390a2d3e752dfe2 : SUCCESS in 1h 08m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1665fe41b1bf4f3d81afa37f1d11b373 : SUCCESS in 3m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/03167b2ecd334e31b34205ee36958c5b : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80a1d8440b6d42eda6a3d4ea32aa8b03 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5da5a5fa74ed4fd8846779928354a0af : SUCCESS in 7m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f10b74046e9141bc8d0cb5a4632629fc : SUCCESS in 7m 27s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3dfe8918841b47598af56ae029fbc5ce : SUCCESS in 1h 18m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/86260df6b5fc471cb9fbea498f32df18 : SUCCESS in 1h 32m 45s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/27c48c8a9954449eba72e15c7ee665f9 : SUCCESS in 1h 34m 35s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6df56db453224b60a5f110a60bdd6cc3 : SUCCESS in 1h 22m 37s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d0c15889dbec41a5a8c06e8fa9f2efdc : SUCCESS in 1h 57m 26s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac4989b6843249d38a64098e730c9dcb : SUCCESS in 1h 01m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/44b0b3e3055b4446960ecdbb19f1877d : SUCCESS in 1h 43m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/90d304a27a1e4481b6773368f8ce6b1c : SUCCESS in 44m 45s","accounts_in_message":[],"_revision_number":1},{"id":"9cdaaa9e5b62dae8d5a66758a838c40949e8ea23","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 08:33:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4946fd63ee3a59dd58f7cab926364518a6cbce8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:36:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f1e55e756b9886999cbd751ef0d5f24649f75777","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:42:23.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4d678fec22141c5a461d23629b04d6a\n\n- grenade https://zuul.opendev.org/t/openstack/build/e4d79b785c5d40dfab93dcf759832fdc : SUCCESS in 1h 06m 19s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/704946eabd744cb39339d135e7461048 : SUCCESS in 55m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/52a276bb0a09488da88ef67c2acf4ac5 : SUCCESS in 1h 11m 47s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e4d9106f99846a8a60dcc84fdc01193 : SUCCESS in 1h 04m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b21824c6c2e46229f5ff06efdef24e2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102a0fea234a458099d5d0fd230c1658 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c11d52f4b9f8415394709fa9f3a2f418 : SUCCESS in 7m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d0a4b2e1535b4eb2ace22e1b658f762e : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/92d2e184ab1d45e69740be76727dd7da : SUCCESS in 7m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2a528def8475428dbe4d0fcc23aeb5a1 : SUCCESS in 1h 27m 44s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f697106297fd4b04859496203f36fa0d : SUCCESS in 1h 33m 16s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60b2337d36d6402a92cc3d395413a5c0 : SUCCESS in 1h 23m 24s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/490733de26ed4a3c86fa3d943c655354 : SUCCESS in 34m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e4b6b7afce5d4ca79d213853d2b1bb78 : SUCCESS in 1h 56m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15b3a8739acf47a2a83844fd03213cc0 : SUCCESS in 34m 18s","accounts_in_message":[],"_revision_number":1},{"id":"fc31de8bffbd6c0ed814a26b328a6a49a47f00e0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:42:23.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"415f549cd5f0f16e36f46cc05e06b075ae9a4706","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 10:43:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a6e53c5d8d24bd69e58253dc19b49e0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1726bab0f8a247399a174dcb3e74476b : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9b3faf376bfc911c7cabde458507a0c65d63d70c","revisions":{"9b3faf376bfc911c7cabde458507a0c65d63d70c":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/81/923281/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/81/923281/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/923281/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/923281/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/81/923281/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/81/923281/1"}}},"commit":{"parents":[{"commit":"c1c54abeaba91ae0030d4acf01a91339b60a2d7d","subject":"Reject unsafe qcow and vmdk files","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c1c54abeaba91ae0030d4acf01a91339b60a2d7d"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-27 16:33:55.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:31:14.000000000","tz":0},"subject":"Add QED format detection to format_inspector","message":"Add QED format detection to format_inspector\n\nThis merely recognizes this format and always marks it as unsafe\nbecause no service supports it. This prevents someone from uploading\none that we will ask qemu-img to inspect.\n\nChange-Id: Ieea7b7eb0f380571bd4937cded920776e05f7ec4\n(cherry picked from commit 4096c5aff1d046d5c28d0e4a69b3c9574e9e8cc8)\n(cherry picked from commit ba98022b98ef5b9c98e1d7d20c88e1ca4b23fa80)\n(cherry picked from commit a8dadcd7994c88a61f94341286b4bcd693b51b32)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b3faf376bfc911c7cabde458507a0c65d63d70c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9b3faf376bfc911c7cabde458507a0c65d63d70c"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
