)]}'
{"id":"openstack%2Fglance~923283","triplet_id":"openstack%2Fglance~stable%2F2023.1~I447e7e51315472f8fa6013d4c4852f54c1e0c43d","project":"openstack/glance","branch":"stable/2023.1","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 15:55:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I447e7e51315472f8fa6013d4c4852f54c1e0c43d","subject":"Add safety check and detection support to FI tool","status":"MERGED","created":"2024-07-02 14:10:45.000000000","updated":"2024-07-05 16:58:14.000000000","submitted":"2024-07-05 15:55:04.000000000","submitter":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923283-bug/2059809","meta_rev_id":"7df36b1bef8820bf8d0c518af41acce98b9aba04","_number":923283,"virtual_id_number":923283,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"all":[{"value":2,"date":"2024-07-05 15:55:04.000000000","_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-07-02 17:28:40.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":2,"date":"2024-07-05 15:54:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":1,"date":"2024-07-05 15:54:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 14:10:45.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 14:51:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-02 17:28:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 15:55:04.000000000","updated_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"reviewer":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"state":"REVIEWER"}],"messages":[{"id":"9c56d7202efae384be66fed295da7e2462b114f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-02 14:10:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6efedcbed1a128165812b55bc9f443d05cbdffc6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 14:51:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/345178c340d54d5584882ae3bfd9ca97\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/918629753db642a28751d991e313f522 : SUCCESS in 13m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"bc067228ed65c1596e794f75db02fb4bc113f021","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 17:28:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0db39843d7d4ec8914ec0f38ef9e790\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/2ecb267f25284a9ab669fe23fe3cc22c : SUCCESS in 55m 07s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3928cbe81e5f4e11b4dc100040f43629 : SUCCESS in 54m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8274b183489d4e71b68686a506fd8231 : SUCCESS in 4m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8a7a1419d744b7abd5628e39ff300c0 : SUCCESS in 4m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a03a69adbeab4e29b4d6749dbc97466e : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/768cd7d670864e34a0f49473acbb5b0b : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3e7ede43afb74239a0182e488ba9f0f4 : SUCCESS in 9m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/685faaa6d626434c80a8ce7dfb27c29a : SUCCESS in 34m 42s","accounts_in_message":[],"_revision_number":1},{"id":"9376f2bed351089c3fd1d766c803a9fe5d92cf98","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 10:02:01.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e7ec2385214cf6439880b447a66c81737e3f2eb1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 15:54:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee2d0bb195d71abdc3010aa0b9680974b764718e","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-07-05 15:55:04.000000000","message":"Patch Set 1: Verified+2\n\nBypassing Zuul to merge this.","accounts_in_message":[],"_revision_number":1},{"id":"3394aacbe9e0556fe5430ebe8cc5144603c6a2ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"},"date":"2024-07-05 15:55:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5091f2c2d22c2e899fbf2d9cfddd0b4bd903a12e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 15:56:45.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/abca542b58cc42b7bed5ddaff27b3494\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bd007d7db9e84444a3aef349818e31d4 : FAILURE in 1m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2f0903474ec1850485abd70bc1419f45852a7b8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 15:58:26.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/af5494cad92c4d31aabab2e591d3c15d\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c15dc7fa97e94a9cb20774b4245c4428 : FAILURE in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7df36b1bef8820bf8d0c518af41acce98b9aba04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 16:58:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a99759e9d2dd477eb56a2d72fc9d8caf\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d8bb453e050e4046be36dd6f26fff81d : SUCCESS in 53m 42s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b64634d264b4585b1cea2b170a64b16 : SUCCESS in 57m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a8d585df4ccb4efe8736ef5a848fe365 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38fe4f96dbde4c929535442929eeed06 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f07f11630aa4f2c9ca2421d1edd3bd8 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db92749173404498976ee2241c670dba : SUCCESS in 10m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a8a80f4f1dbb45d2a28216a85e651730 : SUCCESS in 8m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b0c1bc0cf4524b1a8d2406c51756f346 : SUCCESS in 36m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a1c94b47d02a7bd3ac076ea02a23e456da2d5a62","revisions":{"a1c94b47d02a7bd3ac076ea02a23e456da2d5a62":{"kind":"REWORK","_number":1,"created":"2024-07-02 14:10:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/83/923283/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/923283/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923283/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923283/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923283/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/923283/1"}}},"commit":{"parents":[{"commit":"1656fe8832db6cfff96dad09b13584b0037187c5","subject":"Add file format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1656fe8832db6cfff96dad09b13584b0037187c5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-26 15:41:02.000000000","tz":-420},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-01 13:31:14.000000000","tz":0},"subject":"Add safety check and detection support to FI tool","message":"Add safety check and detection support to FI tool\n\nThis adds a safety check and detection mechanism to the\ntools/test_format_inspector.py utility for verifying those features\noutside of glance.\n\nChange-Id: I447e7e51315472f8fa6013d4c4852f54c1e0c43d\n(cherry picked from commit b27040b87e43ff4acfb6870ef0d13d54e5ca5caa)\n(cherry picked from commit b394ef00c3426771092d099cf1d96077bfa4b919)\n(cherry picked from commit aa12d39b453068d9ee8367591eb60d4a15cddece)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1c94b47d02a7bd3ac076ea02a23e456da2d5a62"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a1c94b47d02a7bd3ac076ea02a23e456da2d5a62"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":32613,"name":"fungi.admin","username":"fungi.admin"}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
