)]}'
{"id":"openstack%2Fglance~923304","triplet_id":"openstack%2Fglance~unmaintained%2Fzed~I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","project":"openstack/glance","branch":"unmaintained/zed","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-15 19:05:58.000000000","reason":"Change was submitted"},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-07-15 12:29:12.000000000","reason":"\u003cGERRIT_ACCOUNT_1004\u003e replied on the change","reason_account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}}},"hashtags":[],"change_id":"I6326a3e85c1ba4cb1da944a4323769f2399ed2c1","subject":"Reject qcow files with data-file attributes","status":"MERGED","created":"2024-07-02 15:03:47.000000000","updated":"2024-07-15 19:08:08.000000000","submitted":"2024-07-15 19:05:58.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923304-bug/2059809","meta_rev_id":"29e5b59cf4c163af54b961867208561cfd16465d","_number":923304,"virtual_id_number":923304,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-15 19:05:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-07-15 08:31:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-07-12 12:45:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-07-12 12:45:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-15 12:29:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-07-15 08:31:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 15:03:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 19:35:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 08:41:05.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-08 16:40:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-15 08:31:40.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"bba0c7c9b354a54e8c56e8905c057925015e3ea0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-02 15:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6227d02c0c47ad61d4a2a30daee497faa81b9928","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:35:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/13ae0aee2b504f3ea76b9959196bc282\n\n- grenade https://zuul.opendev.org/t/openstack/build/dd960c84c7414bfab37daa172d846ce6 : SUCCESS in 59m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c84b9b079d9047ccb29d0bb775090dab : SUCCESS in 1h 26m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/843d6ca5733d4db1a1fda13e15413130 : SUCCESS in 1h 01m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef1f81c07fc64bbeabfb4d562217f083 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/31365e1449764cd1917a775880a518e0 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/297d615eed224953b5af37a09af8addb : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cdea87888e924f13811923a458a68210 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19f21afffa5d4b98b2c137412caa263f : SUCCESS in 5m 37s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e78ae943670743d696682b4f847e39fd : RETRY_LIMIT in 1m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb1737f8715b4fe692816c599d9bfd06 : SUCCESS in 8m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/55772120312743abb295c04bf224d0d9 : SUCCESS in 8m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7d427f190df04f5dbcad0c3425ff2074 : TIMED_OUT in 2h 04m 13s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b7eac559326a4852a116b97f00982a13 : POST_FAILURE in 1h 18m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/72de2dd4d0954d7e9e66ed05c8c54594 : POST_FAILURE in 1h 06m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d93f6ecbda7a4a61bd682c0dcdf096e9 : POST_FAILURE in 1h 27m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f43aba7a8def4566b23c0310ee47da49 : SUCCESS in 57m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5393c2fdc7c147519ec68c3de5d57959 : FAILURE in 1h 52m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/90e76e24f9c14cebbc6f9dbbb9c28b68 : SUCCESS in 26m 43s","accounts_in_message":[],"_revision_number":1},{"id":"3ee1b0fe6cae9e2079c23dd950ec4d988f0fc155","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 08:41:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63c365d6c99e32d9aede5506284c72f806df0afb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-08 16:15:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2905444746e667b3e7e90cd0f20b1595e528aaa2","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-08 16:52:55.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01f383b7811c06164a7b91cea26b9f70e04f8dda","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-12 12:45:53.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"16b331c1d8f14d42a63c22c1af563d008ec77d52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 16:15:26.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7779e18549140adaccfb3cddbf4aa6d\n\n- grenade https://zuul.opendev.org/t/openstack/build/7df8b4ab552f44ffa6e735a20dafd5c2 : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/71036dd5e76849f98b1caacf6e0abd1c : SUCCESS in 1h 20m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4af38d4c46204a0f83ab2a3809477986 : SUCCESS in 55m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0539bb48c908473aa989b258e0a4a5ab : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77956fdf344a4351b090638a65cdf128 : SUCCESS in 7m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de4a0b50a40c4620928aa1276773c237 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5c8fed3dd9ee4ba58e439595b56c4d92 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acbe70ced4054d328650a5f3483efbc7 : SUCCESS in 5m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c49b15821c0c4ae1947e70797b736cd6 : SUCCESS in 9m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/59831ed9c8cb4e7e8ec7b2346a26fb24 : SUCCESS in 9m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/69504d055159446da914a737723b2348 : FAILURE in 1h 48m 57s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e514070ec24245b5b5f22cf9f55d48c4 : SUCCESS in 1h 43m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/327f314d60b54673a08da3f8a4e73a21 : SUCCESS in 1h 30m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/799a5f117b0243eb90225adbc51ac4b1 : SUCCESS in 1h 33m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6a19368182d4cb0936362135b4d4358 : SUCCESS in 1h 00m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/72cdf94c62ef43229d097b6dc55ec9ab : SUCCESS in 1h 41m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e5ee7a0ad382460c9971035d4b6e88dd : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":2},{"id":"8301ee20f5e844d6d5fcbed5c304c8816de7429e","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-15 08:31:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2c20eba0d874632b95c19b1b042a592d5f77933c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 08:35:42.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"48137f8eb95395cfe162f44351333e4fc9cdbf0d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 10:15:21.000000000","message":"Patch Set 2: Verified-2\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/923303 is needed.","accounts_in_message":[],"_revision_number":2},{"id":"520584e7b1b05d1776839ba665dd1adcbb4388e4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-15 12:29:12.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2dc640a72ab5671da34582126716c6525091a52a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 15:36:54.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8c6735a17d24bebbbc5fd5151fb0dc6\n\n- grenade https://zuul.opendev.org/t/openstack/build/2234676ea9da4f3b8d2aeea6d2d63e00 : SUCCESS in 55m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1711ff01851444d5a24bb6354291c950 : SUCCESS in 1h 14m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02a19ced76134a7ba8d01f832dfff5e2 : SUCCESS in 50m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a08a31a2490e424bb390e4bff0801df5 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/baf1fc53878b4b7c97855fa19bb97b0c : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4fdef3df5d348aba4462fe6a34bd96a : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fcd9e239fdea405f937bd9e84475125c : SUCCESS in 5m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dc0f62f5085240b98d208e93b63bc15b : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6da985b4ee73455a9b3a45b3ab74ac21 : SUCCESS in 8m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/09ad71e69c1b40d08117b7b999a4787c : SUCCESS in 8m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08c4e290103442ae9580cebf449767dc : FAILURE in 1h 25m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/75c4fdb30a4c48f8bc5f4752f707b64e : SUCCESS in 1h 07m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c27d273f677e44728fcce8742caf67c3 : SUCCESS in 1h 36m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6fe3a936abb945a0b24212cf9b5182aa : SUCCESS in 1h 20m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3e9a1f95e1b401596f49bfed806299d : SUCCESS in 40m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c292e7160254a98933c76a6888add8e : SUCCESS in 1h 15m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1cc05a23e4654b53a93372ebf6f99755 : SUCCESS in 30m 31s","accounts_in_message":[],"_revision_number":2},{"id":"3f8e25ddf30af1239251e59c6057501a635cd784","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 15:38:38.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"187eb7f2f1fbd21f1eb48e8fb0f9756ad7d3ec09","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 19:05:57.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da9b08deb93b46088890cb6bee010e72\n\n- grenade https://zuul.opendev.org/t/openstack/build/95237d1ed3fb4e9188b0b97b9ca354dc : SUCCESS in 48m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b08096670e2f406fa0e3cfb41e6187a0 : SUCCESS in 1h 02m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6d184470941f4352aece9839260283b2 : SUCCESS in 51m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ef9ae352b144b08b833fe690749e2ab : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3864a5dd71b54ae7a75a4767642f0797 : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c74c7c4bb674392a81d8b83aab82c7a : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7c080d9ac644c03bf76c3338debd239 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/db5a36684cb147ad9f7d2640d17051c2 : SUCCESS in 8m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea61621c4994488698436eb71d654079 : SUCCESS in 9m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e45b9bac750a45639f28664956b193eb : SUCCESS in 1h 23m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/324656c6e231428a9f3c93f25435d2c8 : SUCCESS in 1h 32m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1c48029c703458f9b0d3430e0b42711 : SUCCESS in 42m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2d975714d7554526bb1f1aa42f323b6e : SUCCESS in 1h 33m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6f9178b583a44815a646e30d7a1b5213 : SUCCESS in 28m 43s","accounts_in_message":[],"_revision_number":2},{"id":"dcfbc638c7dfa33a40324a8f0c7d18e75cad6f1e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 19:05:58.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"29e5b59cf4c163af54b961867208561cfd16465d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 19:08:08.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c187f6f8f9864765b02f7d6ad2eb9a26\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cdc59726de3b4eb8996d5c9c20ec9296 : FAILURE in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"6a38aef8baaf5caecbd8c866f1cf922d939dfbcc","revisions":{"5576ee7b7e094e33cae78aad85b24bacbed850d3":{"kind":"REWORK","_number":1,"created":"2024-07-02 15:03:47.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/04/923304/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/923304/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/923304/1"}}},"commit":{"parents":[{"commit":"f6b5041a384e9bc82baf76b2dcef0fb9b3b76e20","subject":"Support Stream Optimized VMDKs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f6b5041a384e9bc82baf76b2dcef0fb9b3b76e20"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-02 15:03:41.000000000","tz":-240},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n(cherry picked from commit dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5576ee7b7e094e33cae78aad85b24bacbed850d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5576ee7b7e094e33cae78aad85b24bacbed850d3"}]},"branch":"refs/heads/unmaintained/zed"},"6a38aef8baaf5caecbd8c866f1cf922d939dfbcc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 12:45:53.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/04/923304/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/04/923304/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/04/923304/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/04/923304/2"}}},"commit":{"parents":[{"commit":"bcfc1e4dee268dc182d73f54729a840194c90d49","subject":"Support Stream Optimized VMDKs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bcfc1e4dee268dc182d73f54729a840194c90d49"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-01 15:06:31.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-12 12:45:53.000000000","tz":0},"subject":"Reject qcow files with data-file attributes","message":"Reject qcow files with data-file attributes\n\nChange-Id: I6326a3e85c1ba4cb1da944a4323769f2399ed2c1\nCloses-Bug: #2059809\n(cherry picked from commit 2ca29af4433e9fa99a0a48e230d8d25d6eaa4a87)\n(cherry picked from commit c3586f3a122f6cb0663217b12b52203e74e2e4fa)\n(cherry picked from commit a92c438fb5ba55440b38cae7c8b4361b58daa9dd)\n(cherry picked from commit dba3bdb458aa8a5d0193f12b7f1e374a89ed34a2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a38aef8baaf5caecbd8c866f1cf922d939dfbcc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a38aef8baaf5caecbd8c866f1cf922d939dfbcc"}]},"branch":"refs/heads/unmaintained/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
