)]}'
{"id":"openstack%2Fglance~923307","triplet_id":"openstack%2Fglance~unmaintained%2Fzed~I0554706368e573e11f649c09569f7c21cbc8634b","project":"openstack/glance","branch":"unmaintained/zed","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-08-28 08:51:25.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-07-26 17:43:48.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_10342\u003e by clicking the attention icon","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-09-12 12:30:00.000000000","reason":"Change was submitted"},"8313":{"account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"last_update":"2024-09-12 07:51:13.000000000","reason":"\u003cGERRIT_ACCOUNT_8313\u003e replied on the change","reason_account":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-09-12 12:30:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0554706368e573e11f649c09569f7c21cbc8634b","subject":"Reject unsafe qcow and vmdk files","status":"MERGED","created":"2024-07-02 15:03:47.000000000","updated":"2024-09-12 12:31:42.000000000","submitted":"2024-09-12 12:30:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":23,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"923307-bug/2059809","meta_rev_id":"b2483c46a403288bbe9a3a1289269e2136dc8e0a","_number":923307,"virtual_id_number":923307,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-12 12:30:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-08-28 08:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-08-22 12:08:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-08-28 08:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 15:03:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 19:42:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 08:46:39.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-08 16:39:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-16 05:49:36.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"CC"},{"updated":"2024-07-16 14:56:21.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-08-22 12:08:09.000000000","updated_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"reviewer":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"state":"REVIEWER"},{"updated":"2024-08-28 08:51:25.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"95062672f341511edbb4d9cc53267a36c38bc20d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-02 15:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"fd36e303ac21ac083043b69dd3fef40aa64d6fb6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 19:42:52.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/02351c56eeab4f5c93dc4564e4a97c68\n\n- grenade https://zuul.opendev.org/t/openstack/build/6548f56c6be84d70b82898441f2bbcb3 : SUCCESS in 47m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ddfea4608b1942379fdb7fa2b5545190 : FAILURE in 1h 22m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/00d5c1ab659542b6ae2048cd02cc40b4 : SUCCESS in 52m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/859e0bdc24974bd49a344481edde9b81 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cdecfb8f8329437d971fe42557f7e7f9 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/476fe14b579041d69b9967b20dfb8731 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b636b94dad154f828111d0020410c817 : SUCCESS in 5m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e99098859e5349b8aa39c926b34543ff : SUCCESS in 6m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/e6819e19bd95426bb3be7805b280371f : RETRY_LIMIT in 2m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2810e143bd2450da61464ef802742e3 : SUCCESS in 8m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2e5200c1d16d4d18bc82bb1e3f20c719 : SUCCESS in 8m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6d368917e8e84d33a279b5caea92fab7 : FAILURE in 1h 43m 38s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/483abde6b38746d78657f68826d5c6ed : POST_FAILURE in 1h 23m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4fcb8ca6149b4f3dabd0ea7969429ddf : POST_FAILURE in 1h 10m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/29a5452179c0417f90b6cd97efaf08d2 : POST_FAILURE in 1h 34m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/974cf015a3834dd2bdcfd3709cc805ca : SUCCESS in 49m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ec55a45ce1a4d1d91891a32871035a9 : TIMED_OUT in 2h 04m 50s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2e2b54029a2d4f0b83910c79503a1a8b : SUCCESS in 38m 44s","accounts_in_message":[],"_revision_number":1},{"id":"139828979024faf2a8a73da42dbc72e0bb4b7143","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 08:46:39.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"335ae6d1cc69786bc586f6813984cc1208ca209a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-08 16:18:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"662d7bc941ca9e1fdeb67551b1ddcd08ffd157a5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 16:39:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d44c9047444ef26fcce3a43bee8b8d8ba525c40d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-08 16:40:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"61faa0b6b1610a662f575b6fa47e3ab8cc32ba73","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-12 12:47:21.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a254dae5b55583182d062590cb49641b445d55da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 14:43:18.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/31ecc5a98621479cb3f9410ef48636e1\n\n- grenade https://zuul.opendev.org/t/openstack/build/f3dd47e4d1b04d7a80f3927b4da6de3a : SUCCESS in 51m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/04d9709b1bea4f129aedaac9d8eca5ad : SUCCESS in 1h 15m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f50d0c588e654d9db2ba4420a87f575f : SUCCESS in 50m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b75bf4818dc343bc83c3281ac32617ee : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb517026c36f40c3b5808b07fdd8e380 : SUCCESS in 4m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/26eaa304d555435f85812cef39f3e0ab : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f99d633c2ad04f6d8924119b2d9e2200 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b0f0ab748274016b80b40d126587822 : SUCCESS in 5m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e53f4fa3c9ed4f31be8d6b4c210e6249 : SUCCESS in 8m 21s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fd5f4f3bd91c4318be149ec3c083ecd1 : SUCCESS in 8m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c756bffad0bb494b902301ca81d7b54d : FAILURE in 1h 50m 36s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cdf5754eba88493481e3cad163daf98a : FAILURE in 23m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5f86ee84e8c24a4ba0e2b514d524a0ec : SUCCESS in 1h 35m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2eac2da882904948b8df56cf35c9d4c9 : SUCCESS in 1h 37m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/63237af4ea334457ac191c38fba077a4 : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/75ef3f2110f344ab86d244238f8c762a : FAILURE in 1h 40m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7c09d931f3c24eb68c28dc27fc3ec016 : SUCCESS in 29m 17s","accounts_in_message":[],"_revision_number":2},{"id":"a22ddf8b57fbb0f8e6ce3f2b0de5450cfa1dbc8a","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-12 21:35:42.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ed0bc01e35ba14ae3cb9e7aaf866c647d7134792","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 23:27:23.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5257130bcb264566bfcb05294c28f96c\n\n- grenade https://zuul.opendev.org/t/openstack/build/5c6cf6d89b0949aca04f15a87bc2d04e : SUCCESS in 56m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d2e48c79947d4dd58902ca4a2bc8fa66 : SUCCESS in 57m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b02351b41a1d40a498e010f77b3dfdae : SUCCESS in 45m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8506f846146e49a689b7a3f3693c3bb8 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3f6a1d008a24e98aa2612af98766ebe : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/75ce3cb254b24089b2370aa524e4708c : SUCCESS in 5m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d32d4a73416a404e9d97d9c52c2cecfd : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6158823985ff49519d7e6cb71b82658e : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10d98b02b52d402ab9662af8cb674cae : SUCCESS in 9m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0bafd732b3bf4d03bb9b61469d3dffba : SUCCESS in 9m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/89bf213a83ef48dd940ae55e0c10b9a7 : FAILURE in 1h 50m 33s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/76ecf15e04084835bd41ba9cf11841df : SUCCESS in 1h 24m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b3d60e08f421417f805ef9f01a36a2fe : SUCCESS in 1h 24m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a685a2e8a1774569be081a52f5002df8 : SUCCESS in 1h 32m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/76f38e58a9694696ba057e3c01e91ef3 : SUCCESS in 52m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dceae4e1118f48f5a99d888d1601d400 : FAILURE in 1h 39m 32s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37a84fd62cd045858ac8f0857c79ff5b : SUCCESS in 33m 04s","accounts_in_message":[],"_revision_number":2},{"id":"37f971d55db9850aba079c97e9d027532b363af4","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-13 03:15:21.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7ea847d829d1b095dc32ddfb5a38bb57a8c5be9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 04:59:54.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1eb5ae241f314489a88c942f2ce72f61\n\n- grenade https://zuul.opendev.org/t/openstack/build/77f6c40798e04abdaa9dd61a7f60fe3d : SUCCESS in 1h 01m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0338a54542c34db69925a6f7f834e8c3 : SUCCESS in 1h 12m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c3518e7a4d448e0ba3c59cd332cc0c8 : SUCCESS in 47m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/352d87b602c340ed8655b6f4a0403787 : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/abb336051a124ad9854f7ccc94cef6ca : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480cb6fcad7347f4bba2f60f405bea5a : SUCCESS in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66b48e75dc2741a78f0b541e29c0b8a0 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc5a1b41d7c9464db0dd86e21e5ff16c : SUCCESS in 6m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4544561b32414884803a4a89ee145975 : SUCCESS in 7m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1f185d94c57a45a49d66ff0babcd237c : SUCCESS in 8m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c12a689b3f9e49e497a5750f75b44a0d : FAILURE in 1h 43m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4eb451414d034a148dbf2475b62566e0 : SUCCESS in 1h 27m 15s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d234c1323eda489ba2f9e95c067b9f64 : SUCCESS in 1h 08m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3eed6d31756f45039f9bc49dd0f0505b : SUCCESS in 1h 30m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4dfde23f7e3244cb83d21ce9e2e0f3ae : SUCCESS in 56m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a6c262c8c7504b52ab4f0cf13f63b6af : FAILURE in 1h 15m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0a46fdca67e24a888318d55557ac0c44 : SUCCESS in 37m 49s","accounts_in_message":[],"_revision_number":2},{"id":"1ea763cd3cb5391d756174630193b35bc54d5edb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-13 14:53:29.000000000","message":"Patch Set 3: Commit message was updated.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"33adf3484471da5bbbf1d7ece688433f0234fcfb","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-13 14:54:08.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"f9e878c4bd662f1fa25abee2722e64023d06bf4d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 16:52:39.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9d3f4cfb6bc44cf89d4a326645fe1bbe\n\n- grenade https://zuul.opendev.org/t/openstack/build/823f27242e23425190ca03dbcc235bef : SUCCESS in 44m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d877cd1f427447ecbedec071a139e83e : SUCCESS in 1h 03m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/319d5e46ec864d75ab879a437d157344 : SUCCESS in 38m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c99c061bcabd4a3f9bc5c39791cca82e : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58fb2985081946b7b40146f8861b32fd : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/96254e387e8e441084b0cd5e6a66abba : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fd4ac055d6d7412c9284361dd6156745 : SUCCESS in 4m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36cf58ab83d744bcb29ed68018df2f54 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e540c2c7f4cc4d80b1dc3c64bb0247aa : SUCCESS in 8m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/3810d0b8088940e49a349e6a65d694b0 : SUCCESS in 10m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f3f239750c7e40eeae54d4e1a7c0c8a1 : FAILURE in 1h 42m 47s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/087703c07758467a982161593763acad : SUCCESS in 1h 16m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/69f9a9d025e74ecfa7d1814c2091ffdc : SUCCESS in 1h 26m 18s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/03bd599814fd449db3363246a8210613 : SUCCESS in 1h 22m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8a18620d3b0849449f0b0639156f5611 : SUCCESS in 36m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b1efacd1828642c4b6fa74712de8e756 : FAILURE in 1h 33m 49s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/200731103bfe49e4aee0b29304d77ce8 : SUCCESS in 40m 07s","accounts_in_message":[],"_revision_number":3},{"id":"cf9f6319f6734aaaed1bae84c64de818e847a970","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-15 21:33:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"606f781bf10449de6cf672185a19c37bffc6293f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 23:37:37.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d228abaf2bfb41fc9d3210b8a320e94b\n\n- grenade https://zuul.opendev.org/t/openstack/build/c71fcb46f5174179b81f7c492a04e713 : SUCCESS in 43m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dfd669454c884f7e9f3e7b46db7d240b : SUCCESS in 1h 07m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ded13f4a478549c69121862d95dd1537 : SUCCESS in 51m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7873770848e84dd3b3c33ef43a52333c : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09ed533ac17f4131b789000940313e42 : SUCCESS in 4m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0313e04ce1504cf2983a3b58f27c8c3c : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/04fe707698fe419985db0d540975c005 : SUCCESS in 5m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9769a31646a347188d5437c1e5d6bf41 : SUCCESS in 5m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fe3c988fddb241ad901cc3019fcfe91c : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2d10e2a99fdd4a57a626536312c06502 : SUCCESS in 8m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8bd052fd186749358d8179ae5853b5c3 : FAILURE in 1h 47m 20s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ddf9e6d5409947ec81ae4eb98e3e2bfd : SUCCESS in 1h 30m 40s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31ba0f74ff3a4000ad80b4bfc726e546 : SUCCESS in 1h 30m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2d23d064a80149a3ad4cbf736e111dd2 : SUCCESS in 1h 58m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9455bdfd9a4caa961443f936791402 : SUCCESS in 57m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bec10f5431fc46f8ba978db9356fa709 : FAILURE in 1h 37m 18s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/28c15aed1ed543e6bf2e90b849f5775f : SUCCESS in 40m 12s","accounts_in_message":[],"_revision_number":3},{"id":"b03d8b177d872d270930f319cc5750d352574ece","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-16 05:49:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"eff3a6a3ecfe8fc99c95b53a3796693b61ca4eab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 08:00:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d6ba8cb912c042b8a142a9e76b4362ba\n\n- grenade https://zuul.opendev.org/t/openstack/build/88e020f5bdb5478fa899d626c25d675e : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f0c87165de2a4dfcac61fa39a4f6c3d8 : FAILURE in 1h 01m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5012b3e892ab41d5b852fdf75b095472 : SUCCESS in 37m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a451ffa3dbe84ac79907b755380fcb7d : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ea5bfdee822040de9d9f84ebdb31401f : SUCCESS in 4m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ec52ac0ae474410a79a1e511f34b217 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cc51b3225e3a41e0a77c6499b54343d7 : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/99d9983ec78b49ac89d2a7befb4d55b8 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d1a6da0b5b0949ccb268bf787bba12b5 : SUCCESS in 7m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/807b5c604e1e4280bc80bd8ae6406e35 : SUCCESS in 8m 47s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f907f5943b0d4d2e9a5ede0eb1231a94 : FAILURE in 1h 56m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7a74bfb0eca64c40b8d966a2c0e616a3 : SUCCESS in 1h 14m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2537fc38f1234cfc9215edd3fdfbfa1f : SUCCESS in 1h 23m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5b7bef467d244882b58739bf858dc0ad : SUCCESS in 1h 36m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0331414d87fd4687b1e2857871429d42 : SUCCESS in 41m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f25b82a950584e779ef956a76f076fd3 : TIMED_OUT in 2h 03m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f6cffbaad03c4655af68ef841b41c012 : SUCCESS in 37m 03s","accounts_in_message":[],"_revision_number":3},{"id":"d14b3aa9a079dea786a0f7a9282e0ff81331cc34","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-16 14:56:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c11b336574f33eb62dc1ac262e33256bbddaf9b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 17:09:00.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/aa1b68de780d4ddbaa0716eb556f47b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/f9b7d3fa6f8e49e39cb3ab34ad4a1392 : SUCCESS in 1h 04m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4d229fece4e44455bb45812c4254f0a6 : SUCCESS in 1h 19m 56s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aa7a1dbc54174253a8a0d57e017d6864 : SUCCESS in 51m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23df79f3b806440c89c11141f5478a47 : SUCCESS in 6m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/191f8264021c4ab7ae57a7d61f642673 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71ade753fa774ea3b2cafa2722d1d530 : SUCCESS in 5m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bc6cce8bf00d4a65817ddd54872679ba : SUCCESS in 4m 53s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b0fc6bbb392492cac577e6e0f439758 : SUCCESS in 7m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/abdbfa27d1f946ce92d1a75c593e43c6 : SUCCESS in 8m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bc795f3e9bd546f08c149bc53629f5b6 : SUCCESS in 9m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2662939ff993488d8d6fb4c66d427620 : TIMED_OUT in 2h 03m 27s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a96b01e5c72e4bc9b3702ad09ef661c6 : SUCCESS in 1h 34m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/56602c1a97f3472aaef8817b4d9926b2 : SUCCESS in 1h 53m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d21b19cf4b56498aa37bcd9cc39d1a84 : SUCCESS in 2h 03m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/79a8a4ea216d4392aacfd22849f2314d : SUCCESS in 51m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17549b81530c4571846f712617cfc340 : FAILURE in 1h 42m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52124b1ebda24aa495521aadb950636b : SUCCESS in 38m 20s","accounts_in_message":[],"_revision_number":3},{"id":"2864a6287563dd621719264a6ece71b04a5d95f5","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-16 17:21:14.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"caf5dc8c577945caa0fc766f3fc828a71f220608","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-16 17:55:01.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"54097e8c06343cce142cfad837396279a1c10dba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 21:25:26.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/888a207d2b704818802875206c9fb816\n\n- grenade https://zuul.opendev.org/t/openstack/build/894cd97e1ab0464cbe1656f3c10b4531 : SUCCESS in 53m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9260ecd4f5c840f7813519bd80a99ac9 : SUCCESS in 1h 23m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/046661bc965f4cc69908adba364b91f0 : SUCCESS in 53m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d082a161fafa48cc8acab35027356a9b : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2441a857235e4049aedee8c73c1f6fc9 : SUCCESS in 6m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4b07e65af4b4961831940a606c0eef1 : SUCCESS in 4m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/af3df5198835404785a5326f1b246a79 : SUCCESS in 6m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64ec7c89575940938f11e8c5c6d9eba2 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/07d1dde72f6f421485f7eea32c4efcd5 : SUCCESS in 9m 18s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/36217192228641f197e4d5ba9f422532 : SUCCESS in 8m 08s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/909d1220097c4173a2e95c042ca9c9b4 : FAILURE in 1h 38m 22s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6fefabbb863f4e4da8ab759af9709940 : SUCCESS in 1h 29m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/93039c18264241579163517bf6f773bd : SUCCESS in 1h 25m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a11970541d9e4ebe951b63185d5e3e38 : SUCCESS in 1h 29m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9c42231223439ba74593d32986c68e : SUCCESS in 50m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/73f86e4371ce4ce1ab2ef895e87cf334 : FAILURE in 1h 41m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/04600205a338427d9c3dc8fcb188a5b2 : SUCCESS in 27m 55s","accounts_in_message":[],"_revision_number":3},{"id":"74f3029d1d89854ca29425564ac732e2463a200c","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-17 02:40:12.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72e3162947c9d086cf7f40db460b45192cdb9ba4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-17 02:55:02.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"2b6551f9e0d66c08455faa9fc8ab74326b82d02d","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 09:10:47.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"13740e2d1438918d805cc0027ca797de7eddf3fd","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 11:27:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"0e24237d65b15d404284e8630cb2a8fc08893b68","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 11:29:34.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"5f3ee487a8b81d3eefbca5482fe426ab66bf73e3","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-18 11:35:19.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"70309298346d1b0562b6d6e1627d23fb37a50d1a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-23 11:00:54.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"e873b24ea49c713d848e0c28408e592f56820a15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-23 11:52:41.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/034fa38d099647a2958cbf5e0ed96f5d\n\n- grenade https://zuul.opendev.org/t/openstack/build/3919ced447494642a245f0269e03aa65 : FAILURE in 17m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2f792da0b84d46ffbefb2c5c10021cce : FAILURE in 18m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/27e877902c3f450b91be5c6941603d8d : POST_FAILURE in 18m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d32df8ed7e24401bfad7234e2cda860 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2da868aace714ad0854e1eb493bae055 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3501813f641645b9b01ab99bf339ba9b : SUCCESS in 5m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dfcbd5f8b2f44f1198bd994b7a41786e : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5976317383b54f008cca16c1139eaea7 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f11b6ef7a0343d59a6cbd32d4a9cb5a : SUCCESS in 9m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/97adddbf1b494db6bbf967c8f9190c2d : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f55022432fc9467ba7cd264238a60633 : FAILURE in 26m 26s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/92bb21def5fe4e638651069b11c2b6df : FAILURE in 17m 54s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8f4af6281a474f3583657c4ca8862478 : FAILURE in 14m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3161fc4baf2f48e3a9ba805709fa8534 : FAILURE in 18m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c4c8af1587614c938da019f77435354a : FAILURE in 14m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cea6ea5dd1842c9973f96809bf64283 : FAILURE in 17m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/55ab44a13e684896af996750a54444d4 : FAILURE in 26m 43s","accounts_in_message":[],"_revision_number":4},{"id":"2d3f40a104147641cc1cbeb9762c3d3e5c69b810","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-07-26 07:35:25.000000000","message":"Patch Set 5: Patch Set 4 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"d928ba129c4c4308a1e235c32b0d8730ee70102a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 09:17:59.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/dac5faec40d24ff9921271411593a7a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/656a50da053e4685b48a9820ba1293d4 : SUCCESS in 54m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4470af15e7004c968258d827178e7f83 : SUCCESS in 1h 05m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/60b5e9009a6b47d9bcbda5d5cb6eeea3 : SUCCESS in 53m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/163f83cfd1c845aca2ac25a1fb625f08 : SUCCESS in 4m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7c98480b0d9d4945b1372da84bb0e7dc : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f442bbb5c9245a19c5c62f2d6f770cd : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2177632086e442ca3911b692020fa8e : SUCCESS in 4m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6456420f72f2408bad4b9b4227e9871e : SUCCESS in 5m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5215856b067544c68831f362bd7c95ca : SUCCESS in 8m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/170264e2f1704d99b18e555c68c15e81 : SUCCESS in 9m 31s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/337bd0ebb548488e82258b9a9f448170 : FAILURE in 1h 37m 25s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e809dd11bd8a4807a954b5d9216d2315 : SUCCESS in 1h 35m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f64dd2dde92f4a87ab61cbfcce8c5540 : SUCCESS in 1h 34m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b75dd26351048538e7fcd3468f76881 : SUCCESS in 1h 32m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b6dcefb716d8448991a75f03ee894fd7 : SUCCESS in 48m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ad067f3e7e354d00b2b56c7322f91ac0 : FAILURE in 1h 25m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6a9c7f20fcbd423f810aeee9c4de17b3 : SUCCESS in 41m 20s","accounts_in_message":[],"_revision_number":5},{"id":"9cbf566f10be398de7f2b7c51d0f02c1abd2fdcb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-21 09:24:55.000000000","message":"Patch Set 6: Patch Set 5 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e5c03de8b7e55af160b1710caf5c97c2c9b81bef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 11:05:29.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf1cfd9a06ed498ead8185225ddc1a8d\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff4e61e66a854e249db7a7b3d4b0924a : SUCCESS in 53m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a01a907e958046d0abb8c888a8f8808f : SUCCESS in 1h 08m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0f22d5b7d4fe43a1b1ef708de2c4d1b2 : SUCCESS in 50m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4edc45d78a5d4c3387eea3a866474b6f : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66ac3c71732a4d6a9bc79c18ec288e7d : SUCCESS in 5m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/313c708269134bafa6bccec2d4fdbe4c : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c74b453c3c38478b8babf293602ffcc9 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fcdc27d2c5946079059767e67cdd536 : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7725dfa4dfd7477ab14f91ba6b3599e9 : SUCCESS in 8m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b83bb9e9608b4068b6bc4c91978db22c : SUCCESS in 8m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1ee1749dc11c4e7483bc6df479b2aed3 : FAILURE in 1h 34m 50s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f352da1b41a249c79081ddf95596aa26 : SUCCESS in 1h 19m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e02342fe52d440a2b0571f57aabfc386 : SUCCESS in 1h 28m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9edfcdf5119e4d08ba73f8ac5a55b823 : SUCCESS in 1h 33m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/03d15bd5c5054eb2a99c6def51a894a2 : SUCCESS in 49m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bffac718cf784fb8ad01d4dca7e716d7 : SUCCESS in 1h 30m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4c0561f786994cc2866b2a643bcada1e : SUCCESS in 36m 01s","accounts_in_message":[],"_revision_number":6},{"id":"c203f26466de6a1e890bea58910039e16d92fc47","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-08-22 12:08:09.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"2514a51eafcde5a7b54f89ccc77417e895ff7a08","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-28 08:51:25.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9034fee503ba9997d2840cce0ac4789247d2ca65","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 08:52:11.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"7cfeb8a9838b6d3c5c02f4c2bf7a4b6499211412","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 11:52:20.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a5447e5df204491b8f54752eb74ef79\n\n- grenade https://zuul.opendev.org/t/openstack/build/517f8b28e41e4907a8f3978cc6767d9e : SUCCESS in 51m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5abfef6a7a4447d0b54855c555516f78 : SUCCESS in 1h 10m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ae0069abc934e0ca952473960eeaebf : SUCCESS in 57m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9a2a2136290941d081a4f89bb1f9c6f4 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f643cfb1b1aa4d64a8971e2cc31338ad : SUCCESS in 5m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c69d5311fac54b659d71b40ae481cea2 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c004361d86542dcbc3a7424e52b99b6 : SUCCESS in 5m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/005cc1c14d2a4a659f9ece73e8783c31 : SUCCESS in 7m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/928f07cfdc624831b9581febb89e6fff : SUCCESS in 8m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/63e5b2b11331440c9387199b7b9a5294 : SUCCESS in 1h 39m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/89a868577eb4406cb18f093d86271bf5 : SUCCESS in 1h 20m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/16606b72150a4f01bbd30e5963e449b9 : SUCCESS in 46m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f89c0cd0c9254afeb8ce0b69f51532fd : FAILURE in 1h 43m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2247d8acf9da4218ba615043e57224a0 : SUCCESS in 35m 09s","accounts_in_message":[],"_revision_number":6},{"id":"427376c3aeeaa26a6659180f5ba94adcf8f21631","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-28 13:14:31.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"632f2abde6b14b70a483069e0db11d85abc23082","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:10:51.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a94b983ef9542de979f82e71c761b8f\n\n- grenade https://zuul.opendev.org/t/openstack/build/022da84b086b471c825f7ff16b4d2c0a : SUCCESS in 52m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8894eb10ce044e12b93c25debec0bb5b : SUCCESS in 1h 19m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56a395780bc7489c8a0dc4f9745d52c2 : SUCCESS in 48m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/69561eb68f324e5f9426cd1f2ba05a2b : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85d215730cc3448bb5d85b26636f10c9 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0936e516d3924e42b99844fa5086d288 : SUCCESS in 7m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/92d3a4fe4a4d47fa825159f9b3be9cd2 : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b4ca48bf16c143baaf55005dcaa1f604 : SUCCESS in 5m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/72fdcff3296a45a39a3c1b0279cfc16e : SUCCESS in 9m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/03223dd2b32249109a88aefbdc45517e : SUCCESS in 7m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c29cd807195a45bfaf8d769253888335 : FAILURE in 1h 49m 39s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/796a5f94427140ae97935b62c75cd5a0 : SUCCESS in 1h 16m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f2ecea1c72f543f7a2e7a665cf409436 : SUCCESS in 1h 29m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2462b03f6ba14d4ea9dcb3a5a2c35eaa : SUCCESS in 1h 48m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/45daf82a2fc94eb183cf344d677a19d5 : SUCCESS in 53m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b3fe1439e2854f36b8395b3549afe83d : SUCCESS in 1h 38m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0a71d87bc754f9c91e200f3f46da1ac : SUCCESS in 49m 20s","accounts_in_message":[],"_revision_number":6},{"id":"770a0619ac8da60670c8333044aeed327ecbec2f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 15:11:32.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"3a857ffb07cdd83f990906294ff182dce1d532fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 17:22:01.000000000","message":"Patch Set 6: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d2e1fe277b1048b2803f9770798dcee2\n\n- grenade https://zuul.opendev.org/t/openstack/build/ff287a23fd2b416c9a3f5aa6cef7f502 : SUCCESS in 1h 08m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d168dac4efb43f5b8af87fb53259622 : SUCCESS in 1h 18m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3df0ddd99d954cd089e065fd8cb07b99 : SUCCESS in 45m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f9685f4286549348b282b3df839d1c2 : SUCCESS in 4m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c81e9e8673c406180d28b7453a67651 : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cf77342f7c29433a98fe14b13f069e2c : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f844448d9fcc44a096b2b39ff88a1ea9 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ce9a2afda91649d4b4b94b2819ede936 : SUCCESS in 8m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9cd35ad00a824480913d4195273e1dc0 : SUCCESS in 9m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/54810d1179354d9fa9f00f1d26ea54b8 : SUCCESS in 1h 39m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/69c2a94b73a3483d9563b52a00e5e038 : SUCCESS in 1h 36m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c5225b5b6cd42c4a76ff3b43468eadd : SUCCESS in 55m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c7ccbb8a57fe4fb7944e7d93cb15cc8f : TIMED_OUT in 2h 02m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/abef46c917ed4df4931eb64d88eaed0c : SUCCESS in 31m 56s","accounts_in_message":[],"_revision_number":6},{"id":"c04303dc4c3f8fffdbd5beabe30720e4e03f1045","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-08-29 09:56:52.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"3f7c657f5fdc5ba44e1a96a2e88e15b2f6c7a45e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 12:00:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0a864ca6cf524fd1aff79dfb6d7801b6\n\n- grenade https://zuul.opendev.org/t/openstack/build/644bbc131b3b4f50b8087dd222549b1e : SUCCESS in 50m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d608731abde045d2abd8d3201c1bac3c : SUCCESS in 1h 08m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/12b6bb0bfdfa4754b43b1837ff577261 : SUCCESS in 44m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80d137e36f3c4abebe2b55c4f03af46e : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/54e2cc1330924b438d18585d2e9a4719 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/79b929a5c7a1404d91ca0ac8d0832601 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aaea33af59f848b4ae0d78ea050b9e4d : SUCCESS in 4m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/469975b0a20a42fd86af3bbb6c49fac0 : SUCCESS in 6m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd922060fb0f44eca0f94e2800637948 : SUCCESS in 8m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7789cd39ea9d4b719ede91dee514909d : SUCCESS in 8m 59s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b4f6700ba722423d931ac9a4fc2e5f05 : FAILURE in 1h 46m 24s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2a3bbe68cae94754aabf2e7f5de57e93 : SUCCESS in 1h 26m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/14ac84e91e224f5e873d3b3363b7671e : SUCCESS in 1h 20m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/365f7a666c1c487c8161ba075008467b : SUCCESS in 1h 55m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/679e80a6ae4c425ca53938c7c6159d1f : FAILURE in 24m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/65c06e38fac247939a007f486ad460ff : SUCCESS in 1h 57m 37s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cdc04ef22eb64edb9c1cd16ababfea59 : SUCCESS in 31m 30s","accounts_in_message":[],"_revision_number":6},{"id":"181fed8530dbee48dd9173496d5b58885202145f","author":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"date":"2024-09-12 07:51:13.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"8f01bdaa6131a8730ff46d0da34ddbdafe4ab5df","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 09:57:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1f2b01b9de0471a88d87f5417a97429\n\n- grenade https://zuul.opendev.org/t/openstack/build/369afff18fa64fb58f8d77b336eedae5 : SUCCESS in 54m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/cc1d947d508745fbba0e6519ddf8bc2c : SUCCESS in 1h 28m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/630920f1d176437d828afaf10a35c8b0 : SUCCESS in 54m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e8b1cec7ec345e2af4f22056c52c012 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/46c30c70976b4962a6a9482fc7bb7c85 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/674d9a0ab1a44521a8eeb262d236d645 : SUCCESS in 6m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fee44384cade4ca99ccce736810f68ac : SUCCESS in 4m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6b2edc1aad534f738a9948f76866124b : SUCCESS in 5m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c5e35cb7470a431aacde5a32410c7c60 : SUCCESS in 10m 20s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a761111e50ce488a8cbd1154c442735d : SUCCESS in 9m 38s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ec824769252641f28e348cc17ec2c9f8 : FAILURE in 1h 49m 49s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3ff8cf81edbb439e9df0f285e7cac6be : SUCCESS in 1h 29m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ffea8153167a4c2c95ace72c227e701a : SUCCESS in 1h 20m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cceca49a7a0e4636b6340349229de94c : SUCCESS in 1h 43m 08s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7dbb9f9f4a504835ad0602853adf8c67 : SUCCESS in 56m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1e371a1dbfc41428a2a4c1d712b091c : SUCCESS in 1h 39m 38s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1ab3d49cd77d4928882dc897bc5eded6 : SUCCESS in 22m 53s","accounts_in_message":[],"_revision_number":6},{"id":"5df4a2b42279a3fc934cdaeeedf38397c89ad5ea","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 09:58:43.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1a2c0c02ed139c2f9f5cc972a83a9f70d325dd76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 12:30:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/253f3182aeee40a491ee371ba72fefa7\n\n- grenade https://zuul.opendev.org/t/openstack/build/33617c6c433646789e379bc1d4e52e90 : SUCCESS in 48m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4f4522bbcfbd4cf4999af7fbb99568d5 : SUCCESS in 1h 06m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ee337ecb9fea42bd868d25f9005c23a1 : SUCCESS in 41m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae15cf54d8bb4ec1873ca7fc3d6d99d7 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/49a79f679ebc4e88a2ecc596cfb8dd3f : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9be497a514747ad9a886fb55b7e32c4 : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9fbe0d77677b490bb519f5bff80c4eb7 : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/84ef5f7e4af04894b6493c9963a16146 : SUCCESS in 8m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4651d36e6ba344f8abe1be02b1b4d479 : SUCCESS in 8m 08s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/c0584b43ae544fdc9a23711d402947cc : SUCCESS in 1h 19m 22s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/864dbbafadfa4946ba83fb2746359a1b : SUCCESS in 1h 25m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a5383dec3b7d47cfbfd2c6a68876bb05 : SUCCESS in 41m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f562fb89eaa94b3bbd5d0a6f9e8bb291 : SUCCESS in 1h 28m 51s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/83251d7e3ea345f180bfc05b2ef5152c : SUCCESS in 33m 38s","accounts_in_message":[],"_revision_number":6},{"id":"c25b5f8058ed9442fd35f4afddb5cb840ccbc705","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 12:30:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"b2483c46a403288bbe9a3a1289269e2136dc8e0a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 12:31:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85808f21ccd546da9d9fbc2bcfd81c44\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/702804c40bb9461d8a14e56db80026d7 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"63ac35e3248ceed3176a65cdccbbf06f69b3c58e","revisions":{"82a3de3322ba47b7f545cfa9257dfcdaf6309cca":{"kind":"REWORK","_number":1,"created":"2024-07-02 15:03:47.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/923307/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/1"}}},"commit":{"parents":[{"commit":"e0960f3ea69ee678a33a1952f846fc93064a31e5","subject":"Add VMDK safety check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e0960f3ea69ee678a33a1952f846fc93064a31e5"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-02 15:03:41.000000000","tz":-240},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82a3de3322ba47b7f545cfa9257dfcdaf6309cca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82a3de3322ba47b7f545cfa9257dfcdaf6309cca"}]},"branch":"refs/heads/unmaintained/zed"},"8944bdf21bdc39cf667448d299ab1fa4e5c92d2f":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-12 12:47:21.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/923307/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/2"}}},"commit":{"parents":[{"commit":"05650e16f1dc6fde71b51e9e9bd9494226047c4a","subject":"Add VMDK safety check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05650e16f1dc6fde71b51e9e9bd9494226047c4a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-12 12:47:21.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8944bdf21bdc39cf667448d299ab1fa4e5c92d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8944bdf21bdc39cf667448d299ab1fa4e5c92d2f"}]},"branch":"refs/heads/unmaintained/zed","description":"Rebase"},"9cef946218d699ab4fdf32fbe0533e6ee1ba7f5a":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2024-07-13 14:53:29.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/07/923307/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/3"}}},"commit":{"parents":[{"commit":"05650e16f1dc6fde71b51e9e9bd9494226047c4a","subject":"Add VMDK safety check","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/05650e16f1dc6fde71b51e9e9bd9494226047c4a"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-13 14:53:29.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/923861\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cef946218d699ab4fdf32fbe0533e6ee1ba7f5a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9cef946218d699ab4fdf32fbe0533e6ee1ba7f5a"}]},"branch":"refs/heads/unmaintained/zed","description":"Edit commit message"},"3c000e23ccb5a5a3c708ccde8e9867a09f4d42ee":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-07-23 11:00:54.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/07/923307/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/4"}}},"commit":{"parents":[{"commit":"2186cd3805bae74f10f4016f3272b91894f06b22","subject":"Fix disk_formats in ceph job tempest config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2186cd3805bae74f10f4016f3272b91894f06b22"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"elajkat","email":"lajos.katona@est.tech","date":"2024-07-23 10:59:27.000000000","tz":120},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/923861\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c000e23ccb5a5a3c708ccde8e9867a09f4d42ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3c000e23ccb5a5a3c708ccde8e9867a09f4d42ee"}]},"branch":"refs/heads/unmaintained/zed"},"b55f146c7ccfa542994528181390a5f5aadf904d":{"kind":"NO_CHANGE","_number":5,"created":"2024-07-26 07:35:25.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/07/923307/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/5"}}},"commit":{"parents":[{"commit":"2bad95d72a65dee4520d74e45c49757ae7fe2572","subject":"Fix disk_formats in ceph job tempest config","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2bad95d72a65dee4520d74e45c49757ae7fe2572"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-07-26 07:35:25.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/923861\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b55f146c7ccfa542994528181390a5f5aadf904d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b55f146c7ccfa542994528181390a5f5aadf904d"}]},"branch":"refs/heads/unmaintained/zed","description":"Rebase"},"63ac35e3248ceed3176a65cdccbbf06f69b3c58e":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2024-08-21 09:24:55.000000000","uploader":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"},"ref":"refs/changes/07/923307/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/07/923307/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/07/923307/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/07/923307/6"}}},"commit":{"parents":[{"commit":"d4ab3b8f627f80d74a3d98e5bae70676c21d7a8f","subject":"Merge \"Add VMDK safety check\" into unmaintained/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d4ab3b8f627f80d74a3d98e5bae70676c21d7a8f"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-04-16 18:20:48.000000000","tz":-420},"committer":{"name":"Lajos Katona","email":"lajos.katona@est.tech","date":"2024-08-21 09:24:55.000000000","tz":0},"subject":"Reject unsafe qcow and vmdk files","message":"Reject unsafe qcow and vmdk files\n\nThis causes us to use the format inspector to pre-examine qcow and\nvmdk files for safe configurations before even using qemu-img\non them.\n\nDepends-On: https://review.opendev.org/c/openstack/glance/+/923861\nChange-Id: I0554706368e573e11f649c09569f7c21cbc8634b\nCloses-Bug: #2059809\n(cherry picked from commit a95f335bca1dfdd1c904ae475e9fe8c6806f2c56)\n(cherry picked from commit 55fc42563818fcf88b474233df242a796c918b3a)\n(cherry picked from commit f1f53075d69a9a1c006b3e25506e30eb0210de1f)\n(cherry picked from commit c1c54abeaba91ae0030d4acf01a91339b60a2d7d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63ac35e3248ceed3176a65cdccbbf06f69b3c58e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/63ac35e3248ceed3176a65cdccbbf06f69b3c58e"}]},"branch":"refs/heads/unmaintained/zed","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8313,"name":"Lajos Katona","display_name":"lajoskatona","email":"katonalala@gmail.com","username":"elajkat","status":"Ericsson Software Technology"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
