)]}'
{"id":"openstack%2Fglance~923310","triplet_id":"openstack%2Fglance~unmaintained%2Fzed~I447e7e51315472f8fa6013d4c4852f54c1e0c43d","project":"openstack/glance","branch":"unmaintained/zed","topic":"bug/2059809","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-10 14:04:53.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2024-07-10 14:51:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I447e7e51315472f8fa6013d4c4852f54c1e0c43d","subject":"Add safety check and detection support to FI tool","status":"MERGED","created":"2024-07-02 15:03:47.000000000","updated":"2024-07-10 14:53:02.000000000","submitted":"2024-07-10 14:51:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923310-bug/2059809","meta_rev_id":"98350de1f82c1a0fa2fe51ab11cfed878342b1ca","_number":923310,"virtual_id_number":923310,"owner":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-10 14:51:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-07-10 14:04:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":2,"date":"2024-07-09 11:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2024-07-09 11:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-07-10 13:58:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-07-10 14:04:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-02 15:03:47.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"CC"},{"updated":"2024-07-02 18:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 08:47:33.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-08 16:35:22.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-07-09 11:42:33.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"},{"updated":"2024-07-10 14:04:53.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"REVIEWER"}],"messages":[{"id":"1d7edb5e01a1da309a975e464a8d62377616d617","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-02 15:03:47.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bdd35fdb99e8df7f15f8cc47581e17ba5af9a60a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-02 18:29:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/242156e999bd406999782b69f7f6134e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4697eca231de440fa91331fba42e8b25 : SUCCESS in 53m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d599e5b9df1d497fbe24a7c4be6149a1 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f8fc62dd52e94ea1a32049483e40c838 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/825b6195eb534bd782420e485f72eeba : SUCCESS in 5m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/130129a6f6cb443c9fc230408475de6a : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ab1baf38b8a47b2a8ca6e1bfa9292d7 : SUCCESS in 5m 44s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/551050fad67a4f59bbf665c66d76e063 : RETRY_LIMIT in 1m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2971223e0498445986be6aaa8b104936 : SUCCESS in 7m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/02ae0c32a39644cc959beb6a56e0de4e : SUCCESS in 10m 13s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fc3cad368aa94a1abf2c1ffb50b84581 : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3385a240138c09589cee6cb0b8c98e4c83e8badc","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 08:47:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"740308fb32baf351b4bf402197d09695db76b7ac","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-08 16:19:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6bdce3fe69a1f4346e28c420873c230bbd16bebd","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 16:21:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a2ee036ec4f007dcad5867b520b1d86ec0cbecb0","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-07-08 16:35:22.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae5085467533a823bc6cd6dccc9729798cffdda4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 17:20:10.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d00ee454d1fd45508999096271c8a55c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3ce3c937c19340f3b9a9a0ef0c4d55d6 : SUCCESS in 43m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a05945fc791402981a208daa8a34cfc : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5331e409cf4449f1a5aa4644f01824b0 : SUCCESS in 6m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8510475b2f684378b874740f2ba9bf0d : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e094ad311c614579bd8bc270fc1cb839 : SUCCESS in 4m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b91e3a728f44915b2203c53344fbd57 : SUCCESS in 7m 30s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/64c126d04c7742b1ba6c4d255ddb36da : RETRY_LIMIT in 1m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e2462f911e4d40c0948c35ca51de447f : SUCCESS in 8m 37s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/514899c9ccac4ccc979d2aeaa6d269bb : SUCCESS in 9m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f5ffac39436641d1b4234f275bd23e25 : SUCCESS in 32m 22s","accounts_in_message":[],"_revision_number":1},{"id":"2ea37eb3b36e24b8e2cf33a22e675a05d54b2801","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-09 11:36:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"c44bb6916dca37c29c6d5208a27cd84224c14acd","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-09 11:42:33.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94112ef46fc8db85a4c29468592ccc1a720e829d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 12:31:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/06b07726ca144fda9a6f0580891c72f3\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d5ee3e676b4b4e9f9a127aa406eb36e5 : SUCCESS in 45m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7ff9a0e9f8ed459a81e4a662f7a0fb63 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/636b3e7f5d624ec598bc27d949836ffe : SUCCESS in 6m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f23772116223440187c050f1b9c092c0 : SUCCESS in 5m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/94ccc42d60b2427f9839875a85ea3acb : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/148302fef2c0434aa84fa6d265eb80b0 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/612036fe460e473c9ad04426afc7157e : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/06d562bbddcc4ea7a376f6c5aeaa692d : SUCCESS in 8m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b475a4aedfa74857a3a363118a2dc75b : SUCCESS in 38m 54s","accounts_in_message":[],"_revision_number":2},{"id":"20796a11219a70a9f257db197c0ab7aa8ac864b2","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-10 13:58:04.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"84f0fd88e76cd92be20a67cdf3810dd9c888923c","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 14:04:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76a6cd2aaf11cc0b0696715452ae5467ba517519","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 14:05:23.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"207fea295d85276499e6333941c9d600372c976f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 14:51:48.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a50cccd6e034d158d42030f10759641\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/41f5e28a65d54f6492c55df2260b6ab4 : SUCCESS in 45m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa31c46bf7cc418bb1ec121dda1a0231 : SUCCESS in 4m 33s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37376bc27e0442778aa9ad10829cbd0b : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/76eabfc19c384621b2c42292f8e2f65e : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a0c3dd680c2c4ca69758d56f84b2fb4c : SUCCESS in 5m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/99831dbe00a74dd4aa1e443dcd937237 : SUCCESS in 7m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a4af3553ae84dea9150a961ddbfdcf0 : SUCCESS in 10m 17s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/21d35173671f4523ab53047466748c7a : SUCCESS in 36m 29s","accounts_in_message":[],"_revision_number":2},{"id":"d3714bdcce4779952bce0402cfe92a4648898127","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 14:51:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"98350de1f82c1a0fa2fe51ab11cfed878342b1ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 14:53:02.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed63d72dfe144dfcb6f92755286acfbd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8ef605a95af4d1cb5e18ff8325a2f52 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a5ec69ba1c0a92cc9b91b52b4101b81a330ea684","revisions":{"2b6631e8a6a1adb8ab77709e4c5827ee0b8acf1e":{"kind":"REWORK","_number":1,"created":"2024-07-02 15:03:47.000000000","uploader":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"ref":"refs/changes/10/923310/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/923310/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/923310/1"}}},"commit":{"parents":[{"commit":"d90b3678c9087ffa39a34ac5c51048b064c29837","subject":"Add file format detection to format_inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d90b3678c9087ffa39a34ac5c51048b064c29837"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-26 15:41:02.000000000","tz":-420},"committer":{"name":"Mohammed Naser","email":"mnaser@vexxhost.com","date":"2024-07-02 15:03:41.000000000","tz":-240},"subject":"Add safety check and detection support to FI tool","message":"Add safety check and detection support to FI tool\n\nThis adds a safety check and detection mechanism to the\ntools/test_format_inspector.py utility for verifying those features\noutside of glance.\n\nChange-Id: I447e7e51315472f8fa6013d4c4852f54c1e0c43d\n(cherry picked from commit b27040b87e43ff4acfb6870ef0d13d54e5ca5caa)\n(cherry picked from commit b394ef00c3426771092d099cf1d96077bfa4b919)\n(cherry picked from commit aa12d39b453068d9ee8367591eb60d4a15cddece)\n(cherry picked from commit a1c94b47d02a7bd3ac076ea02a23e456da2d5a62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6631e8a6a1adb8ab77709e4c5827ee0b8acf1e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6631e8a6a1adb8ab77709e4c5827ee0b8acf1e"}]},"branch":"refs/heads/unmaintained/zed"},"a5ec69ba1c0a92cc9b91b52b4101b81a330ea684":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-07-09 11:36:01.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/10/923310/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/10/923310/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/10/923310/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/10/923310/2"}}},"commit":{"parents":[{"commit":"b830abacdc0f9d2db9100b05981efb786c6cfe6e","subject":"Remove py38-fips jobs because stream8 is dead","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b830abacdc0f9d2db9100b05981efb786c6cfe6e"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-06-26 15:41:02.000000000","tz":-420},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-07-09 11:33:12.000000000","tz":120},"subject":"Add safety check and detection support to FI tool","message":"Add safety check and detection support to FI tool\n\nThis adds a safety check and detection mechanism to the\ntools/test_format_inspector.py utility for verifying those features\noutside of glance.\n\nChange-Id: I447e7e51315472f8fa6013d4c4852f54c1e0c43d\n(cherry picked from commit b27040b87e43ff4acfb6870ef0d13d54e5ca5caa)\n(cherry picked from commit b394ef00c3426771092d099cf1d96077bfa4b919)\n(cherry picked from commit aa12d39b453068d9ee8367591eb60d4a15cddece)\n(cherry picked from commit a1c94b47d02a7bd3ac076ea02a23e456da2d5a62)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5ec69ba1c0a92cc9b91b52b4101b81a330ea684"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5ec69ba1c0a92cc9b91b52b4101b81a330ea684"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
