)]}'
{"id":"openstack%2Fglance~923339","triplet_id":"openstack%2Fglance~stable%2F2024.1~I712b304ebe168705ec2793626f5d64135ad6d458","project":"openstack/glance","branch":"stable/2024.1","topic":"fix/disk-formats-tempes-stable/2024.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-03 11:00:46.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I712b304ebe168705ec2793626f5d64135ad6d458","subject":"Fix import job to provide valid disk-formats list to tempest","status":"MERGED","created":"2024-07-03 01:46:32.000000000","updated":"2024-07-18 12:44:42.000000000","submitted":"2024-07-03 20:13:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923339-fix/disk-formats-tempes-stable/2024.1","cherry_pick_of_change":923321,"cherry_pick_of_patch_set":3,"meta_rev_id":"2130cb60fb49543d2d7121889b2aefb761c9f4fa","_number":923339,"virtual_id_number":923339,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 20:13:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 01:51:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-03 01:50:25.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 01:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 01:50:25.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-07-03 01:51:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-03 04:05:05.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"8f271f08a38e93d5233db1d0289b17a50f875124","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:46:32.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"17b3da6a39c28adbefcf1555efa6b487aa23a28e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-03 01:50:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"efe5f0ae998a7192bcd1657117f2252b95e5f2cf","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 01:51:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/63e5439bc35a400a9dbef585de2ef0e0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74dc5cb252d4468d9a2cba0dd4743806 : FAILURE in 3m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b3623abb574945179402856e4be54ee0 : FAILURE in 3m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"adb26ecad251fb8dac4809a4e783fc452719264e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:51:25.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cee86fba3d1f9f8b151750ba1712dfb007c16acc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 04:05:05.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/662300d006a94dc1b544ff91c0be1f54\n\n- grenade https://zuul.opendev.org/t/openstack/build/33773fd3911649c98e16af255d6ec4f4 : SUCCESS in 1h 14m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4b6bdd37850f425da0b4e5d7768a236f : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48eeaf6541524cf999b0c5746268220b : SUCCESS in 1h 22m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ac0e677b7e634eaabb76a99a4e1f2264 : SUCCESS in 1h 07m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adb5e5305a5a4a4e8b45ab8aa0912d97 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/42bac3cc4fa6483b9247f75703bf3b34 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/27f3e89cea494a98bb8d8eb9f2a138af : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ff4aee3d4940b0890835256c40e332 : SUCCESS in 4m 54s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b6729b83565044b681611d1926c3b9a5 : RETRY_LIMIT in 2m 37s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0e6056768c344c883f3d9335ed94028 : SUCCESS in 8m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13d4464445174ce8bc4da23a6a8c73b4 : SUCCESS in 9m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ff68bd686dc54f9281438042d63f0ead : SUCCESS in 9m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/268e5ba40ba84e42932bd18cfb4f5a58 : SUCCESS in 1h 57m 00s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/34a79da06cdf4f2b965584d810e9a886 : SUCCESS in 1h 41m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/14ab21d377324d8eac517defcce7088a : SUCCESS in 1h 48m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ef90df04095c46fa9ababd45794630d2 : SUCCESS in 1h 38m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9dd95529c60c4737977f72abfd816652 : SUCCESS in 2h 09m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2e9a8be370f143f8aaa857921c456a41 : SUCCESS in 1h 17m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b5b4fd5333c142719ac141d85406cb84 : SUCCESS in 1h 48m 30s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/69c3058ce2d24b46a18b1b1399a88f77 : SUCCESS in 1h 52m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f4dd509e600c4e3a95e7ee0b73bfd17c : SUCCESS in 1h 04m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0ef74aaaf9984c3d92a67ecc8c3978be : SUCCESS in 39m 39s","accounts_in_message":[],"_revision_number":1},{"id":"286eaa6a54f7c025a48b1267c8d885938f4aa005","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 04:06:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1157c0c1e4004f045a2408cba0bd30692c62dcd6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:00:46.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed8a97e717a64554972b62df324148ca\n\n- grenade https://zuul.opendev.org/t/openstack/build/c874c75733e14280bc0700f9da98ed83 : SUCCESS in 1h 03m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/ec4c0b534523462d9269c8eda57e0823 : SUCCESS in 1h 08m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/92eed9ae2cb54bb7871e76560f5af7f1 : FAILURE in 1h 21m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54f4027ec96f412ba2e9791d694db1c4 : SUCCESS in 1h 01m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa2f84660f454fdab22a3a98d93603cf : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4a9505d91e8a4d9685cd45b7efd87ab3 : SUCCESS in 6m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/12054142a4564c1c9246076e0a9de8e0 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05ee229bdde7498a86ac40f280db4acc : SUCCESS in 6m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a10a006e760e4f9b8319bb0fbcfbdfbf : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9bb26d101fbf4fc4aea8228ca523a223 : SUCCESS in 9m 11s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fb0e9942b35649f6aeae638d4648e96e : SUCCESS in 1h 27m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ce665c64fe2144aba36b1c7261dc640e : SUCCESS in 1h 40m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d4254cf2c91c4bdfa9bc9728f19ce691 : TIMED_OUT in 2h 04m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/98b13cd2bf3b48ab99188616820548b4 : SUCCESS in 52m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3b395c508e294b32b84f2cad22f2ad44 : SUCCESS in 2h 02m 23s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/48093400238e4a02839068ece67bf989 : SUCCESS in 24m 47s","accounts_in_message":[],"_revision_number":1},{"id":"e7623b833dfcc299594682f54a3a6596620fd782","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 08:04:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"75bffd06cd7eeb54c22fd540d3f2eb9243c1e2f7","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 08:18:43.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f8c08bc290b40ee8d7ed265e566a69f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c0e8bce5503c42dbbb9d9d25bbed6ba5 : SUCCESS in 11m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ff5107ea98fa41d79785412aeae769ea : SUCCESS in 11m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"096a99345c96c53ab9e2eaaf7d377fa6eddf7c49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 10:21:35.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/934ada74a770443d89b1b226544f26cc\n\n- grenade https://zuul.opendev.org/t/openstack/build/96858de795394b9c8b1d718075f14707 : SUCCESS in 1h 03m 21s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/76f3873305a4422692dda31db36f5549 : SUCCESS in 1h 02m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8c0a8312bb834246b043714cc3d750ce : SUCCESS in 1h 24m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d660639c2744736a59912855f9ceaf0 : SUCCESS in 1h 02m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73204fd94dc24fcaa3432df05e7ee63d : SUCCESS in 6m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b76b51687a19498bb43a2413c253079e : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e8426388e91a4e0bbf65332fe35913d8 : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eba64f5413aa452283a0244398d57b2c : SUCCESS in 6m 22s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/b0ea5cd661544adb99df993b12a201f7 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b08b8cdcf3994c3c87faba7ae508e00a : SUCCESS in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c502285c9b474258bddbf43e10db4abf : SUCCESS in 10m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fd5477b67bc7481181b3e596b4d762cd : SUCCESS in 9m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee988e4241624f938fc271d71d626ca9 : TIMED_OUT in 2h 03m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c8d03746346047909c018bf12f3f2de9 : SUCCESS in 1h 41m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/15c3edcb39854d2c8a07ac7c6525a874 : SUCCESS in 1h 37m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/450cd7ceba3d41b5bf6a474bc96f0993 : TIMED_OUT in 2h 06m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/dbb3590797de49e79ccb91994fd984c6 : SUCCESS in 2h 10m 18s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fe530288c2b844519b5a100df4040053 : SUCCESS in 1h 07m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5c5566eeb0c945fc850a453c8989f712 : SUCCESS in 2h 03m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a784ad12789e4249b4a0a6409b7a065e : TIMED_OUT in 2h 06m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1d0bc4b2b567492187b017075087146c : SUCCESS in 1h 03m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/52dc32b581db43c09787aa13fbf0f30d : SUCCESS in 34m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a4f9893e65f130c6f11b5c3ad6422d24ad96e2e0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 11:00:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1a6aea82a27b2730f7ca94264e8e1f19c2fb411","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 11:05:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd318ea6aca24beb86fc40dc5004d81a\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/8fa9d61d07894cd6b39e30c0f011d870 : FAILURE in 3m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/63a8a5a2334d4a7a9dfbd6f67ea028df : FAILURE in 3m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c8acecc51ca46cc2a6c8c1b6bfb04112d86a35a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 13:11:28.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/237c87e015a3440c8c104bc6efab8e16\n\n- grenade https://zuul.opendev.org/t/openstack/build/6556907aa1f9462c93cf646f6fd8c7c6 : SUCCESS in 1h 04m 17s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/80ef35d17c1c452d8e138e8c1c76af29 : SUCCESS in 1h 04m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4b14e1913a074ceab1e5eae431b3a75f : SUCCESS in 1h 36m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bdad02090dc24361921f81dde0ce62d9 : SUCCESS in 1h 07m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d0caeb048b804fe79bca77e6ecf6956d : SUCCESS in 6m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30a3e313e86c428cace55569d7e518a4 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b79e949d77e46b8b00dd555eba4bf78 : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfedbbe61b2439b8f45dd2caf33daba : SUCCESS in 6m 55s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/ec7d87ffcba147339397b204421fb439 : RETRY_LIMIT in 3m 00s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c5718885053464bb3073cf6018c7c49 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6d845ea3b7049e2995e88d11c1bcff3 : SUCCESS in 9m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7c075c3f550a48f797da92738890799b : SUCCESS in 10m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/950ca9239780487684e41dd4c6b34e44 : SUCCESS in 2h 04m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b71839d2278849beb2e39f17ffba0a77 : SUCCESS in 1h 46m 07s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b202b12d287f4c458cbab2c5169a6fb5 : SUCCESS in 2h 01m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1cae22235e8e49cf87d8d8ca79cfb063 : SUCCESS in 1h 56m 02s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f75cc74e14c84024bd5b2b1b9c34d6d3 : SUCCESS in 2h 03m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccc287a416c64bd381a2e5bee5a84689 : SUCCESS in 1h 04m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6e7352591e2f443dbbe9c4e9fa5fabb0 : SUCCESS in 1h 57m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a347720cb8804d228efd4e9aa6fe6622 : SUCCESS in 1h 58m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4e3772251afd4b6f8a4545f1303d5ffe : SUCCESS in 1h 02m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e2703f08620f45fc83a07f483cc8c255 : SUCCESS in 41m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cc5c5c60c504f39cbe7a898b5613138ed7b15b70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 13:12:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c6db742688de60b9ec4ca18ff78d741a2527ba0b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:13:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0edc9f7777b84b4fa9765efea1f0cbd4\n\n- grenade https://zuul.opendev.org/t/openstack/build/06d719f693214f608730a85f7027ec01 : SUCCESS in 1h 08m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3825bf8daf144e48870e49fc7abbd463 : SUCCESS in 1h 05m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1fac5dd4038e46edb916a5c46d6aa607 : SUCCESS in 1h 26m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4d189079138447539e5e72a5b35104de : SUCCESS in 57m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b96b8d8b8614951942d2d6312f23aa8 : SUCCESS in 4m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fbdbe09abdca4337879edb01b2c01ab7 : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cae7dd433c7a41ba8724e6b1eee0cb1c : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0138885c48747ce9284c1746280a325 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4d26e03f79164e07ab9068746faff951 : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/235eba2f86a646028b48d681cdd5c67f : SUCCESS in 7m 59s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ee16586c60a64efb998075298ad4d448 : SUCCESS in 1h 44m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/72c19702491d4a0ea599eea03f1b4e29 : SUCCESS in 1h 30m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/46a3de72af86417cb9eb95f0df2e6411 : SUCCESS in 1h 29m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3053b1d232cf44c49c8a6fdb1e945a40 : SUCCESS in 55m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3fe366ca05d84e689d3c90e0599694f3 : SUCCESS in 1h 56m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/891539922450421b90609e9aa441887a : SUCCESS in 33m 10s","accounts_in_message":[],"_revision_number":1},{"id":"58fb7df676fd6c81b7964faa12e0c010a7669e61","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:13:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6e3658e920f507bc4a548dae44e729a3119c475a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 20:16:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70a9f1e953994de78843376e1a70ebd8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2881cb1985d3490f9cb3f1917b3c93b2 : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1},{"id":"2130cb60fb49543d2d7121889b2aefb761c9f4fa","tag":"autogenerated:gerrit:revert","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-18 12:44:42.000000000","message":"Created a revert of this change as Icdae35c78e85bd9eedfe9130a8742323c58d16a1","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"400345bf88ce507b5401eb706923752b846bbc86","revisions":{"400345bf88ce507b5401eb706923752b846bbc86":{"kind":"REWORK","_number":1,"created":"2024-07-03 01:46:32.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/39/923339/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/39/923339/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/923339/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/923339/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/39/923339/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/39/923339/1"}}},"commit":{"parents":[{"commit":"eb3eb4ea269f608a7b534bb205f917309420cd51","subject":"Remove SQLAlchemy tips jobs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eb3eb4ea269f608a7b534bb205f917309420cd51"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-02 17:39:32.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 01:46:32.000000000","tz":0},"subject":"Fix import job to provide valid disk-formats list to tempest","message":"Fix import job to provide valid disk-formats list to tempest\n\nRecent changes does not allow image conversion to convert an image\nhaving AMI disk format for image. Passing list of disk-formats to\neach import job which will now use qcow2 as default disk-format to\ncreate the image.\n\nChange-Id: I712b304ebe168705ec2793626f5d64135ad6d458\n(cherry picked from commit 244ccb68dd9c1be8b8c6223c50998464a394b688)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/400345bf88ce507b5401eb706923752b846bbc86"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/400345bf88ce507b5401eb706923752b846bbc86"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
