)]}'
{"id":"openstack%2Fglance~923341","triplet_id":"openstack%2Fglance~stable%2F2023.1~I712b304ebe168705ec2793626f5d64135ad6d458","project":"openstack/glance","branch":"stable/2023.1","topic":"fix/disk-formats-tempes-stable/2024.1-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-03 03:02:51.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I712b304ebe168705ec2793626f5d64135ad6d458","subject":"Fix import job to provide valid disk-formats list to tempest","status":"MERGED","created":"2024-07-03 01:50:21.000000000","updated":"2024-07-03 07:35:04.000000000","submitted":"2024-07-03 07:31:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923341-fix/disk-formats-tempes-stable/2024.1-stable/2023.2-stable/2023.1","cherry_pick_of_change":923340,"cherry_pick_of_patch_set":1,"meta_rev_id":"1ea47378be8a3a795c576771e37b14be98c2fe2e","_number":923341,"virtual_id_number":923341,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-03 07:31:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-03 03:02:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-03 01:51:12.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-03 03:02:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-03 01:51:12.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-07-03 01:54:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-03 03:47:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"efe03e7d0217c04b4a27c263b63c02795abaf54a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 01:50:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"779d6433dbb5cc01f57130293c65e8aa8dfcfe9c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-07-03 01:51:12.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"01a0a1f20f6279d11b495e298e62d989b03816c1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 01:54:24.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18325eb974084122a23b466a170d8d8f\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/aebf3158e3094654b80f61122d7c27b5 : FAILURE in 2m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f623fdc07365f5a04061b30d0c025404151934c7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-03 03:02:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5b30220c6d4fdf6c1c7b1f562ee07e2edb66be1f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 03:47:19.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20424cd5dbcd43afbf0b548af607f357\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7996e8fca0cb459eb2085285c6d0b881 : SUCCESS in 57m 10s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d4f41f108f164d97af9cef0095460bd5 : SUCCESS in 57m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce618ade3fc845c88bc3751173f4f484 : SUCCESS in 4m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7fa62ebeba6d44b0b4aeade8fc2336db : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5444ffb4d4b4bb79d80483f42fcc124 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/147edadf01c74641a48aceb0f4e157f9 : SUCCESS in 7m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8c5138ccb95d4379855a158be5fc5f1d : SUCCESS in 9m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6599562eae55453fbb72caf059cf6f7e : SUCCESS in 1h 49m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee8c35e61b4c44e3b4466876938378db : SUCCESS in 1h 31m 15s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/848aa8220f934c1aa5f3645802831361 : SUCCESS in 1h 41m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/00d3922d890d456cb5c29434f701ef5a : SUCCESS in 29m 39s","accounts_in_message":[],"_revision_number":1},{"id":"62dbd4da4ece0fd9e98c76cea1885745d884bb1e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 03:48:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"00bad3187779c7ed79ed3e77e091f6755f9d934c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:31:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/483c4efb2540425ba8a9c53d61362b7e\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7ca5064ac54849a5a439720395a1f398 : SUCCESS in 58m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ade634977869471ea0161bba91586d22 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c99901c7fead49808b9e98d46ae0d978 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b67952eef95b44a0a898ada67cda9248 : SUCCESS in 7m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ff83a7558af4ffca2e3d1af2bbfb3b2 : SUCCESS in 9m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/064f0ef66ed848819c161f52e3dfb8de : SUCCESS in 9m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e5d47c28758c4270acb6d86eb9788169 : SUCCESS in 1h 31m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bd8078aa4e874baa81f6d6dca8081cda : SUCCESS in 1h 31m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/278ff377bf9b4d8abd994e60008355ee : SUCCESS in 29m 53s","accounts_in_message":[],"_revision_number":1},{"id":"c52ee51df5874b229ecfd1b10e2ad147fc27aae9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:31:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1ea47378be8a3a795c576771e37b14be98c2fe2e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-03 07:35:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2aeb7cfbc9c41e28a4ba84e4ca3c404\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4837bb2c00f41ae8961d502cf92064a : SUCCESS in 57s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f98e9ae4a80ab206a8214572df22ac6134908d14","revisions":{"f98e9ae4a80ab206a8214572df22ac6134908d14":{"kind":"REWORK","_number":1,"created":"2024-07-03 01:50:21.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/41/923341/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/923341/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923341/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923341/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923341/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/923341/1"}}},"commit":{"parents":[{"commit":"ef22a7425a5fe4f7006f9434a33dd092e2594830","subject":"Support Stream Optimized VMDKs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ef22a7425a5fe4f7006f9434a33dd092e2594830"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-02 17:39:32.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-03 01:50:21.000000000","tz":0},"subject":"Fix import job to provide valid disk-formats list to tempest","message":"Fix import job to provide valid disk-formats list to tempest\n\nRecent changes does not allow image conversion to convert an image\nhaving AMI disk format for image. Passing list of disk-formats to\neach import job which will now use qcow2 as default disk-format to\ncreate the image.\n\nChange-Id: I712b304ebe168705ec2793626f5d64135ad6d458\n(cherry picked from commit 244ccb68dd9c1be8b8c6223c50998464a394b688)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f98e9ae4a80ab206a8214572df22ac6134908d14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f98e9ae4a80ab206a8214572df22ac6134908d14"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
