)]}'
{"id":"openstack%2Fglance~923471","triplet_id":"openstack%2Fglance~stable%2F2023.1~I03336ae0bbe191d23626abd5208f50fa5adc9407","project":"openstack/glance","branch":"stable/2023.1","topic":"temo-non-voting","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-04 17:31:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I03336ae0bbe191d23626abd5208f50fa5adc9407","subject":"[stable only] Make import jobs non-voting","status":"MERGED","created":"2024-07-04 08:05:27.000000000","updated":"2024-07-05 12:02:18.000000000","submitted":"2024-07-04 17:31:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923471-temo-non-voting","meta_rev_id":"aed9e2e80c27a87be9b149e580512773e14e62d5","_number":923471,"virtual_id_number":923471,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 17:31:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":2,"date":"2024-07-04 12:01:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":1,"date":"2024-07-04 12:01:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 08:09:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-04 10:04:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 12:01:44.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"208b6eca260d5a18c6f1a9b0ac228e8d8d9bafc7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 08:05:27.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23cfc78db14814e5d1bf536b04d74dd209547a62","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 08:09:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d51c6708718f4c76b38e411d94ef87da\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3ce1d810da9943e5a8e3bdf7f3af1bde : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"25f29fc1dd00f0f2908aeffd0ceb5724f5587f03","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:04:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/753db8cddd5f4d06b22016164f29ca7c\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/cd7e9bc453ea43809193b2eaa239520a : SUCCESS in 1h 05m 46s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7a56cd19b64f4453a58eb3dac78baafd : SUCCESS in 1h 14m 44s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/312c9f6679554f258648e53bedf20573 : SUCCESS in 58m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5952a55ec0024a1c992e684551ea0004 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2b56b59297284d24a4e2fee45214e491 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/704f6ecd76de42f28fcebaeca879971d : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/18dac43477d9439db817bbf3e389fbb7 : SUCCESS in 9m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9bfed1a32f7d448aa4db3df5ca488429 : SUCCESS in 8m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f5bafbae67144119bb503166fee6a536 : SUCCESS in 1h 17m 16s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/89492bdd08ec4d9e988166c7b47fa51b : SUCCESS in 1h 42m 00s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c36be67726f44663836d4f543d1359ca : SUCCESS in 1h 36m 49s (non-voting)\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3815818668944b569548f241fab02b9b : SUCCESS in 1h 52m 25s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31dde243d21946b7a2b795cf63e20aa5 : SUCCESS in 1h 47m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b837e3bfb127468a84681acbf84f97fd : SUCCESS in 31m 46s","accounts_in_message":[],"_revision_number":1},{"id":"6b11afccfa52821a4ef747d87d3c5b4bdc594deb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-04 12:01:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"ea9f0e74d7f28eaee9ab00b3eb55466adb45d9fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:13:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"e7e8b46f88fbd1c592b395f411f6fce92de7ed91","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:31:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2da5def0327e46c8a8a4d2b08679cb0a\n\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/67e8e5ed575342cb9659b037aeb620c6 : SUCCESS in 1h 13m 43s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a814a03e8c794f07a81ed5318c92b84c : SUCCESS in 57m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4de2d299bd2424bacb16aeefd10bcfd : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2fefef3c95884a758b0c3eb04e6b928d : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32ce3f4135ed4a0abf761ec049bcd01f : SUCCESS in 6m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3aa551fd47b44facb4d425eaafcccdb3 : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30f900c1cb4b469fbc88d966f0d32d3f : SUCCESS in 8m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e4d3a46cf5ec4e428bef8829a26d2e41 : SUCCESS in 1h 09m 48s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e385fcb8bccb4433942e91c93e3634c6 : SUCCESS in 1h 38m 56s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/060f30e2a69e44cdab2bd102ee838626 : SUCCESS in 1h 46m 12s (non-voting)\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/542d25da9d0b4e08904da5c0751ffb9c : TIMED_OUT in 2h 33m 10s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a99ea0968c54a65b82c565c7f3d4aec : SUCCESS in 37m 35s","accounts_in_message":[],"_revision_number":1},{"id":"7d8fefdf498f6b000d33665939d3f36590b50b96","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:31:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d12993c4b39f7f77100cbb30f8df3c158c9cc44","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 17:34:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15f7663a43e746ee9071a84b49cb52be\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a5c636fa7c141e2b1047a0af587e3eb : SUCCESS in 46s","accounts_in_message":[],"_revision_number":1},{"id":"aed9e2e80c27a87be9b149e580512773e14e62d5","tag":"autogenerated:gerrit:revert","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-05 12:02:18.000000000","message":"Created a revert of this change as I7c4101b2cf588faf2343c65044dfd50889c7a2fe","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f9ccc2d866c080ebff55022e8512d72f5363b611","revisions":{"f9ccc2d866c080ebff55022e8512d72f5363b611":{"kind":"REWORK","_number":1,"created":"2024-07-04 08:05:27.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/71/923471/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/923471/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/923471/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/923471/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/923471/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/923471/1"}}},"commit":{"parents":[{"commit":"f98e9ae4a80ab206a8214572df22ac6134908d14","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f98e9ae4a80ab206a8214572df22ac6134908d14"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-04 08:04:15.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-04 08:05:24.000000000","tz":0},"subject":"[stable only] Make import jobs non-voting","message":"[stable only] Make import jobs non-voting\n\nRevert this once CVE fixes merged.\n\nChange-Id: I03336ae0bbe191d23626abd5208f50fa5adc9407\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9ccc2d866c080ebff55022e8512d72f5363b611"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f9ccc2d866c080ebff55022e8512d72f5363b611"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
