)]}'
{"id":"openstack%2Fglance~923483","triplet_id":"openstack%2Fglance~stable%2F2024.1~I7165bb8c588659362b7b36394111c41772ac802d","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-04 19:16:31.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7165bb8c588659362b7b36394111c41772ac802d","subject":"Increase timeout for tempest-integrated-storage-import job","status":"MERGED","created":"2024-07-04 10:08:46.000000000","updated":"2024-07-04 19:20:40.000000000","submitted":"2024-07-04 19:16:31.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923483","cherry_pick_of_change":923431,"cherry_pick_of_patch_set":2,"meta_rev_id":"f0451e2d729df75d186305dbbf0aceda7e7e3d5a","_number":923483,"virtual_id_number":923483,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 19:16:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 10:11:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 10:11:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 10:11:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-04 10:23:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-04 13:41:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"6205802b53c34ccaf25d55ba51bf2d73430da53f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 10:08:46.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"5874fa7e5dc6fa79421f6ecc18b548db5704cbc8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 10:11:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2a12c183164a1f173608f5bce65fb8234819f512","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:23:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c2d7b4789a14a63857d2acd3be574a7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/0579799da1624a9a925c8d60ec91bb9d : SUCCESS in 12m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/435831bd248e4965afdfc1f561e525c7 : SUCCESS in 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"545620fcc1fdb53aab8ba00fe1013d7b76044813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:41:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b043bb7886164a90bbbaa3bcfe391a1d\n\n- grenade https://zuul.opendev.org/t/openstack/build/3c23643b7ef840ab90e35eab5a619192 : SUCCESS in 1h 03m 45s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9a219fc5b32e4c88b61e4345b6274322 : SUCCESS in 58m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a78003e515df41d9bf5c4986083337a3 : SUCCESS in 1h 26m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3c58228dc1e4837b305a0fc87b2b4e8 : SUCCESS in 1h 08m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a624680cd8934c7d8a7bb54df58fcdc4 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d3dc3b54495945d7938c7fd52b5d62c5 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/03224a54e39a434e983582994f755756 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc4975d68c6c4d50a044dbf49e60e9d2 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/8ced73464e3648b696db7e7a50d8adf8 : RETRY_LIMIT in 2m 55s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d02b9c052b14e63896be59d5d52b06f : SUCCESS in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cd87a2fd32d74e939c679190a822ca53 : SUCCESS in 7m 23s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/09fb6822a4c746bdb04a04bd0b4b9af3 : SUCCESS in 7m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/51db5cd7f5834c64a47ff78c55d30252 : TIMED_OUT in 2h 03m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ebdabe4f8da44139653341fd93a1a73 : SUCCESS in 2h 01m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cc8c558fca2b429d965bcef6278efeef : SUCCESS in 1h 50m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e3cacfc6feb451b869db34c119b290b : SUCCESS in 1h 46m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ed22f0ce33ac45ba8161634fd75e700c : SUCCESS in 1h 55m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ee233ad5bcda41ff8936f185dec5b799 : SUCCESS in 55m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92c440d235b742b1ab44b6905e97ac1d : SUCCESS in 1h 37m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/901067f9c1844f469ca4b5299882ef21 : SUCCESS in 1h 43m 28s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/192ae2869ea4484aa3bc25e59bfab94a : SUCCESS in 1h 08m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2bb4d3c1c93c4d64ae998986fdc65b30 : SUCCESS in 33m 15s","accounts_in_message":[],"_revision_number":1},{"id":"b26c177d8978483ec5b1269db8adfa5c24de6f87","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:42:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4325c1f28a930bead4134f5f68a171a3b2cffa39","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:16:31.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e69db8ab6cfc480f96201b5eab696e89\n\n- grenade https://zuul.opendev.org/t/openstack/build/51de2a2c621a471896c96fb984dec913 : SUCCESS in 1h 06m 31s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/241b6f5daf2c49fbb32bcb1475198e98 : SUCCESS in 1h 03m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8f245a5e4ab74179864ab7b3d5d5579b : SUCCESS in 1h 29m 33s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e138a0e5732d46658d0b646c5b2b1911 : SUCCESS in 1h 04m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff04a19f9e044dba9dae5b466849526f : SUCCESS in 5m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6c5f8336d09e45d5bec8f2209738d11b : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0e03bfbf9a2f45589bc5d952121fc403 : SUCCESS in 5m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c3a55f6f99da48dab9f81ef37990d185 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f38dc1b37ed4455bab4540c168598e45 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4eaf1087d1a347f0b95d5b978115753a : SUCCESS in 8m 08s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a4aea3c88bc049f3b755a350f01f4b32 : SUCCESS in 1h 54m 55s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3835b9fa8ee64f76ab1f11474b93c470 : SUCCESS in 1h 52m 19s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d979f2b2f4e42478b615e46de001653 : SUCCESS in 1h 51m 02s (non-voting)\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8865ea65d93d476bb04d79e72e5f8954 : SUCCESS in 55m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ce3d3022743444698819807a0fd6668e : TIMED_OUT in 2h 34m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd24b4c49f4b407b84e5eb9614df6cc8 : SUCCESS in 34m 38s","accounts_in_message":[],"_revision_number":1},{"id":"99c9308dc98acde2095bcf045c01cd3def42c34a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:16:31.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f0451e2d729df75d186305dbbf0aceda7e7e3d5a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4caa8a4a3a684251a051b7698c9d4101\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ddbc2531f224a6ebc3389037b1319c2 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"35968d7a7a898663fe542429a1e1bf8f14a92099","revisions":{"35968d7a7a898663fe542429a1e1bf8f14a92099":{"kind":"REWORK","_number":1,"created":"2024-07-04 10:08:46.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/83/923483/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/83/923483/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923483/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923483/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/83/923483/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/83/923483/1"}}},"commit":{"parents":[{"commit":"400345bf88ce507b5401eb706923752b846bbc86","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/400345bf88ce507b5401eb706923752b846bbc86"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-03 17:08:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-04 10:08:46.000000000","tz":0},"subject":"Increase timeout for tempest-integrated-storage-import job","message":"Increase timeout for tempest-integrated-storage-import job\n\nThe tempest-integrated-storage-import has been timing out\nfrequently around 2 hour 5 minutes mark and similarly one\nof the jobs inheriting from it,\ntempest-integrated-storage-import-standalone, also times\nout around the same time. The timeout issue has increased\nrecently as can be seen in the job build history[1].\n\nWe had a similar issue with the glance-multistore-cinder-import\njob for which we increased the timeout to 3 hours[2] but the\njob completes successfully on average 2 hour 10 minutes\nand has been completing in that time for more than 8 months\nsince the change to increase timout landed.\n\nThe underlying cause could be due to a lot of reasons:\n1. Lower concurrency (3) as compared to parent job (6)\n2. More number of tests being added to tempest\n3. Slow Devstack installation\n4. Slow tempest test execution\n5. memory consumption of OpenStack services\nincreased making environment slow\n6. no significant improvement in the hardware\nspecifications on which we run the job\n\nThis patch generalizes the timeout to the tempest-integrated-storage-import\njob so all of it\u0027s child jobs also don\u0027t face similar timeouts.\n\nAlso we are updating timeout for tempest-integrated-storage-enforce-scope-new-defaults\njob as we see similar time out issues in that job.\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dtempest-integrated-storage-import\u0026skip\u003d0\n[2] https://github.com/openstack/glance/commit/64e6da6fe1c810befab1e5a59230ce8711dce709\n\nChange-Id: I7165bb8c588659362b7b36394111c41772ac802d\n(cherry picked from commit 9df48a3071dd6fff11bf4e5523e2185e23d6a2e3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35968d7a7a898663fe542429a1e1bf8f14a92099"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35968d7a7a898663fe542429a1e1bf8f14a92099"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
