)]}'
{"id":"openstack%2Fglance~923486","triplet_id":"openstack%2Fglance~stable%2F2023.1~I7165bb8c588659362b7b36394111c41772ac802d","project":"openstack/glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-04 19:16:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I7165bb8c588659362b7b36394111c41772ac802d","subject":"Increase timeout for tempest-integrated-storage-import job","status":"MERGED","created":"2024-07-04 10:12:21.000000000","updated":"2024-07-04 19:20:39.000000000","submitted":"2024-07-04 19:16:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923486","cherry_pick_of_change":923484,"cherry_pick_of_patch_set":1,"meta_rev_id":"92be94e4db9cc06c6248466b3a3c7697f3bfa22b","_number":923486,"virtual_id_number":923486,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-04 19:16:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-04 13:45:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-04 13:45:56.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-04 10:16:58.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-04 12:16:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-04 13:45:56.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"1f0fbfc7a5dccf2ed1a2da6bab4fb437ad06e203","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-04 10:12:21.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"c9c6ae64756a1c1a22d2bcb3f71422818bd27f9a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 10:16:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b034d516079f48978c074f6544032b3b\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/35cdd08fcc4947fc8080b21198db7b5a : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0b665ee911c7ba12f2ef950d9a9e8c26a452d338","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 12:16:04.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02eaab3fec8048fcaf6b928cb98ed7bf\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/9b2caf1796ff4b068c3b8008dfb2c2e4 : SUCCESS in 53m 16s (non-voting)\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/630b7285e7d6480aa4f27fbfb6a595f2 : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ffb90c4092794c778609dabfa1cf2859 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f614a39f806a403a88ccf0f4991baf06 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc265a1e0ceb4b81a519509db667fdb9 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02244ba497dc4193917e7705158be199 : SUCCESS in 8m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6ab21f09270a4d55a4dea1b1c02b2a51 : SUCCESS in 8m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc3ad99ebcc7454199db99babe6c523b : SUCCESS in 1h 23m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/77cd0e84388f48a4bb5aab088d447302 : SUCCESS in 1h 43m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9e4b218d936444d6a2d6161216d43ba4 : SUCCESS in 1h 38m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/171f6cfa9cd841ca8be91b7677069d2d : SUCCESS in 1h 42m 48s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/4d8b919631fb4038b3a5c227c2226432 : SUCCESS in 41m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a55c1523c08a3e3ad1d3047eec167343c6311565","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-04 13:45:56.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5c083156ecd65ce8472c82327e5c32bf5079e49f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 13:47:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its\nconfiguration in the repo openstack/nova on branch unmaintained/zed.  The\nproblem was:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the \"irrelevant-files\" attribute\nof the following project stanza:\n\n  ...\n              irrelevant-files: *nova-base-irrelevant-files\n          - nova-ceph-multistore:\n              irrelevant-files: *nova-base-irrelevant-files\n          - neutron-linuxbridge-tempest:\n              irrelevant-files:\n                # NOTE(mriedem): This job has its own irrelevant-files section\n                # so that we only run it on changes to networking and libvirt/vif\n                # code; we don\u0027t need to run this on all changes.\n                - ^(?!nova/network/.*)(?!nova/virt/libvirt/vif.py).*$\n  ...\n\n  in \"openstack/nova/.zuul.yaml@unmaintained/zed\", line 677","accounts_in_message":[],"_revision_number":1},{"id":"38f2507ebbd9ea393c6401473fa9d9a40f97bbe1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:16:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed4ac326ee8b4cafb411e6ab91b36728\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3f92786c65154d5d8da4eee1ebe17f1a : SUCCESS in 1h 03m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbda15d4a7ca410ab02564356afbc5f3 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d5097f3a6554c24b193990c384037a4 : SUCCESS in 6m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75b9123af7824c7a92f3a7bbb02ca9f6 : SUCCESS in 8m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c4ee13e96a4247e7a8aa5b72b4074d42 : SUCCESS in 9m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9db80713d8045d9b3a48ad3fbe38fc9 : SUCCESS in 8m 04s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/54ed3978a74341278490e9811dc2520d : SUCCESS in 1h 48m 53s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4e0f9c1d78794922b63645f09b086aa0 : SUCCESS in 1h 22m 29s (non-voting)\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ed42e1eb5c724e8689eb1d93ebaba220 : SUCCESS in 1h 50m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/33b63af3f8a34b5393423689ed50d56a : SUCCESS in 34m 08s","accounts_in_message":[],"_revision_number":1},{"id":"28ec1512918c93be3a82a1646389ec66e8b00647","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:16:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"92be94e4db9cc06c6248466b3a3c7697f3bfa22b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-04 19:20:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f0e43db35b6e448e8a8671b4a4fd8683\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89a9cf9aea934f8b98ca0f38eeab71e6 : SUCCESS in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6918d5b577c636b974c28db34640117a6d4e7fd8","revisions":{"6918d5b577c636b974c28db34640117a6d4e7fd8":{"kind":"REWORK","_number":1,"created":"2024-07-04 10:12:21.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/86/923486/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/923486/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/923486/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/923486/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/923486/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/923486/1"}}},"commit":{"parents":[{"commit":"f98e9ae4a80ab206a8214572df22ac6134908d14","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f98e9ae4a80ab206a8214572df22ac6134908d14"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-03 17:08:03.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-04 10:12:21.000000000","tz":0},"subject":"Increase timeout for tempest-integrated-storage-import job","message":"Increase timeout for tempest-integrated-storage-import job\n\nThe tempest-integrated-storage-import has been timing out\nfrequently around 2 hour 5 minutes mark and similarly one\nof the jobs inheriting from it,\ntempest-integrated-storage-import-standalone, also times\nout around the same time. The timeout issue has increased\nrecently as can be seen in the job build history[1].\n\nWe had a similar issue with the glance-multistore-cinder-import\njob for which we increased the timeout to 3 hours[2] but the\njob completes successfully on average 2 hour 10 minutes\nand has been completing in that time for more than 8 months\nsince the change to increase timout landed.\n\nThe underlying cause could be due to a lot of reasons:\n1. Lower concurrency (3) as compared to parent job (6)\n2. More number of tests being added to tempest\n3. Slow Devstack installation\n4. Slow tempest test execution\n5. memory consumption of OpenStack services\nincreased making environment slow\n6. no significant improvement in the hardware\nspecifications on which we run the job\n\nThis patch generalizes the timeout to the tempest-integrated-storage-import\njob so all of it\u0027s child jobs also don\u0027t face similar timeouts.\n\nAlso we are updating timeout for tempest-integrated-storage-enforce-scope-new-defaults\njob as we see similar time out issues in that job.\n\n[1] https://zuul.opendev.org/t/openstack/builds?job_name\u003dtempest-integrated-storage-import\u0026skip\u003d0\n[2] https://github.com/openstack/glance/commit/64e6da6fe1c810befab1e5a59230ce8711dce709\n\nChange-Id: I7165bb8c588659362b7b36394111c41772ac802d\n(cherry picked from commit 9df48a3071dd6fff11bf4e5523e2185e23d6a2e3)\n(cherry picked from commit 35968d7a7a898663fe542429a1e1bf8f14a92099)\n(cherry picked from commit 212426faf32074337095715b0068d39a14b4d626)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6918d5b577c636b974c28db34640117a6d4e7fd8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6918d5b577c636b974c28db34640117a6d4e7fd8"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
