)]}'
{"id":"openstack%2Fglance~923541","triplet_id":"openstack%2Fglance~stable%2F2024.1~I6c141ce126014660b056d7d226baffb326c87a22","project":"openstack/glance","branch":"stable/2024.1","topic":"temp-non-voting","attention_set":{},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 04:07:55.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-05 08:26:35.000000000","reason":"Change was submitted"},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-05 08:26:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6c141ce126014660b056d7d226baffb326c87a22","subject":"Revert \"[stable only] Make import jobs non-voting\"","status":"MERGED","created":"2024-07-05 04:06:34.000000000","updated":"2024-07-05 08:29:42.000000000","submitted":"2024-07-05 08:26:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":923468,"submission_id":"923541-temp-non-voting","meta_rev_id":"e0e7546b457a54d6b273e16b64fe1fedbe6e3f67","_number":923541,"virtual_id_number":923541,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-05 08:26:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-05 04:07:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-05 04:07:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-05 04:06:34.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-05 04:06:34.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-07-05 04:06:34.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ac70e530bde65d70917dd0b7d183849142c1c884","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-05 04:06:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"aaf0be829974b5b17c57c94058cfccdfb577e07a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-05 04:07:55.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5d06e3b05f3bd2a1f253707549fa55659774d8b1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 04:16:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5caf8c850bbf47388c3a047a795ebd04\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f7e45bc74fbb4b76afc3e8b0e47b2ed2 : SUCCESS in 8m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8187e809f10240459dda0d97beed11c2 : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"824f54ecece1c6c5aa4c5f91d02cb4c10426b4f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:16:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/630abad4c79f4dec865da3c6c0fc236a\n\n- grenade https://zuul.opendev.org/t/openstack/build/ec0ab44718ac409080bd503e20e614f6 : SUCCESS in 1h 05m 36s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0a8f0713914c446780a25aa51b1f3c1a : SUCCESS in 1h 01m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b638db8cd32d40d182b3613f35f00cf2 : SUCCESS in 1h 35m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a4eddbef1f046e1b72e46d5b6ee9e63 : SUCCESS in 1h 00m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97c29a288e0f4164bfe8ebd31d1c3e29 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfe6a5db03a04e0c82cdcb65b435a157 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2910bf6bc3a84d8da78a2f833f5ed9e3 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3dcca6a4c35b458199b31f82131aaf10 : SUCCESS in 6m 20s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/338e97961dc14fd8b315fb63d117a095 : RETRY_LIMIT in 1m 35s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8bd013778a624ab3af9d6bf74e21dbe5 : SUCCESS in 9m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d78c1ed633fe40fd8314d12109e4a578 : SUCCESS in 7m 08s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/80091d96873a4a2598082646fac2d2d2 : SUCCESS in 9m 36s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a3e445ddd48d4648b7df13d8bdee79c0 : TIMED_OUT in 2h 03m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f626b9c25b63425c8100d8e613d035d9 : SUCCESS in 1h 50m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/52d040b6f6624f7fad34a811e55d74dd : SUCCESS in 1h 39m 11s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2b809f1cd0d748dca8b525c67963d5aa : SUCCESS in 1h 49m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/954f5df36f904f398512ff192b81437f : SUCCESS in 1h 50m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c703a7fdecc3459a9c95b39e6ebb3d2d : SUCCESS in 51m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d6203667133d48c1babcab9325c6accb : SUCCESS in 1h 46m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/d22d1f7e36b9478ba114f06711bb98a2 : SUCCESS in 1h 35m 53s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9f4236a824914e72886cc7c6f3002f51 : SUCCESS in 1h 00m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/38cca237e2404bef8086ea00cf2cea59 : SUCCESS in 37m 00s","accounts_in_message":[],"_revision_number":1},{"id":"f0b8aee758f8e86ed3afe028b0423b182e1da8a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 06:17:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"56fa8aa38a6e256dbae91df2d7081ac95638203d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:26:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/517806edea404bd899bbb064b32ca2c8\n\n- grenade https://zuul.opendev.org/t/openstack/build/f13d67d1132049bab29ad3ae7302a1e7 : SUCCESS in 1h 05m 06s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/34386e3398234173b2b6436b719f15e1 : SUCCESS in 1h 03m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/041cc0a2e7724d42acc785bec5f4ab19 : SUCCESS in 1h 30m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eff0efc0dc1d49ccadd0982e04b6d098 : SUCCESS in 59m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/275391cfc22b46928a793ffe3c614d12 : SUCCESS in 4m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbc987a850424bcc9df3d2b1b8cc4d65 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c727895da180471b80cd3e55a6119966 : SUCCESS in 6m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4812b7bb4ab6424498051511a6361520 : SUCCESS in 4m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7bb666abc56b4cdd847c0ffd169fb562 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e0ee78cc5bb417cb2363723eff7eb73 : SUCCESS in 7m 10s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/72e5682cca1f4047918185f61e5f5049 : SUCCESS in 1h 52m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/91a8bba1abdb465194ad432f3873b56b : SUCCESS in 1h 50m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/9834c72ad5dc42a5aab44cc9c79fb4a1 : SUCCESS in 1h 47m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/95270df1461e4e338720600b97d1daab : SUCCESS in 49m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d844aa352f5c49969ac3ae7435b52572 : SUCCESS in 2h 01m 21s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e0c1d7bf03934fc9a8aaea34f3524a72 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":1},{"id":"351ac05b27171bd1ab4ce0bd38b53cdb20edbdc3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:26:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e0e7546b457a54d6b273e16b64fe1fedbe6e3f67","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 08:29:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/836af418d90f47249134a8b3bd6caa30\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9d8681a54bd14401b5f255551eb5d11d : SUCCESS in 1m 02s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4168ee5afe88df453c2db92f1afec6d7b54dd849","revisions":{"4168ee5afe88df453c2db92f1afec6d7b54dd849":{"kind":"REWORK","_number":1,"created":"2024-07-05 04:06:34.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/41/923541/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/41/923541/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923541/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923541/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/41/923541/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/41/923541/1"}}},"commit":{"parents":[{"commit":"bc996f20e07e5d022c0dd356ca47121cc38126cf","subject":"[stable only] Make import jobs non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc996f20e07e5d022c0dd356ca47121cc38126cf"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-05 04:06:34.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-07-05 04:06:34.000000000","tz":0},"subject":"Revert \"[stable only] Make import jobs non-voting\"","message":"Revert \"[stable only] Make import jobs non-voting\"\n\nThis reverts commit bc996f20e07e5d022c0dd356ca47121cc38126cf.\n\nReason for revert: Make all jobs voting again as all the required CVE patches have merged.\n\nChange-Id: I6c141ce126014660b056d7d226baffb326c87a22\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4168ee5afe88df453c2db92f1afec6d7b54dd849"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4168ee5afe88df453c2db92f1afec6d7b54dd849"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
