)]}'
{"id":"openstack%2Fglance~923575","triplet_id":"openstack%2Fglance~stable%2F2023.2~I3259dd013ba5e3fefd0e172bf0e7cc502158c8db","project":"openstack/glance","branch":"stable/2023.2","topic":"bug/2059809-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-08 10:34:17.000000000","reason":"Change was submitted"},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-05 14:42:33.000000000","reason":"Change was abandoned"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-05 14:42:33.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I3259dd013ba5e3fefd0e172bf0e7cc502158c8db","subject":"Add releasenote for CVE-2024-32498 fix","status":"MERGED","created":"2024-07-05 14:04:07.000000000","updated":"2024-07-08 10:35:43.000000000","submitted":"2024-07-08 10:34:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923575-bug/2059809-stable/2024.1-stable/2023.2","cherry_pick_of_change":923574,"cherry_pick_of_patch_set":3,"meta_rev_id":"b45a90a11b7dfc1d4ff39e5bb19c1304bc4ac558","_number":923575,"virtual_id_number":923575,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-08 10:34:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-08 06:49:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-08 06:49:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-05 14:18:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-05 14:23:32.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-05 14:23:32.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"bfecc8739159ea670b2feded8441cfe09fbda42d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-05 14:04:07.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"227d69bd0077f24a6329e8cf5c8e931697a26c04","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-05 14:18:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/343853e3d91e458da51f5b850640d25c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f8e6095f01b48e8842d5c5271c7e7b9 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1def691eec4f0caedd9318b9325c5b : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4928bd891f734aa18bfbee528ced65ab : SUCCESS in 9m 25s","accounts_in_message":[],"_revision_number":1},{"id":"f035b6e220e6e87840bb27534bfe86734fa50f1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-05 14:42:33.000000000","message":"Abandoned\n\nsince missing cherry picked from commit","accounts_in_message":[],"_revision_number":1},{"id":"8f2214c32dbae223f3a37ade1587bfbbf693db80","tag":"autogenerated:gerrit:restore","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 05:18:26.000000000","message":"Restored","accounts_in_message":[],"_revision_number":1},{"id":"8bde9bc0cfd5ec94acf8c7613f7015d4578f8f8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 05:30:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed111765f476472291e1e95d858cdfe1\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4f10660f7dd49949a6d0a8a0c8ddc04 : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b2cb8603f7aa45c7a3f13443fe60c659 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b8de25d4aa64636a15aecad1cd7ed6a : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":1},{"id":"9407f131aec0c8422f6f49319a2fae95bc91df64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 06:48:16.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"71e8e7cd4bc721b04cfce984af0a4d5abaec639a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 06:49:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"6c868f04959074ec84dc6e5f05f2cf29bc4fdc09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 07:26:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/83be0686a72149068f75e5ce91a081de\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3114351e6b25465d8bbd7895639e361b : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c5b996d0b4f468285deebae0ccf834a : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/54e0e7d0068147f197f87fba7593ca7f : SUCCESS in 6m 40s","accounts_in_message":[],"_revision_number":2},{"id":"50ccbde9bc1ac8301a67290e7f3983d9980c7b4f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 07:27:01.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"b24afff969c6b2fdd9423d273b23ac8e9414c720","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 10:34:17.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6aad93f681424769bca9a7d69d104ae7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/617d69546d774b3f8887ae8c5d70fba2 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a725a40447924d0c916001f7c52b3b87 : SUCCESS in 6m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18b32e8708d14d6c8f6a4ca661828418 : SUCCESS in 6m 15s","accounts_in_message":[],"_revision_number":2},{"id":"fbb2bb6c0e0223b898662b3a2284d34def2b8dcc","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 10:34:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"b45a90a11b7dfc1d4ff39e5bb19c1304bc4ac558","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 10:35:43.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38f3056da06a4b6ebac9ff19f45aa7d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/08407be232b14844bdfcd67a4c219054 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/67164843bad442b5a93ccc963da8de8d : SUCCESS in 50s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"835c89c711993c51aa8752d96b4c8effa1a97009","revisions":{"8b53db3e7f853f65fa1b1eb15fa92669f0ff0493":{"kind":"REWORK","_number":1,"created":"2024-07-05 14:04:07.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/75/923575/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/923575/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/923575/1"}}},"commit":{"parents":[{"commit":"3d0fcae49f5fcb9b5c9c342103bf492368046f88","subject":"Merge \"Add file format detection to format_inspector\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3d0fcae49f5fcb9b5c9c342103bf492368046f88"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-04 09:59:18.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-05 14:04:07.000000000","tz":0},"subject":"Add releasenote for CVE-2024-32498 fix","message":"Add releasenote for CVE-2024-32498 fix\n\nRelated-Bug: #2059809\nChange-Id: I3259dd013ba5e3fefd0e172bf0e7cc502158c8db\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b53db3e7f853f65fa1b1eb15fa92669f0ff0493"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8b53db3e7f853f65fa1b1eb15fa92669f0ff0493"}]},"branch":"refs/heads/stable/2023.2"},"835c89c711993c51aa8752d96b4c8effa1a97009":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":2,"created":"2024-07-08 06:48:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/75/923575/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/75/923575/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/75/923575/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/75/923575/2"}}},"commit":{"parents":[{"commit":"1a3debae8f651612d09a95faf2cf7654dfd2a067","subject":"Merge \"Revert \"[stable only] Make import jobs non-voting\"\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1a3debae8f651612d09a95faf2cf7654dfd2a067"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-04 09:59:18.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 06:48:16.000000000","tz":0},"subject":"Add releasenote for CVE-2024-32498 fix","message":"Add releasenote for CVE-2024-32498 fix\n\nRelated-Bug: #2059809\nChange-Id: I3259dd013ba5e3fefd0e172bf0e7cc502158c8db\n(cherry picked from commit 867d1dd8b6e4f5774257a98c7c33061fbbbde973)\n(cherry picked from commit b5b29a0ae15d44424f04db0686328d0a01a14853)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/835c89c711993c51aa8752d96b4c8effa1a97009"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/835c89c711993c51aa8752d96b4c8effa1a97009"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
