)]}'
{"id":"openstack%2Fglance~923599","triplet_id":"openstack%2Fglance~master~Ie419f2cf9f685bba770b6b2c88571e41758cb212","project":"openstack/glance","branch":"master","topic":"drop-fips","attention_set":{},"removed_from_attention_set":{"13252":{"account":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"last_update":"2024-07-18 08:21:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ie419f2cf9f685bba770b6b2c88571e41758cb212","subject":"zuul: Drop openstack-tox-functional-py38-fips job","status":"MERGED","created":"2024-07-06 17:37:04.000000000","updated":"2024-07-18 08:24:12.000000000","submitted":"2024-07-18 08:21:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923599-drop-fips","meta_rev_id":"4eb5029eb2c4bcba3acf7236791272398ae9f7e6","_number":923599,"virtual_id_number":923599,"owner":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-18 08:21:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-07-08 11:10:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-07-18 06:20:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-07-18 06:20:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-06 17:43:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-06 19:46:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-08 11:10:47.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-07-18 06:20:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"ac978573b9e114f79f422ca820d86df5f6b499ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-06 17:37:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6e2f396182b342f8de1c5681118e7f7da76cf4da","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 17:43:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ad1f0271200d41d8a96663c1833a41a0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ffb5047196644c7f88f5ab16bce6753d : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f35970ed642e4304883258cba5e8770b : FAILURE in 3m 34s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c7ebe9dbe23813845fd8d410b3db1f086a4fa77a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-06 19:46:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b7d6011123f8452681b670d2022aed29\n\n- grenade https://zuul.opendev.org/t/openstack/build/3d375424d8974418979430ebcdf5907c : SUCCESS in 1h 02m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c7236b4c165f4f76a3272ab5522c5ea7 : SUCCESS in 1h 23m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aee0955639e644bb9c58bee478ae8560 : SUCCESS in 50m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea5f2e5f4ce4bb6814957594645ddd9 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a1666da6d07340d9a74dc1b8f824d233 : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9ab2624bfe704e94861d8fd7a5ddf4a5 : SUCCESS in 5m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/88b1e0d8517e4fcb983ca0724962f890 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85fff90cb07249c29dbdeb2a2e47d75e : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae8b90bbebb74c20844751fbd51f42ef : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b504d103253441bcafb34458bfdc70fd : SUCCESS in 8m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ea9a3bc50a4a45bbb39d3d2225d54c1f : SUCCESS in 12m 28s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ca3a564417284c11b5111cc5c28c5c75 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1458e46b5a89415eb69d5a3ce53718a8 : SUCCESS in 1h 58m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ba246fa2ebf7482cb219d31ac6d302b7 : SUCCESS in 1h 54m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/66b98085ca704804a4ee4858c525d2aa : SUCCESS in 2h 01m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a41c5bc4d36f48e18dfac30573a169bf : SUCCESS in 2h 03m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/847cce942b5048a4888be149d7d5238c : FAILURE in 1h 55m 05s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/22f96a89b7904babb03937e17c930c02 : SUCCESS in 1h 01m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ae5394b50444bb59f8f8732800808e5 : SUCCESS in 1h 58m 49s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/04851da8cfb64c6aa8d58b6f591c43fa : SUCCESS in 2h 00m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b6ed500aad35408991702553650c219c : SUCCESS in 1h 08m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/678b274c82dd4c19a6c4170a8cff58aa : SUCCESS in 34m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a1c989aa9929058635d32fac35c706bc503d54d0","author":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"date":"2024-07-08 05:07:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9a7b6d9043bf0f37faf1f0f04fd2f8594c937e23","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 05:11:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6bccd35c952f4ed4b56ebaab90210688\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cb5bcb0cfade4ae28636e6d717418c04 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/b5ad92ebe7b84806969ef83acb6041a8 : FAILURE in 2m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"701ab09c5963da4d93c28adb3b9151252adbf3aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 07:25:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2adf51f29a7d4058b4927e57ad600b0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/613368715b1444e5bb06d072be522fd5 : SUCCESS in 1h 01m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e1ae791f7954c2da8baf4647617f52a : SUCCESS in 1h 37m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0875268f11f44a92b7556c3239962372 : SUCCESS in 47m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3070107828e4152b4e5471d48d07384 : SUCCESS in 4m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a24bae57db6d457586e745db61f8ceee : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dbe7f345bf2342c89fe2062f498f7eb6 : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5d6cdd8c9a904f6195fc6f6786e742bd : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0fcf9cfe5da04b56832561fc5b0f3492 : SUCCESS in 5m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d98ec61d04054a759a1f6e9b87b41007 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/18d42e85acbc494d8b9798f1770fa66c : SUCCESS in 8m 05s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/067b56e6225f4d1d82f56334ec46b752 : SUCCESS in 12m 54s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2c71b7ca3d1d47cc94719a2b45294289 : SUCCESS in 8m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/887f7b8aa48845ec9e84f110df094aa0 : SUCCESS in 1h 42m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7b00a1d48734bb995cea63394d29e13 : SUCCESS in 2h 00m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4b9681f95ae1473597a3c2d05e252134 : SUCCESS in 1h 54m 13s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ce8fcdb43408491fa2866161dd609599 : SUCCESS in 2h 08m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d4cfc6a3c8ba4b1fbbfdccbc59abb4e5 : SUCCESS in 2h 16m 36s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8974ce289f374d9ba0945a52c293d0ed : SUCCESS in 58m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/1d2cfd07d00f4f05a7b3d64b5d870d9f : SUCCESS in 1h 39m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0365e16fd37c4735a20a38bedf430139 : SUCCESS in 2h 06m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/011f64d37c474f368338cbd3ac45c8cb : SUCCESS in 50m 37s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef0157f5fe264959972a8c8d37bae2ff : SUCCESS in 32m 30s","accounts_in_message":[],"_revision_number":1},{"id":"aefff237da6fc85173cedf2a4289fc0ef5572b89","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-08 11:10:47.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"225b6840ba2bade94865ea6922455c53b1857deb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-18 06:20:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02ac6507078a496d3a9abe43f43cebacfc74ada1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 06:20:54.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"044c79292e5159d8a34d2911af6ccaa9c3eda7fe","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 08:21:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04cc4928dfc9470c8c1235d09883cba2\n\n- grenade https://zuul.opendev.org/t/openstack/build/9b4102c33778434595ce7f1a77319fd0 : SUCCESS in 57m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/10b60d766d784a2c8ef15e7fd9c31e23 : SUCCESS in 1h 06m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7342ee66a10d4a15b1666c3ce9ae3f87 : SUCCESS in 58m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e8581ff5ee7846f1b454b306e2fb54f9 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7c7e011b5f3f4d4ca319be1d110c2441 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbb4c167a6a4440ea8761d975c0d289d : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90411952fe374af18f997fc14fbf8eb4 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/58f0f16c33704ae79653f40eb1667c3a : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7158fe63eecd46c8a3da371708aad9d9 : SUCCESS in 8m 36s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c32245493dd8450a9e31e274d5c6e022 : SUCCESS in 10m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/eed75680543c4b37a418a7c52bb8c128 : SUCCESS in 1h 38m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d6ded8d6b46b4338953c17389b1e79b5 : SUCCESS in 1h 55m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/34289fb7ecbd406ea628222615dc3cb0 : SUCCESS in 1h 51m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e3ba7b3cbd4444f4babdb0fe0e1bb7fa : SUCCESS in 1h 00m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c81e7aa8fc654dcaac1dd75539dd7fc9 : SUCCESS in 1h 41m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/99a450f810cf4620afe7142ad0e18322 : SUCCESS in 29m 02s","accounts_in_message":[],"_revision_number":1},{"id":"d02b3910581cce7a63b723af2f3be4012fc2c86c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 08:21:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"4eb5029eb2c4bcba3acf7236791272398ae9f7e6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 08:24:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db47a5dafa5948ab9f70a69a333b193c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9b814a3528f34e26917dd60030cf4bea : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"726b186e34a59988f83cef23f64d4b1d71dd2387","revisions":{"726b186e34a59988f83cef23f64d4b1d71dd2387":{"kind":"REWORK","_number":1,"created":"2024-07-06 17:37:04.000000000","uploader":{"_account_id":13252,"name":"Dr. Jens Harbott","display_name":"Jens Harbott (frickler)","email":"frickler@offenerstapel.de","username":"jrosenboom"},"ref":"refs/changes/99/923599/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/99/923599/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/923599/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/923599/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/99/923599/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/99/923599/1"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-06 17:36:01.000000000","tz":120},"committer":{"name":"Dr. Jens Harbott","email":"harbott@osism.tech","date":"2024-07-06 17:36:01.000000000","tz":120},"subject":"zuul: Drop openstack-tox-functional-py38-fips job","message":"zuul: Drop openstack-tox-functional-py38-fips job\n\nThe openstack-tox-functional-py38-fips job is based on centos-8-stream,\nwhich is no longer working and will soon be dropped completely, so\nremove the job.\n\nChange-Id: Ie419f2cf9f685bba770b6b2c88571e41758cb212\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/726b186e34a59988f83cef23f64d4b1d71dd2387"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/726b186e34a59988f83cef23f64d4b1d71dd2387"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
