)]}'
{"id":"openstack%2Fglance~923676","triplet_id":"openstack%2Fglance~master~Id706480e31687d8ade6f7199b600aff3ad7c68f7","project":"openstack/glance","branch":"master","topic":"format-inspector","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-08-28 23:23:37.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-08-28 23:23:37.000000000","reason":"Change was submitted"},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-08-19 14:19:15.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-08-28 19:03:20.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"Id706480e31687d8ade6f7199b600aff3ad7c68f7","subject":"Add iso file format inspector","status":"MERGED","created":"2024-07-08 19:46:31.000000000","updated":"2024-08-28 23:25:39.000000000","submitted":"2024-08-28 23:23:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":66,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923676-format-inspector","meta_rev_id":"7c8adabe7499faddf23fe7866b4ff8b757de11cb","_number":923676,"virtual_id_number":923676,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-28 23:23:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-08-19 14:19:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-08-28 19:03:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-08-28 19:03:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-19 14:17:04.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-08 19:50:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-08 21:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-12 05:51:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-07-12 05:51:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-12 05:51:21.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"f187b2df593142bc9e8d89ad2992a36f06c64d1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 19:46:31.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5b7d25c9c504070f20a0ccb00b78805eb41ba1c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-08 19:46:49.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"c330ce8d29e3718d61692be6339844593e1fabe3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 19:50:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81c8ac979c5948fc9a4a9fb382a1fcc1\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/fcd97b163bce4c66958a84f84c5772f2 : FAILURE in 3m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f51642e22da94259a638e163de250bbd : FAILURE in 3m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"745fbe8303d8852183beca7db1f0c93b5887e579","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-08 21:56:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9b9c62314189433e9c2408dc313ec7c5\n\n- grenade https://zuul.opendev.org/t/openstack/build/9e72f66686224fc1ac2fcb406dcd36a7 : SUCCESS in 49m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/99fb66fa90904e08b64dbf6a3c6e8fa3 : SUCCESS in 1h 27m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/01c8193b3e6145d9aa71100e7e98101e : SUCCESS in 1h 00m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19289dae8cf494a9b49c9fa7a8133e8 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77ba6e9a968745bb8a58ae233535999f : FAILURE in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b4ce570d2f134ee48059c50b9e475aa5 : FAILURE in 6m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2593929639704b2ebd8781a72fa5cfdf : FAILURE in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/68524de41b724f168889261915345465 : SUCCESS in 5m 45s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/4cd1269d1de448e293e6b718b9d3d3d7 : RETRY_LIMIT in 1m 50s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ff4bb6f95cc441e698050f15ff3dce4a : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aeabe2572f664a9088ff1209c2a3b6ec : SUCCESS in 9m 34s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5d8d7c98264641a8b6c71791b05269fe : SUCCESS in 12m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c411d059133f4edb9cc7788ac0a87d72 : SUCCESS in 9m 05s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7b544279837244dfb4ae0ded2c748a5d : SUCCESS in 1h 53m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fa6a7ea36055475a8ee58adb8ff6a831 : SUCCESS in 2h 01m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6899d2cbe62e4c889f49e436582d6fb4 : SUCCESS in 1h 41m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3b65527046da46a1815d781368bca5ad : SUCCESS in 1h 40m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f8f46f3210004e3d896e7942fc3652e5 : SUCCESS in 2h 04m 09s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/885460c698a04561ad28623913536c6e : SUCCESS in 59m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/732283c41c874fce87857b1ef71d52e5 : SUCCESS in 1h 48m 28s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/470294e0a55f4ff3a146ecea2852619c : SUCCESS in 1h 39m 15s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ffad7c362dfb403b9c51e296d726ff44 : SUCCESS in 52m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e7f0bbba2fb54c04b0516f40893e933a : SUCCESS in 34m 16s","accounts_in_message":[],"_revision_number":1},{"id":"fbb93a381a788a64cd2e08dce25f5266b086586e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-09 10:21:34.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"83cc17cac2a9b29edde85900c9b254beea3ee6a0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-09 10:22:16.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2ff8d2d41d96fe95e2a463f6ce5658cc0adc4a8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 10:24:58.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c2705f51e3ce479b8ce3efa206387b99\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/5a1a461f00b04687926414bb62ddef63 : FAILURE in 2m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f96d3233f71475d94bc23cad9d7aac1 : FAILURE in 2m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"97a327a155e4d240808e06de250d20fb68193cff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-09 12:17:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b631fed22c974664992414a27b444479\n\n- grenade https://zuul.opendev.org/t/openstack/build/74293e63606f4e29b8d3eff42c36ff8e : SUCCESS in 1h 05m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6cc3bff0cb7e4ea9ad2a182466fc75b4 : SUCCESS in 1h 25m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c03dc4d8c184725b887cc0894320bc5 : SUCCESS in 55m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6900d55c3f4f4d6ea3ee410ed912013e : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1d6ef1d110f4b35ac844989480a0d95 : SUCCESS in 6m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/95edf289cadb45f09d34e9e56ac4916a : SUCCESS in 5m 21s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2b7d5e1b23954d4e8ecf8a02d7b2ba92 : SUCCESS in 9m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ac4ff5f5fc9643f9858ffc7db9d10f7d : SUCCESS in 6m 06s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/0f5a0970062243b1b1cac452a45365de : RETRY_LIMIT in 1m 48s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/43051e7421744f089a9c8e525fbd23e7 : SUCCESS in 8m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b13a075c363499ba76397d61b3111d1 : SUCCESS in 8m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3b37292ce70a420c9b040619d038cf8c : SUCCESS in 13m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/05cb9c02291345038ad8530c732bcbd2 : SUCCESS in 7m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7bed024beba64fd6a955af4c07956d83 : FAILURE in 1h 55m 02s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e8f227d49dab45679b323289514d677a : SUCCESS in 1h 43m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b585dbeb8c74d8192318cc1a887c667 : SUCCESS in 1h 36m 53s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b0ebddbcfd7f41ed94b9317df01c35f9 : SUCCESS in 1h 45m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8135852b4f994d84b867370904160605 : SUCCESS in 1h 48m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1836de4e93bd4e1d946f387b2ac894df : SUCCESS in 48m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21258a9231954e9bb561e1a650c2cdf1 : SUCCESS in 1h 48m 56s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/820dcd3ef95e4f0aa76d021b91db520c : SUCCESS in 1h 46m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5871a4c83d7c4740b79a44b409d8e13a : SUCCESS in 1h 09m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af1d73c82bd84bc1a19ccaf80ad494e6 : SUCCESS in 25m 41s","accounts_in_message":[],"_revision_number":2},{"id":"701c8b8a349474c6a2215eb828a76a8c9dd392e8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-10 14:50:41.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"76feff801db0c4d0a01c15440097d862f2c5355b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 15:04:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16a7668dc5604940a7b827b20df37158\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/372a64ac7ed94e1abea74ca10744f694 : FAILURE in 3m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d47a6af06675426a9e8c37a1047da254 : SUCCESS in 9m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e585abe596bd174e52402d012ceaa1cfef5fa99a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 16:55:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c4ace1f84444d5dbf1dc366f566892d\n\n- grenade https://zuul.opendev.org/t/openstack/build/497366a88fe74248846fe05d53f37815 : SUCCESS in 57m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84a7c5b17fd64c20a4588da0dd5d1dcd : SUCCESS in 1h 21m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c584cb51d1e495ab4808656959215be : SUCCESS in 43m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a36890a4b0eb43deb3816cafc9413cbc : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d16f4b7a99224684bb20aa0c9735d75c : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a10f4e7b957f4509b82efd41329c69b0 : SUCCESS in 4m 26s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b305ddf881ea408ea4cd16354f0e8d69 : SUCCESS in 11m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70ddf385761841c4a04a4ba85275ad79 : SUCCESS in 5m 25s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/1ed3bf11a6b3431c8e315ad61c8b38df : RETRY_LIMIT in 1m 57s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea207cd193844471a6276db1334dfda5 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/85ec84b0c65746c4a627efda7640a8f3 : SUCCESS in 10m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/693799316b1043278ad1aa6f3cf83cf2 : SUCCESS in 12m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ae6591a812846d3b92f04755e6c47b8 : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/293da0ee73264bd7a47b55adb3a0c72c : SUCCESS in 1h 33m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1b37d30d86a54d37b74da847f033ab04 : SUCCESS in 1h 35m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4507a839d5214101b39f84fdb5f3a562 : SUCCESS in 1h 35m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67ae97ac687145a69a1b8a45f78f43ec : SUCCESS in 1h 56m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4d9bcc1434964694829cfd6dcb07ca4a : SUCCESS in 2h 03m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/764d162d36734849a5604d3ca3c85fc7 : SUCCESS in 1h 00m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0f60210a31a54ff68c812cb4e3444838 : SUCCESS in 1h 42m 19s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3fc2d0e3c1404387ae246933866b1181 : SUCCESS in 1h 40m 53s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/086b3f55345d4638bcb79e1cf4b9315e : SUCCESS in 54m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0e0665c78d29409084f58dd178ce679a : SUCCESS in 23m 33s","accounts_in_message":[],"_revision_number":2},{"id":"4190a4a7b29cffe2bfd2e953828dcf903950fa8f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-12 05:49:31.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"1244a0fe6b8aa8b81e998bf19bb9678782e0e536","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-12 05:51:21.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"189a49b545a63a8feab842541e0316d5a5d09c42","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 06:03:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/728780ebcc9541e0991f73ec74268ce3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2ecc5d2e46a146d5adf71023c8914bd5 : FAILURE in 3m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a8921bc3dfb84b81b37f3f32c2b031df : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"c14c34adb8e2ab6ca466bcb80301531a304e3afb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 08:08:20.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d87f6b877834102a1ab7d85d99a01ff\n\n- grenade https://zuul.opendev.org/t/openstack/build/3fb5dcc658044fef94ed33c4950d0477 : SUCCESS in 49m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f81976cf95d4ffdba5f57b2b43b8b01 : SUCCESS in 1h 34m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8009b7e5aa8d4b59b8082c7df39837d3 : SUCCESS in 1h 07m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eeea7aaa50194a0ab02b8cff128920ad : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b6568bc20a6448a6a0821c6a13e6d144 : SUCCESS in 5m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d480c0f0b68d4c259a1a82aa426eb629 : SUCCESS in 7m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a96166c555574e6b9c478c873e8b980a : SUCCESS in 11m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76740edd87a7488caa6743213608467a : SUCCESS in 5m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9985de84bb134abba73cb3cf509fc9ec : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4c5a32f5cb0c4774bffe935cf1a91d1a : SUCCESS in 9m 25s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/70d227ce870c421a97a06a311d3a41ad : SUCCESS in 10m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/db4b4f895c7a4b10a1be3f80ce799948 : SUCCESS in 9m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee589d40de434aceb278eda7d175cffc : TIMED_OUT in 2h 03m 06s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/17f17dd932514cc6b9aaedc1a0c9fa00 : SUCCESS in 1h 38m 39s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d98cbe27fd844ad836b63394b3e80bb : SUCCESS in 1h 57m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/eaef4af04c4e485b8cdb093fe13ca119 : SUCCESS in 1h 58m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e9952882ec194bb081090c4bb5bcdd48 : SUCCESS in 2h 16m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8d2dc99ba38c49cf8a34284906fdb6b4 : SUCCESS in 39m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/27c10f9681254e0bb85508c61c5237f7 : SUCCESS in 1h 51m 32s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6e234dfaf9664c1da0eb87163d7b1a09 : SUCCESS in 1h 52m 46s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3d5507dcbe834750b4c68034440cba65 : SUCCESS in 45m 03s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bca5ad1200c342f6b91b8d00868d7021 : SUCCESS in 31m 41s","accounts_in_message":[],"_revision_number":3},{"id":"73b10567269fad6638d872c11fcfa6321ea281ed","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-15 14:28:20.000000000","message":"Patch Set 3: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"0180cecc5d9086c26a024822f4d475e9f7c951c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-15 18:12:38.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"a77d3a7754ba2c7ba9a8ce43e8e438665619e636","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-15 18:17:34.000000000","message":"Patch Set 3: -Code-Review\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"909ea1c141589d7adb78632c5d8a24382daed9ab","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 18:22:23.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/080f8c25c864493d81958ac6b966ba73\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9479bc2de10b4fff80f9e28cd565a94b : SUCCESS in 5m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9ca099182f50423f9b9ab39fdfb313f4 : SUCCESS in 8m 02s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"5bca26453c8491f284e1fb41c89884de4df61b22","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 20:22:02.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4a212c63883b4837863b05e210155ba3\n\n- grenade https://zuul.opendev.org/t/openstack/build/fa1a48a8a74d48499dd2955b7efae4e0 : SUCCESS in 1h 00m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07bad5d4eca447f48196df10e70ec2d0 : FAILURE in 1h 00m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7b8d2ee6a784b00b021409a7fc6e37f : SUCCESS in 45m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c3c096e895ac476eb1d89afc60059fc1 : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c43439a2d4c14799911398c2a8c17c44 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70ff0fe018ed46399d72b43cc1f40c0e : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e88c456f25c4f17b73ddfd57fa31c3e : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/433fd5d3db3e4aab8ca4ac43555a33a5 : SUCCESS in 5m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/24710aa7f8334d94910c5ef8b0718221 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05f02a2d71dd40d29271707bba4856bd : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/61aa157bb1194025a52e5a9085bd4fbc : SUCCESS in 12m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/36267c3cbcb64eb2be230a8f92156bdd : SUCCESS in 8m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f80bd079739d479c9e2bebebe7aab3b5 : SUCCESS in 1h 43m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/53bab8d9dfe24cd6b3775943e89a1b91 : SUCCESS in 1h 59m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/38916c3ba58b4ec5a20be69127825df6 : SUCCESS in 1h 55m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6e48411f66a0412182cbb8f8979c9a1b : SUCCESS in 1h 44m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a7ad0156c7314e06ae695e44efd1bb47 : SUCCESS in 1h 53m 22s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e2fb3e6a265942349869964506790558 : SUCCESS in 55m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/31dd2bfe88904e95bc59f031e53a173d : SUCCESS in 1h 35m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/35b22f6e974e4e89b5020e4851561669 : SUCCESS in 1h 31m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/87d449fd17fe4a3d9a644ecea3f186ff : SUCCESS in 1h 10m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c2193c78c8de4e0a9b0b79820a075c1c : SUCCESS in 36m 26s","accounts_in_message":[],"_revision_number":4},{"id":"a82082c20bb7ded3b6fb19a62bb730fff577e184","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-16 01:19:14.000000000","message":"Patch Set 4:\n\nrecheck","accounts_in_message":[],"_revision_number":4},{"id":"48048c0d9ec2a0e99e38a261d237af52b9b0d45f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 01:24:55.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e94008e37bfb40648c55a711a97a8d05\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/9fbde4a4e38b4f7594acc75220fc1a03 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3ad270bb80324c0a82e42fe71e0ed6a8 : SUCCESS in 4m 56s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a017b0daca908cc1cc2c3c3763cf4a06fdab7726","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-16 03:46:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0629526ae9d74901af841549bcf2160d\n\n- grenade https://zuul.opendev.org/t/openstack/build/b24e03bc18884fc4b26e60facfb37144 : SUCCESS in 53m 38s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9e4251afab445929a9caae80d7884bb : SUCCESS in 1h 09m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d9e91fddccf442cbe7469d7b8adac74 : SUCCESS in 49m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/989d0d83e6094363a4f903d95a7cca5f : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/853e28a7876442ed9805dd8d6cc65020 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7c60cca878384825a8377471cba3ca2d : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f991e30f42f4eb5ab224906ab44afab : SUCCESS in 8m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6ff86a6dc7d641f7b960d529c344d149 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e9050da59b64482c83f74e012219dabf : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28d46cce7f9f4ca0b2accc3cb1595dc0 : SUCCESS in 7m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d13da03d376b4083b7e2dd53472275eb : SUCCESS in 12m 15s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/62e917955ab741a5a92d81fa0bfcecee : SUCCESS in 8m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0a7a9b93072c48f9a2b5385ad6003e36 : SUCCESS in 1h 47m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/4ab6afeeee4d46c699d4d08cdf85ee3a : SUCCESS in 1h 30m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/092c9ba88d3242f18559fb5f3a6628ab : SUCCESS in 2h 18m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/54f0760a08f64ef2973fff1892c7a63c : SUCCESS in 1h 40m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c1c5b54a6a564028be39ef36971919ed : SUCCESS in 2h 01m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0dd3fbd214f440258297895bcd01e54f : SUCCESS in 56m 18s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/820ba0bd09264dc799c3c5ec6890a190 : SUCCESS in 1h 29m 20s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ea4120ad3a3e464d98bedf303b5017cb : SUCCESS in 1h 37m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5cdb4e7992c14f67808cd0233326b831 : SUCCESS in 1h 07m 02s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5450743f97384683a59db10235468288 : SUCCESS in 33m 37s","accounts_in_message":[],"_revision_number":4},{"id":"4c12dee30a30a36784544c46c70df6ce619d1592","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-16 05:55:38.000000000","message":"Patch Set 4: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6bc296b74f8d09d50a4a3beb6508bb431cc9ce0e","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-08 17:25:46.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bb584dfe478c822e0cac20f4f038850e10e5718d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-08 18:59:14.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"d0b2bf6c6b1c2a40f018a920a789a3c3e131d2b2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-08 19:01:20.000000000","message":"Patch Set 4: -Code-Review\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"6ce2af401acf505a87eeeb21cf7b15845e39981e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 20:10:32.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a664ec97b27f482ca31a859782c66e44\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/545f61ad0be34f75b2e9ccd2bb6cb377 : SUCCESS in 23m 52s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a1e7f2ca0d6e4684b9f1aa34fab7416e : SUCCESS in 15m 40s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"bd38ac87337ed85f2b8a33792c29b8d481c3ee15","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-08 21:42:03.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1adc7513ae144ef692089110d54bd0e0\n\n- grenade https://zuul.opendev.org/t/openstack/build/f73cdbe4fd324ef3b561e96d06825b64 : SUCCESS in 54m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2b54e632fc5144e1a15ccd91e8443f88 : FAILURE in 1h 00m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/43bce71cd06e40ab9bcbefdeadb82d86 : SUCCESS in 57m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f18b68095ff2498888bed9904b1d8035 : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f162e1659ab44b58a4da19a024cf8324 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18d9805ce4474ade82612ec71f03cecf : SUCCESS in 4m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/de3c27f4d4304d0e9462b57e348e6e6e : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f11fc5b3a6be4b51a0c4a3c018707201 : SUCCESS in 4m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd308d0e66cf4b2e9a7053ff4e71df6f : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b873f6f3ed6447debc1e488913738782 : SUCCESS in 7m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/edaee75acbf74f58a9564e7299963064 : SUCCESS in 13m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e75dc59f043743159701e6bd3c74f67c : SUCCESS in 9m 23s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fee9ccc0cc440b68d7b65ab42f5febf : SUCCESS in 1h 40m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/db188529cec24457ba0b051995b3def9 : SUCCESS in 2h 08m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/84947f76b05147459c97b7f07ea128e0 : SUCCESS in 1h 58m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5e6e4326f1d043ceab97f630d0d4023d : SUCCESS in 1h 58m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bbc2ee828d494d819af3db25b29e2138 : FAILURE in 2h 34m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f14ecac6ca6040098b503352aa7f25cf : SUCCESS in 1h 02m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f91ca9dfe83b47bba288220ca38bb44d : SUCCESS in 1h 39m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6be3e748c46b42b48056a677cfed0132 : SUCCESS in 1h 55m 16s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fee507d638724abd9226e96dda7f4395 : SUCCESS in 1h 05m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0adc9f59f189408ba21b605150841102 : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":5},{"id":"1b04e7510168216bf8e96adb095310206bb0d88b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 05:18:15.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"ac083f5086a43b29e91c18641e9946b292da2d8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 05:38:19.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e31339d0f814dcfbc600d3d104b0726\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f72be40f0c8c4d0da3a02a921269d38d : SUCCESS in 17m 49s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/683c20bacbf44b5fb972738e06a7aacd : SUCCESS in 14m 04s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"0a53777f2b8e5d2eead7b74c56cb3a3a87b53ce6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:29:37.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b8bc7f14ed7d48d7b2339473866c6ee5\n\n- grenade https://zuul.opendev.org/t/openstack/build/ea435591b7d04fcfad32ac53ce36ca49 : SUCCESS in 1h 02m 14s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/26fe4b115097430cbf3b64acdb8305f0 : SUCCESS in 1h 19m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a725b69392b3408cb8f7af4b2d0f43c6 : FAILURE in 59m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22a0998fbbb34c11872f8dd1ba2b038c : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/800c89e9121442d2a9890b94bc48485e : SUCCESS in 6m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b42f4499853d4400b747c55972688785 : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a05a182ce48543b4a30e34325c1c9ff5 : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f691c020b10d4f2eba9942798703dc8d : SUCCESS in 5m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dde85ea92b994376ae5aa7d7c321f44e : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52d17c4800cb4140ab8a984847992f03 : SUCCESS in 7m 51s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/97610d2ba3bd4caa953d178d90aa3f9e : SUCCESS in 13m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a4f3e3e82bdb4fc9abae4f63bd14426d : SUCCESS in 8m 55s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f112bdc82c3f4ce1a2734d79fcc08bbd : SUCCESS in 1h 40m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/393c2633420a4cafb2f5ffed84f2b4a6 : SUCCESS in 2h 10m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1277535ddf1348eb8ee41dc447434220 : SUCCESS in 1h 37m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e763bf2609434d39a4079da145913b95 : SUCCESS in 1h 41m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/62eedd0dd9474f7bb3c7039644b02ca1 : SUCCESS in 2h 05m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/165dbf8308d54bc79bf0d71d3dcb8634 : SUCCESS in 53m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07ea2aad6fcb492f80abde726d6e0201 : SUCCESS in 1h 53m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/733071339bdb4f2c9b2f0176c71173d1 : SUCCESS in 1h 47m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9528e0fd3dc3464da1e441712b0796c9 : SUCCESS in 1h 00m 30s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/20fc3a0c16764779b0be2bc8bff29f9b : SUCCESS in 22m 45s","accounts_in_message":[],"_revision_number":5},{"id":"2a58c893a4e9f9930bbb5221268e0f1ffcc9354f","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 07:38:40.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"19c31e62f76585ec1f9fd178609bfa82aeaf4165","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:58:44.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29bef600717422ca9ba503d605b6372\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/dc5556ab9afc4d32a0f8974ce99c759b : SUCCESS in 12m 00s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/6d29150e625a4b3da4879d63a20aa801 : SUCCESS in 8m 43s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c3f924bc61493b612367b4285dd1a43c4e644b13","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 09:56:16.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce3f5a075fa24ab6a0e4dca6e2ec8b9d\n\n- grenade https://zuul.opendev.org/t/openstack/build/fe1afc681fff4acd9a5f9b7ada232287 : SUCCESS in 1h 07m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de066238c0ec498ca1c5491e9098d289 : SUCCESS in 1h 16m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/02830c94783b4449a95c745b08062cff : SUCCESS in 59m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/743ab9dd2d514301bcf263dc3f8467d3 : SUCCESS in 4m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0c51a098c81c459a99d2b1ff0677cfec : SUCCESS in 5m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c519bec45ac945d6b72ebc82b138f9a2 : SUCCESS in 5m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e27dcdcbc0484c01a8c3f3ef61b81e83 : SUCCESS in 9m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/acfe47d4220743d39fdcfeb6f515395d : SUCCESS in 5m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7e4d8720981f43feba82d90e58c44b4f : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/90a3699fa59647638801f482c6003087 : SUCCESS in 9m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f0cb01c61c53432ea06a4bc77e99793d : SUCCESS in 12m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bdc3795b460e4417923cd4f5d69eb0c2 : SUCCESS in 10m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a939fc2751344ccca16cb471376f2f53 : SUCCESS in 1h 57m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/977ed9037137452e8f34e4d4e28fe070 : SUCCESS in 1h 38m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/33c5d2f414bc493c8585ea264d520ecf : SUCCESS in 2h 12m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c27f32a15441429d932ca77877de4e51 : SUCCESS in 2h 01m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5f6e2297135248d1b1c7e74804507ef1 : SUCCESS in 1h 48m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d439d3ce84a34223aedcca070d4bacfc : SUCCESS in 48m 15s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d2709ad19add47f1888458f372493b92 : SUCCESS in 1h 36m 53s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f250290cdbb14ce983387ee5aacbe2c8 : SUCCESS in 1h 57m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e750614a7cf542a0a71050b044ba2876 : SUCCESS in 1h 10m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/894f469dfc794494bdbd7dc19625df77 : SUCCESS in 31m 19s","accounts_in_message":[],"_revision_number":5},{"id":"9552b20156ffa4de5c0b862c067e7205d05e4fe5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-09 14:13:23.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"0df3161d297c3c713693ed6d6ac042888debd6e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 15:09:45.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"54f2e01b55a5d1c9b9fe39e6a30b2a43f8ce02f3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 15:10:34.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"06422ed2016f2348d772b7da6ca389c1273343da","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-09 16:33:29.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"ed7720580d60f92908abf1af94abf5dfdfa37e8e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 16:35:37.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77ef63c3461f4d90ac849d15535802a7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/495f4eb9c7ad4e93a8e6fede065df1f7 : SUCCESS in 15m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/034b3946c81e4f058e63b8af8e682efc : SUCCESS in 14m 12s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c90338af8837f3c54bf040796dec0ac066554e2e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 16:41:22.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"b3eb407ca1187704b9e70a5bb8d561c6967c90c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 17:15:17.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ff2e739f7524477d8b7de570919f1fc8\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0de51cca49e4638a0bee1ce0d5b0c95 : SUCCESS in 1h 11m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/36507ef3d2df456dbdd45a9b2d396a39 : SUCCESS in 1h 27m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1e51014cfce247e996341a5cdab78d12 : SUCCESS in 59m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1a135601627419e9c77ec5ac99dec58 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d79bba67be745bd836331a5d9981f9a : SUCCESS in 5m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd0aa3cbe0c04c0eb433b07d4bdb1f21 : SUCCESS in 5m 25s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a468b7ab39e54b838ad88335c0290b00 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b36df2c1be5643959d7e8d2c46494c7e : SUCCESS in 5m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efd4fcd297c64819903f16f4a1cce162 : SUCCESS in 8m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7ed21ed7ee204fc49b73cdfc3929fd5b : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5077fb9398a74eefbd0f56352fd9834e : SUCCESS in 12m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/7ed65177f7944a0bbb192e181bda7b9c : SUCCESS in 10m 34s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d53447e465234be699a5db72be4b593f : SUCCESS in 1h 28m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1062a48dc8934540bc42e880c8aba6ea : SUCCESS in 1h 53m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8591390937ce4dd4993b3abf22b34deb : SUCCESS in 1h 56m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/67a82ecb96d44338b1c3b8acc64f8a26 : SUCCESS in 1h 52m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8dc813d4ef5b41a494c07620a28050b5 : SUCCESS in 2h 04m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9c4cdc0014194180b7d937c0df58baff : SUCCESS in 40m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/13d473ad3a8443e4b3f12c9737c93262 : SUCCESS in 1h 45m 51s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/109f8c8c42c749bfa0f75efc11bef144 : SUCCESS in 1h 45m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/87ac8205cccb4aada78c4d6ff184676a : SUCCESS in 1h 11m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d21309ad41ad41e8a2754d469d5dd00b : SUCCESS in 24m 25s","accounts_in_message":[],"_revision_number":6},{"id":"9aeb56a0610fcee48fa9ecebca9bca6402e3f1b9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 18:35:00.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"556a0277db102291caae5cf43b9ccebe2e553c86","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 18:35:31.000000000","message":"Patch Set 7:\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"726d54c6af2c20f4a6e8478447f95896381155ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:05:02.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a500f022a5444f8db403c01275c0267b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/2accd686fba54a1fa7ec00f42a60e08d : SUCCESS in 10m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e033d3e8ca494c11993afd81eb7b5913 : SUCCESS in 9m 39s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"a4b8ff32011e0083d277a8f412d6bbcdf3bed130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:56:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec77beeff4014b408e09c8bd4f9c9e74\n\n- grenade https://zuul.opendev.org/t/openstack/build/790524b91e894d13b40d72d856d4ab41 : SUCCESS in 44m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ea87d5cb1bd944f28dd83f7432855dac : SUCCESS in 1h 30m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c4ad970cc7e14df680aa87a86b19ccfe : SUCCESS in 56m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa27ced3758746be9e17d235355f0878 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/343358b55f1241cabf7fe6ea4d09bd38 : SUCCESS in 7m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/073f0cf2319b46b4a8a2ee010d088b18 : SUCCESS in 5m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32c34800d6b341b9a55d9e747300e2fc : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba2b3fecbc0d4be6bbf34295b25d7ae8 : SUCCESS in 5m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0fc2045af56a472c9627e6ad31c9a3a7 : SUCCESS in 12m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c4270ca409124336ae7c285cf46fe93d : SUCCESS in 9m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/348c5da783dc468a88a2d3fac2eb7631 : SUCCESS in 12m 09s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6391f58673944ca7bc08c6702c267dd0 : SUCCESS in 9m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/39773ef8199c4f8eaf1b1af1a0cad803 : SUCCESS in 1h 41m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/fda19ebcc25f4d6a96108c770597bbf7 : SUCCESS in 1h 47m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fae925342d541248ac4eddd2d93efe3 : SUCCESS in 1h 27m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ea94c53b34b64ca8ab06ddd8dcf6ff2b : SUCCESS in 1h 44m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f7d1598d06224c1d96684507522b1e4d : SUCCESS in 2h 15m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aed987d0ccbf48f298fd45801c4ee1c7 : SUCCESS in 51m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/24f48598139d4c368635c19e5a9b3eac : SUCCESS in 1h 36m 39s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bfae696619644816a3bf0025016c4a34 : SUCCESS in 1h 44m 01s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/28e09ded59254a6fa97f645f5f144481 : SUCCESS in 1h 03m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dfcf59c828c1483fb142ad2c54558bfe : SUCCESS in 20m 12s","accounts_in_message":[],"_revision_number":7},{"id":"1065c985fef610d3a86ddd0fbac9fe768b3d78a5","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-12 14:05:49.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"8d7fdca2a91e14389bf53c371d66cde033df8390","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-12 14:33:15.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"6289d2674fea061c1bf66764b0c9bb79841c5ae2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-13 06:54:25.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"f87d28cb96529d72485bd955bd6a819d7c98d088","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 13:47:48.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"c04a60a7a45f66f9fa4737d6207bec9010868cca","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-13 16:45:21.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b530185211ce51dce31f9bf2be7e949e9b7de586","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 16:48:26.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"4e295c7751226df06403ce853c02f52c7914a93c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-13 17:01:16.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"4d8c8849dd801f03fa4968abf82e2e89e0d04b4c","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-13 17:02:54.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"14f013df060ec8aaab4dfb5d44fef5f3647f6311","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 17:19:34.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ddd8f163d496417b9a93130e848ea245\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f3ce68e61c2e42fca69a895aadd2b5f4 : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf4965a237094e78a72bd30cfec449e2 : SUCCESS in 15m 04s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"9cae3fe40792c9130ccb1548613f52cda4d88028","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:55:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4324c39ed02a478eabdae1d7f642672c\n\n- grenade https://zuul.opendev.org/t/openstack/build/89f28e51673f4873aad3471073af72c2 : SUCCESS in 55m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bcb990a30e934d6d9f58175cdfa88056 : SUCCESS in 1h 19m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e26c300e95854248b72a2610f77012b3 : SUCCESS in 55m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/848957a4c6a04676b1b71a727b30a801 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a511a0d11a644ba18f151b940aed5a55 : SUCCESS in 7m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42d12ce626a842aba0e9d2262f916c43 : SUCCESS in 5m 04s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1f41ba911b8b4853abbfc623d7f6d94d : SUCCESS in 10m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bcecd3f33547b0936a664e40af844e : SUCCESS in 5m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb629e06b1654727878f5f7b7cfa2dc2 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2344a3c477f5494aaefb387944ce1930 : SUCCESS in 9m 28s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/27d3aec8cde74e0dba1056ece93a2dad : SUCCESS in 15m 02s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/963ec1bb7ea340028478203a7536cdc3 : SUCCESS in 8m 16s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0ddc0e32e17144549607f31bab9dde53 : FAILURE in 1h 28m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ffe5eabf3b51497b947fd297330f4ac6 : SUCCESS in 1h 33m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8f89e284d7fb42a3949468ef04deb429 : FAILURE in 1h 44m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2d506e77570a483eb998a3d16f03f2c8 : FAILURE in 1h 52m 07s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/75702b04cec84891ae933a40c76f17fe : FAILURE in 1h 46m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0cfae0f489ce4834b7ad025a2c8a4fa7 : SUCCESS in 49m 26s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/92dd6285e87b4e95bdc6b72b5ed3e0ce : FAILURE in 1h 46m 35s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2af5d689ec65401aaa1881cce6618d66 : FAILURE in 1h 43m 40s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/562259fe8951401ca86c21aa18954426 : SUCCESS in 55m 36s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1f89ff9e028a44a688bf582bb9cde50f : SUCCESS in 28m 18s","accounts_in_message":[],"_revision_number":8},{"id":"c8555da076407601ea19db248ebe370a45108e96","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 21:28:21.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"7134c444b49081d830039326770482b1712f39f1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 21:44:45.000000000","message":"Uploaded patch set 9: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"17b28f3c8965424549dba613bd501e33527883c3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 21:57:24.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/047988a198114187a6dfa3b2cb4bf8b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/523a36beaae643cca76648e02d780228 : SUCCESS in 10m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/055f43e0f0164efda3e3f7a1670d2482 : SUCCESS in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"397d5c4ad03b59321a8975377649bc952cb80560","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 00:39:18.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/027a835a70b94e569db938f9270b8963\n\n- grenade https://zuul.opendev.org/t/openstack/build/04d2a14efbfc4d51bd1712be8d4b254f : SUCCESS in 46m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f10a5471dd3d4bfe9b31481b13293f45 : SUCCESS in 1h 16m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8df4bcec96a54002922f0ea2e3a3b353 : SUCCESS in 45m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/62cb3bbf019f412991a8a1f0b376368a : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4ba18fbb1a8f4935b0c35d3c7d3e8e4a : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/86738c519ae14b4d823f6f7f205d07c6 : SUCCESS in 5m 24s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/dd3a9f75861645f0920cf19fcf64ef64 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/523da4dcd1b74f128d2331d1e209f309 : SUCCESS in 5m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f5d709142d5845f58ecc41ac48f97872 : SUCCESS in 9m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/92f6e9eba792496782f8d3d8991e1be6 : SUCCESS in 7m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3aaea7eeb3394461bb1959d73fe3cb0b : SUCCESS in 13m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/ea24123ad7a54b7c9188414f48d07f21 : SUCCESS in 9m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4da11cb1c20d4fc3badc9531f8132d84 : FAILURE in 1h 48m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14eb1a367911493cbda15d45595eee6e : SUCCESS in 1h 54m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/733da8885ef34b9a89321e0efce1b507 : FAILURE in 2h 14m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dcbbc168e6ac46e19f3afa6df155188c : FAILURE in 1h 46m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a322118238ea4858b3e59868161f48f2 : FAILURE in 2h 46m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a90b86ea264842059f9e118e0f04f081 : SUCCESS in 54m 02s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d44461c6a521480b94ef06d79c739404 : FAILURE in 1h 53m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e7ab9032f00f467f8d56b9f659d1d801 : FAILURE in 1h 45m 49s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8918c728c0344074813fe2947bdb340e : SUCCESS in 44m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9fb7a32846c74cf688a6cd4563b512ea : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":9},{"id":"76569c138279964d156cddef682788da66e549a7","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-14 13:34:36.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"8d39099bb932e3fcf8a9c027fea0941d1e30d8a4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 13:48:32.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f259c525e2b740d1af8452af99a5a0b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bb2d7f1df9ff4203a3ac69f498e3ff2e : SUCCESS in 12m 18s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0caabb31e3de46d6b90520ec1b401801 : SUCCESS in 11m 02s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"dd472217527f07adf5f8039570fd910f557f5e85","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-14 15:41:25.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"fb1aa6ada1945124f49bde51d54c4ec7ed5cc8f3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-14 15:52:10.000000000","message":"Patch Set 9: Code-Review-1\n\n(8 comments)","accounts_in_message":[],"_revision_number":9},{"id":"c08e62d7c9d31bbd6dfbd8c5911c39b52782cc79","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:09:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/df334ba92cea469faf46e43dd1e28c0d\n\n- grenade https://zuul.opendev.org/t/openstack/build/c2c4101533ee45d39b1723558cfb6215 : SUCCESS in 1h 05m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6df10ed27b6749698ae1c9a002c3df8d : SUCCESS in 1h 22m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1edacb1b6ff940479449d7ff460cf4b1 : SUCCESS in 1h 02m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2d545c320f634f69b3de503a763eaece : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2d42b738ff244f1cad1b3bc000c273e1 : SUCCESS in 6m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f0efda9c3c54d638cba00a66e89b9d5 : SUCCESS in 7m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/477be5dd778241b39e5272407721e8ad : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d518892789ef468e92e2fa838ee251c6 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5618ade3dbcf4d4e8bb0ee456e2cc136 : SUCCESS in 10m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b64972776c0d4407bc3aaaa1357b42ec : SUCCESS in 8m 17s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/284dc16fcec84cfd98f29cbf22658b4f : SUCCESS in 12m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/48e75ba4a43e4396a0c7d9b699dfc299 : SUCCESS in 10m 10s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0914867da9b741c9b53f16a97e99911c : FAILURE in 1h 45m 32s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b61a649d16f8478cb2022384447869ee : SUCCESS in 2h 23m 40s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28c8f059e66c462e919cce4e73d3d297 : SUCCESS in 1h 51m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b98709494068484f99502718572d2444 : SUCCESS in 1h 50m 27s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/23309dba18d648ee95742d9fead0e028 : SUCCESS in 2h 28m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a20b8dce1f240348e6fdd38193728d1 : SUCCESS in 50m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e0b4ae30e5744152a8833215d1a2d218 : FAILURE in 1h 43m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e2e4146f01fe45058f1dc8901e540e58 : SUCCESS in 1h 58m 25s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/510c3e1415974176b18258b8ee88aea0 : SUCCESS in 1h 05m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7f7c7ba4048442088d9975eb4800a411 : SUCCESS in 29m 37s","accounts_in_message":[],"_revision_number":9},{"id":"05608f1ab008e1a6c733df156c07facdbd236786","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-14 18:37:02.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"add8f68ea45514247bc088955501238ce3e35a1d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:47:17.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926238 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"93306f333b88afa67a5f70a0daba0194da72975e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:47:21.000000000","message":"Patch Set 9:\n\nThis change depends on a change that failed to merge.\n\nChange https://review.opendev.org/926238 is needed.","accounts_in_message":[],"_revision_number":9},{"id":"53b65c9a38e99f9e777b7d22b575f57d7c5f5318","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-14 18:49:07.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"ad74c900cc99499baa57c5314f5c0f96f8139e47","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 20:09:39.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/780e6254a791456996f97613fda32872\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/eb7a33bcd21642c7b3cc70c1ffbb6327 : SUCCESS in 14m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3c310eec9b2e4e79b12aee3a05c1145a : SUCCESS in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"110e2a72118a43473de94804c680b8b4c034c7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 21:03:20.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a9143b426d0d4f47b83acc6845f1d8ad\n\n- grenade https://zuul.opendev.org/t/openstack/build/65830b279665498a8391af4501770d4a : SUCCESS in 1h 05m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce4ac74da6e444669baa052bad792809 : SUCCESS in 1h 24m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dfaadd862d54468bb02639eebfaed08 : SUCCESS in 1h 01m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/accd715e815e4725aaa832826c72c3d4 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1ba1cb72fa044b986b867d24ad4cac4 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a73b36d34bd944afbb707b71d3675410 : SUCCESS in 4m 39s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50da312a2cb3420d80c93352eaee9672 : SUCCESS in 7m 48s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d4c70aaa444aeaa08be79f5e7434e6 : SUCCESS in 4m 55s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/689dbae562a64444b9383a8f82d1274c : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8121f967801c42aaa5c69e9f5589b5d8 : SUCCESS in 7m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1728e47609b14343b3bc650b6af3d497 : SUCCESS in 12m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f669d8583e1042289ed38834f10e5c94 : SUCCESS in 7m 37s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d1971458f4b84d68a8761446dd1b18df : FAILURE in 1h 51m 29s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/65e3221d08834f619aea104f4659e324 : SUCCESS in 1h 33m 48s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/87681a950c5a482a8dec563175449b16 : FAILURE in 1h 39m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e2c16f363b4340e3b4daf65c243d9269 : FAILURE in 1h 25m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/cbcc67140c7c4619a67c25948ef71314 : FAILURE in 2h 08m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/601cdc423c3a4cca90582852ae4544d5 : SUCCESS in 48m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/642cff51494e46918d7ae7650d627767 : SUCCESS in 1h 32m 08s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/475d07ad7128494d8caa81a2d7511a24 : FAILURE in 1h 39m 29s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0aad2d302412462192ec396abc707140 : SUCCESS in 1h 00m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0b7548f5c8ae46f7aac1cd032c18bd41 : SUCCESS in 28m 09s","accounts_in_message":[],"_revision_number":9},{"id":"6abcee09431fe96bad02dbde4ab56615057e30cf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-15 14:59:37.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"1b558cdc7e49c046f7871c563e2a71f1140f40db","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 15:14:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/be4e1d3e00ee4840a11be54c3c40dcba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/15df68dbe2fc408d9bee95b9d90b7269 : SUCCESS in 13m 31s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ecd0defb4a9a4b97a5be4f3b416f9e23 : SUCCESS in 10m 15s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"3cf6411f13a21fe56b880000c639ed22a2cc0269","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-15 16:57:20.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/64c6c0093cf740bd8607e4494437966c\n\n- grenade https://zuul.opendev.org/t/openstack/build/e40e9e9b9d32403d88c893d873f9745b : SUCCESS in 1h 09m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1692dab85ba24087a4c49f66432549d7 : SUCCESS in 1h 17m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c261e08e66b4296b52439d56b5c3fca : SUCCESS in 59m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202bca99b22241a99257e9f82708ce97 : SUCCESS in 3m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/831dff6249cb49fa956ffd8284a7ed13 : SUCCESS in 6m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c43a5465f34460d8bfcb024609f1246 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/50c7fcd440fd46cfa795034472ca4122 : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e774d75d2aa84b96a448006874dcb233 : SUCCESS in 5m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/69532cfe107b455ca2e32dabd49ba011 : SUCCESS in 9m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/05fde19d44864b4ea93a8f8f15df13a8 : SUCCESS in 8m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d32731782e424bdd9952a28b75923e98 : SUCCESS in 13m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cf9fc8778ecd416da3bc2c50bb043933 : SUCCESS in 8m 21s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ced977f604c64e62a77feb5074df02d3 : FAILURE in 1h 43m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/43f2fa7e14c14c73978ed4e8482b75cc : SUCCESS in 1h 56m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a39d9327c10a472892b19a27e9354dfd : SUCCESS in 1h 51m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/60634c2c514d4ee6bc9d73f5b1a1cc4a : SUCCESS in 1h 48m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/15818e5af3d94c0b820690b4f85f8e65 : SUCCESS in 1h 53m 44s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/36f30875361c4db5a102e9aeadb5f731 : SUCCESS in 53m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a2da8b16b68c43288c218c00eab01507 : SUCCESS in 1h 38m 13s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/dabaac92d2cc425582ceddcd5c16ffbd : FAILURE in 1h 51m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4573c3d1a44c4ca590163fc4ffeba59d : SUCCESS in 1h 10m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/dddcfba8dc2f4edb8891edbb273138bd : SUCCESS in 26m 06s","accounts_in_message":[],"_revision_number":9},{"id":"5e7ff7b6c8d51bc1be5bab28043a28372a0f003d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-16 06:16:49.000000000","message":"Patch Set 9:\n\n(9 comments)","accounts_in_message":[],"_revision_number":9},{"id":"29659cc72bcbefc08980a12a9bcf6b4dfffd1ffd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-16 06:20:47.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"7fc24c6456806455157107d899d1ba4ac31d1943","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 06:35:06.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04df4c5a09354b1292cdd1127741400f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/df5943d2ca0046cc9097ca5dfa24dee3 : SUCCESS in 12m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f72b6401de6463485bdc92df09ffb81 : SUCCESS in 10m 04s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"ae5ad396ff177df48c72112a6178e64de5f46f6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-16 08:35:08.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b887f5d7d3041ba8357773daa1f79f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/62bf930f52ec4207ace5a65cfafe5dc1 : SUCCESS in 42m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/69549e2d765146888b288d7a2d0ee4b0 : SUCCESS in 1h 14m 54s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d7d5941e221d45cf872e10b979e385a7 : SUCCESS in 54m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c85dd5e140b49d6bfe380d311e942e4 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0324994f3ab24659b4fcd3b3af2e2a91 : SUCCESS in 6m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1acbdf79b604638af9da75c5c2c31e5 : SUCCESS in 7m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ad69f371d145441998a0bcaea0c1c4f6 : SUCCESS in 8m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f77a265696246a3a9cd19cd35b8fca3 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/30a9b7871e534ecdb112e18fdde7e24c : SUCCESS in 8m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e6facbf7f7874e71bc28648e01a187b3 : SUCCESS in 7m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a1135691f0bc45f180c3c15313050297 : SUCCESS in 14m 13s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/83d0df36d41c4fc4ac82ed0299c8db9b : SUCCESS in 8m 49s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9f7e854e12d24b5c84a8b0543ea7a1fb : FAILURE in 1h 45m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b58d20394f004f28a2715d561b7fd185 : SUCCESS in 1h 52m 35s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8578b9c4d6184620a63c480b04eadff5 : SUCCESS in 1h 51m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/51c6fbec64804a269eaa41510bbcd7de : SUCCESS in 2h 06m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7e85f0a5107c4b09b607e1c878faa095 : SUCCESS in 2h 13m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ada7442fb3904cf7927157bae9a4b2c0 : SUCCESS in 50m 27s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/618080f7fa5c4a11a31c013f8aa7b095 : SUCCESS in 1h 43m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b848bd3f7bb346689a53bcd0c3bf087d : SUCCESS in 1h 54m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/84b7794a7dac4b75a5e9949a45381cf6 : SUCCESS in 44m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ef42dd3af0ca4c05b27da065830dd0bf : SUCCESS in 33m 52s","accounts_in_message":[],"_revision_number":10},{"id":"925c994f96430fcbb73d6bf182ba86740f661090","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-16 17:05:23.000000000","message":"Patch Set 10: Review-Priority+1","accounts_in_message":[],"_revision_number":10},{"id":"70594c3fe0aa7e946ba2eba82d88b8533c34072c","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-19 13:57:34.000000000","message":"Patch Set 10: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"38a502a4c2f5b29ae766bf397e23589a3e11abb0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-19 14:07:18.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"08ee7c251c819f1fc878bdaa3ecb5e4321ab2b5f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-19 14:17:04.000000000","message":"Uploaded patch set 11.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"5620d63d48ade3b37c31f31b19aafdde52cc0a19","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-19 14:19:15.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"8eecf46129b870d5206a15dc6ed409b32214d72e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 14:53:25.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43695e8b92a54b07bf101802a759a80e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ba23db8b648343c681c571e4c8c8e124 : SUCCESS in 13m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3552e2cb3bfd4f5cbe5d293cb5e00349 : SUCCESS in 16m 57s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"aba388cae3ad9cfab0fb18d2d8768b59e6d6f130","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 16:36:20.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29489cd6a7434a1d9dd3054da302177a\n\n- grenade https://zuul.opendev.org/t/openstack/build/2bbdd8927d4c4cee8effb5781b69dfd6 : SUCCESS in 53m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/01c637b07b184ff0b829abd5fca749be : SUCCESS in 1h 14m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/593e7a7af5114a298a05a787460f672b : SUCCESS in 49m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ee99b6e8adf4f1b9f547a4e2bf1c10d : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ff3d5ba6ab194695ab71082c709ecc7b : SUCCESS in 5m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/421ed7c767054c439104c8bdaf58404b : SUCCESS in 6m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f253de62b26f4312ba6408adc20322a9 : SUCCESS in 7m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6355e0b0bc74bd0a80242f86af56e04 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c9afe3fabc9d42b3b4c9ad5ac9ce1450 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/afdfb1a247614a44abdc76f54ffd0045 : SUCCESS in 7m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/88ed52ff2d104776a103bd23ecb96e89 : SUCCESS in 12m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8f48741f08434831a4f946463dfa4ded : SUCCESS in 8m 18s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2440ac24c618410aa09414f22082765e : FAILURE in 2h 00m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/83ea9b0d93a24b7489d606d030ad4592 : SUCCESS in 2h 01m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3806b748f2ab4cf8bd2475777999e6e0 : SUCCESS in 2h 02m 34s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b69791d322a459fa6b84e7cab80b3f3 : SUCCESS in 1h 45m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6e19602e64184e4aafc84e6e228182dc : SUCCESS in 2h 13m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aba85529c89a44faa034bcd9f52b004a : SUCCESS in 41m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9ba02fe2694241c3a43e5a4fa13e39bc : SUCCESS in 1h 27m 57s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/f13f9d9045be47e893630033d6b989a7 : SUCCESS in 1h 29m 04s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/847169a395c44294a614980ca55d8498 : SUCCESS in 53m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8ba3026ea578456d8de1598df02480f5 : SUCCESS in 25m 46s","accounts_in_message":[],"_revision_number":11},{"id":"69011859945e97c7e0d1e9a7f6b3973f1b2297eb","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-08-28 19:03:20.000000000","message":"Patch Set 11: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":11},{"id":"6a2011c62cdf08e7727d12c21255caf9fc5d87d6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 19:04:42.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"77310e92433e9ce03e2aeb76c50523d5239eae2c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:23:37.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b570ce1563144e986bb1a81b116d1ec\n\n- grenade https://zuul.opendev.org/t/openstack/build/2001bab0521f4276bc04cd256e9f0a42 : SUCCESS in 1h 06m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c96d18a938e64d299d9ac4b6454e0dfa : SUCCESS in 1h 29m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5d7073b551b546ec830e39d76045990f : SUCCESS in 1h 01m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22825c6dd5ce48f084c58f83fbced191 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d89e59ecc42841fe88d52e1f2a2bdefa : SUCCESS in 6m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5da5cc2ae4c4464a84a723dfb6be14c6 : SUCCESS in 5m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e444a435a3264ca5a90b294ebff95338 : SUCCESS in 6m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/051f1889aa294339a14ca2c1e39cee78 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a32c3a317d8041789537cb09f5848e5a : SUCCESS in 7m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a21bfacd56794c7a8716e561ad741664 : SUCCESS in 15m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/15e5d3bbef7b43a9a27c2ca1efff9549 : SUCCESS in 1h 28m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1fcdd6ee1a674aa6b043e1a8a68fd812 : SUCCESS in 1h 43m 35s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d7f7dcb7f9584466a4d2ddf31496f032 : SUCCESS in 2h 00m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2b8cc37cf87348b194a1c78701924fca : SUCCESS in 1h 07m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/885a2d9067eb413dbdc490be00a8b8e1 : SUCCESS in 1h 41m 53s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e348403b7f68415c9d65b1cc3a7703d4 : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":11},{"id":"2d4b490b13f4a3481f72182d97bb1b315429a2f2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:23:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"7c8adabe7499faddf23fe7866b4ff8b757de11cb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-28 23:25:39.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/51236c947eb9426985c51d1b1d77a208\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e78208866214387a4f215e44f9a1902 : SUCCESS in 55s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d8de63a5008b811f2e22cc5cdda59bd2cc05b207","revisions":{"2df9591d7f06fdd737874ad1e6316a2cde29e271":{"kind":"REWORK","_number":1,"created":"2024-07-08 19:46:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/1"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:45:26.000000000","tz":0},"subject":"WIP Add Iso format Inspector","message":"WIP Add Iso format Inspector\n\nToDo:  Fix and add new Unit tests\n\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2df9591d7f06fdd737874ad1e6316a2cde29e271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2df9591d7f06fdd737874ad1e6316a2cde29e271"}]},"branch":"refs/heads/master"},"0c25f6ea9281cef768620074fe7261e33fa94bc4":{"kind":"REWORK","_number":2,"created":"2024-07-09 10:21:34.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/2"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-09 10:20:54.000000000","tz":0},"subject":"WIP Add Iso format Inspector","message":"WIP Add Iso format Inspector\n\nToDo:  Add new Unit tests\n\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c25f6ea9281cef768620074fe7261e33fa94bc4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0c25f6ea9281cef768620074fe7261e33fa94bc4"}]},"branch":"refs/heads/master"},"e1f0d80216f20f32e683422a41c8603258f829e4":{"kind":"REWORK","_number":3,"created":"2024-07-12 05:49:31.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/3"}}},"commit":{"parents":[{"commit":"f87f202bad7da5704dcb3006af8aa25ea206b872","subject":"Merge \"Fix 500 if multi-tenant swift is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f87f202bad7da5704dcb3006af8aa25ea206b872"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-12 05:49:16.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1f0d80216f20f32e683422a41c8603258f829e4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e1f0d80216f20f32e683422a41c8603258f829e4"}]},"branch":"refs/heads/master"},"30bcf74f94fc18da9f0dd1d074da91438d4aebb4":{"kind":"REWORK","_number":4,"created":"2024-07-15 18:12:38.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/76/923676/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/4"}}},"commit":{"parents":[{"commit":"f87f202bad7da5704dcb3006af8aa25ea206b872","subject":"Merge \"Fix 500 if multi-tenant swift is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f87f202bad7da5704dcb3006af8aa25ea206b872"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2024-07-15 18:12:26.000000000","tz":120},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30bcf74f94fc18da9f0dd1d074da91438d4aebb4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/30bcf74f94fc18da9f0dd1d074da91438d4aebb4"}]},"branch":"refs/heads/master"},"94d41c8318fa1a8884269f16f9c0991faf1d8659":{"kind":"REWORK","_number":5,"created":"2024-08-08 18:59:14.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/5"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-08 18:56:30.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94d41c8318fa1a8884269f16f9c0991faf1d8659"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/94d41c8318fa1a8884269f16f9c0991faf1d8659"}]},"branch":"refs/heads/master"},"4fa1b27c44e43390fcf6000449749c5ec3b13d34":{"kind":"REWORK","_number":6,"created":"2024-08-09 15:09:45.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/6"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-09 15:06:06.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fa1b27c44e43390fcf6000449749c5ec3b13d34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4fa1b27c44e43390fcf6000449749c5ec3b13d34"}]},"branch":"refs/heads/master"},"fe4b01f965d1a09962f3a729f109a2f39fa1eaf7":{"kind":"REWORK","_number":7,"created":"2024-08-09 18:35:00.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/7"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-09 18:32:10.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe4b01f965d1a09962f3a729f109a2f39fa1eaf7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fe4b01f965d1a09962f3a729f109a2f39fa1eaf7"}]},"branch":"refs/heads/master"},"0f9b6a8f4aa1a16aa16fcdd1a9d37ac63a320a45":{"kind":"REWORK","_number":8,"created":"2024-08-13 17:01:16.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/8"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-13 16:58:10.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f9b6a8f4aa1a16aa16fcdd1a9d37ac63a320a45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0f9b6a8f4aa1a16aa16fcdd1a9d37ac63a320a45"}]},"branch":"refs/heads/master"},"0098e642d5e779949fe031f060d594d00ab3261f":{"kind":"NO_CODE_CHANGE","_number":9,"created":"2024-08-13 21:44:45.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/76/923676/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/9"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-08-13 21:44:31.000000000","tz":-420},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nDepends-On: https://review.opendev.org/c/openstack/tempest/+/926238\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0098e642d5e779949fe031f060d594d00ab3261f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0098e642d5e779949fe031f060d594d00ab3261f"}]},"branch":"refs/heads/master"},"37771a8101b0684dab82443f61ba184e955cde14":{"kind":"REWORK","_number":10,"created":"2024-08-16 06:20:47.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/10"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-16 06:17:47.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37771a8101b0684dab82443f61ba184e955cde14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/37771a8101b0684dab82443f61ba184e955cde14"}]},"branch":"refs/heads/master"},"d8de63a5008b811f2e22cc5cdda59bd2cc05b207":{"kind":"REWORK","_number":11,"created":"2024-08-19 14:17:04.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/76/923676/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/76/923676/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/76/923676/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/76/923676/11"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-08-19 14:13:49.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8de63a5008b811f2e22cc5cdda59bd2cc05b207"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8de63a5008b811f2e22cc5cdda59bd2cc05b207"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
