)]}'
{"id":"openstack%2Fglance~923808","triplet_id":"openstack%2Fglance~master~Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d","project":"openstack/glance","branch":"master","topic":"zanata/translations","hashtags":[],"change_id":"Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d","subject":"Imported Translations from Zanata","status":"MERGED","created":"2024-07-10 03:53:30.000000000","updated":"2024-07-19 07:40:43.000000000","submitted":"2024-07-19 07:39:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923808-zanata/translations","meta_rev_id":"4cc5b42d3bf50628e6fde06d9de697ae2c6779c1","_number":923808,"virtual_id_number":923808,"owner":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-19 07:39:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-19 06:30:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-19 06:30:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 04:08:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-10 04:28:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-19 06:30:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"3f47216f79e06120c123e60281180ded79ba5a20","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-10 03:53:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"387e18b023a3f32ddea8e837913e910043a4ebea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 04:08:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4d4f2864ec444ccc9764ac6d9cce9b8c\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d17393efa65349fd8ba69f77ccb0ee9f : SUCCESS in 8m 19s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f50944edb3d1435c932f98c7ac0346d6 : FAILURE in 2m 40s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33aba68b6cd19f2cbde192972bfc7055734c6b0b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 04:28:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e15e6b62906949b58cf3697460de9bc2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fbf7122bbff4e48b49806779353afd6 : SUCCESS in 3m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fb49478ba9924ed8a7c10e8009a0ba63 : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ae0b6b63da3c487b971104132238f813 : SUCCESS in 4m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c7a1e5c1d95745e8bdc82fc3d488e55c : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2b4b3a01764a16bdafcc402a44dddb : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c8c6ef848a14ac1b350c557236d714a : SUCCESS in 6m 50s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/76e03940b4354991b78564d83958e38f : RETRY_LIMIT in 2m 02s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b036e56fa9954f0ba8ae2969c2e836be : SUCCESS in 10m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/048f78fa1de04f03be7067c1ec45b570 : SUCCESS in 9m 02s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/68776ceb635c445084c98b0603948cc9 : SUCCESS in 15m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/f6f4e494f9a84df3accbf24521d56cbb : SUCCESS in 7m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4b02ceae4b7f6b1794e754a9f8e5ec31e2fad602","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-12 02:47:55.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f66688bfbc79dce7485b9eeb29a618dea262225a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 03:04:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1bf879b602bf4a88aced82c6ad42a914\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6333673fa9d5485786034a72a620f10a : FAILURE in 3m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/752d2fc11a394afaabab775acadee612 : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"882408ee426838b2e4e1a430ae45ce6c47872ef3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 03:09:53.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a3d32563cbb6496495d0df3d6ffa6549\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4764666042d4d609ad503dc17212a2c : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9101b9d5d36442b6b360741b43aefa2b : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3326e1edb79442dba48a2f903b89b774 : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/fa51efd13cf84e97a60c7d805945158d : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a7e17e4b06e46a3a397d37da0806884 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ced0d26c64224e1aa4139452605416fd : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d40c953d834d4d24a7f8efb5d31fdbaf : SUCCESS in 8m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ca427bf8406a4ca68ebed654f8a6e706 : SUCCESS in 8m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3955acca531f4307afaf7e6945282a53 : SUCCESS in 14m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8e7e88fad70f47a4bf3d6cf85c780b6b : SUCCESS in 7m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2577ba2db8daca83ede95fb59cae11edc41accff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-13 02:57:44.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"e72468f6c6cb6c453027f75cac11ab013832904e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 03:12:29.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9547a39ae6f14d74b091459759cdcc43\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8010874ccf184a4b9ac0f58c4d05e990 : SUCCESS in 4m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3161a6320cee483b8a45dda1087ea030 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/02114753511c4337a2c6540c3f90b75d : SUCCESS in 6m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/09d89133b6a04a488b70cb162652375f : SUCCESS in 8m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/578cdbb2d6b44b6ba4b4eb7b6dd0a56e : SUCCESS in 6m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9871bc49f08644daac58339c8c025fe1 : SUCCESS in 5m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29270ac468a14ebeaceef0185594b674 : SUCCESS in 9m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe8bff991b194db886a4715130ccb80f : SUCCESS in 9m 56s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/ebf64301681045ba857bbe768cd97e11 : SUCCESS in 14m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/01f2a1fad626462aa9be5bbdb828cb63 : SUCCESS in 9m 29s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"6338edbe0cb2a42f9002e08a88d44274ff9a5282","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 03:16:10.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/054214ef0af94d03b8a21348570c1199\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/493dff7eba1440c0828cab023db97ee7 : SUCCESS in 13m 13s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/30a6db9f381241b299312bc868b28f6d : SUCCESS in 5m 32s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"87b198fa6e78d1bb289dc5b4997384510b0d3a99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"date":"2024-07-19 03:05:08.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"8fb93a106617d9812a04f044c1a4aba552c32e41","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 03:11:48.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a437923b4644d828d7f9701e9a38f05\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/57bbaa2da79945cc8a7b01a068e1ee97 : SUCCESS in 5m 44s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/dbe0f0c883ec473480a97d77826d133c : SUCCESS in 5m 49s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"a4057ba55b6a8efbb0f023128315098db1f5ea24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 03:26:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/102b3c3603494b228911e95e76d06413\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/79aa03597df84b1a9aad2ec14d13e8ab : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ad5a0d2728842b4bccc59c1b1ae8eca : SUCCESS in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/982a3525fdcc4909aab68b19e4eb5487 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2d1ea17fcea943339b9776f6a0204786 : SUCCESS in 12m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc77310c6069469bab5c842b1dddb71e : SUCCESS in 8m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dded4b1a630a488883599753ba2f0161 : SUCCESS in 9m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0bddb243c6834e2086d3bf4f1400be92 : SUCCESS in 11m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/37afdea0781e404a9ef09430019425cd : SUCCESS in 10m 29s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3d0738520c86489890ffc2e94a7f7423 : SUCCESS in 12m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d8fac0b09f864a6087da34e60ec25c8c : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"33995a781505060c89af59cbf585ed5167297517","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-19 06:30:41.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d5dda457e20d574ed0df8258361c36c38b798835","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 06:31:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"b01fd30fd1bdf8097d77b188bcabdc1c5e1af79d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:39:07.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f83ee37d480a45adbc00eef5b8e864e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c25fbcab94d74cbeb5397609c4775400 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f1d6bbb1f44c462c93bf6fbb917edb8b : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3fbb10d6ec048c6a9aaa831a07ff7f5 : SUCCESS in 5m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c8e58cbd421b44c78ba8c2e228987efe : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2d6fed2680094e4b9b5da641ae4d8f5c : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c44771c36c4f45ebb19c4abeacc26437 : SUCCESS in 9m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aede8de72d9649f4b7d777f206987133 : SUCCESS in 7m 44s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/81613297f2744963a46566787a86fe7f : SUCCESS in 12m 35s","accounts_in_message":[],"_revision_number":4},{"id":"bec3756ec53bf6dffc9336ad87bbdbe4a99681e7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:39:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"4cc5b42d3bf50628e6fde06d9de697ae2c6779c1","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-19 07:40:43.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b97eed82ac6d4669bb75299114d9ed5c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d685af3b38a4595ab20cd3af904e163 : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e5ec617ba85c4ad4a56f487a25aedb88 : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce","revisions":{"1e93d8b298467ba09c0ad7d207b49a1dd4bbafef":{"kind":"REWORK","_number":1,"created":"2024-07-10 03:53:30.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/923808/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923808/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923808/1"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-10 03:53:28.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-10 03:53:28.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e93d8b298467ba09c0ad7d207b49a1dd4bbafef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1e93d8b298467ba09c0ad7d207b49a1dd4bbafef"}]},"branch":"refs/heads/master"},"d3ed1cc9d076412b9164338863106154cbf93e7e":{"kind":"REWORK","_number":2,"created":"2024-07-12 02:47:55.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/923808/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923808/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923808/2"}}},"commit":{"parents":[{"commit":"f87f202bad7da5704dcb3006af8aa25ea206b872","subject":"Merge \"Fix 500 if multi-tenant swift is used\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f87f202bad7da5704dcb3006af8aa25ea206b872"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-12 02:47:53.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-12 02:47:53.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3ed1cc9d076412b9164338863106154cbf93e7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d3ed1cc9d076412b9164338863106154cbf93e7e"}]},"branch":"refs/heads/master"},"935b86f3dac88127116095c101b00ef768a9a6b8":{"kind":"REWORK","_number":3,"created":"2024-07-13 02:57:44.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/923808/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923808/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923808/3"}}},"commit":{"parents":[{"commit":"a42114c5fdd009a9fd5163b581cb13b1ff2c68ca","subject":"Merge \"Fix /healthcheck url typo in glance-api-paste\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a42114c5fdd009a9fd5163b581cb13b1ff2c68ca"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-13 02:57:42.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-13 02:57:42.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/935b86f3dac88127116095c101b00ef768a9a6b8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/935b86f3dac88127116095c101b00ef768a9a6b8"}]},"branch":"refs/heads/master"},"38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce":{"kind":"REWORK","_number":4,"created":"2024-07-19 03:05:08.000000000","uploader":{"_account_id":11131,"name":"OpenStack Proposal Bot","username":"proposal-bot","tags":["SERVICE_USER"]},"ref":"refs/changes/08/923808/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/08/923808/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/08/923808/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/08/923808/4"}}},"commit":{"parents":[{"commit":"dfab4bc5986604ace9aa7ce4ebc86f0a623d51fc","subject":"Merge \"Revert \"Fix import job to provide valid disk-formats list to tempest\"\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/dfab4bc5986604ace9aa7ce4ebc86f0a623d51fc"}]}],"author":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 03:05:06.000000000","tz":0},"committer":{"name":"OpenStack Proposal Bot","email":"openstack-infra@lists.openstack.org","date":"2024-07-19 03:05:06.000000000","tz":0},"subject":"Imported Translations from Zanata","message":"Imported Translations from Zanata\n\nFor more information about this automatic import see:\nhttps://docs.openstack.org/i18n/latest/reviewing-translation-import.html\n\nChange-Id: Iac3eaade4a6aae48250dfa669c9cc3fb2e0a542d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
