)]}'
{"id":"openstack%2Fglance~923861","triplet_id":"openstack%2Fglance~unmaintained%2Fzed~I712b304ebe168705ec2793626f5d64135ad6d458","project":"openstack/glance","branch":"unmaintained/zed","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2024-07-15 13:13:16.000000000","reason":"\u003cGERRIT_ACCOUNT_17685\u003e replied on the change","reason_account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}}},"hashtags":[],"change_id":"I712b304ebe168705ec2793626f5d64135ad6d458","subject":"Fix import job to provide valid disk-formats list to tempest","status":"MERGED","created":"2024-07-10 14:46:36.000000000","updated":"2024-07-15 14:56:37.000000000","submitted":"2024-07-15 14:53:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923861","meta_rev_id":"adc0c59e6c524de8035ecf05561e07c4717a145b","_number":923861,"virtual_id_number":923861,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-15 14:53:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":2,"date":"2024-07-15 13:13:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"all":[{"value":1,"date":"2024-07-15 13:13:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-10 14:46:36.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-07-10 16:13:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-13 14:52:39.000000000","updated_by":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"CC"}],"messages":[{"id":"48363923763a8327db548846f7937a16f4db377e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-10 14:46:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38c19c93fd198057cd9c8ff2cb45836353aa4a18","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-10 16:13:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c785dd461507465ba85d39e65c03e0aa\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eb7474e567d456397467f2ac202a9f5 : SUCCESS in 56m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47aac36971ac4971a6e24a3d3fd04cd6 : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aea435d68ef940eabd733f04057af9e3 : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b9d33eca57be432db134ec393cf526e2 : SUCCESS in 5m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/959a009fd9b44fddb55b3edd9bc9b118 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98476e8c93a84e1e9b6b2da305e401a2 : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cec419eabe324089b23b20a2a281d0b6 : SUCCESS in 8m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/887a5e3b9f534b8280ae12b1e2eb9bff : SUCCESS in 9m 40s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/51d02577f21b48babf3d30084383c0b9 : POST_FAILURE in 1h 19m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4e3dfa5b5a224e0985228100ed2b02a5 : POST_FAILURE in 1h 12m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2d48ebb3a2af4842b2d6835c6a584b5c : POST_FAILURE in 1h 17m 40s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/91cebf9909ad4b63b44f16b19c8cb39e : SUCCESS in 33m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4864ebd351c1207e62c586b024e2ba3a09c0fb32","author":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"date":"2024-07-13 14:52:39.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"90d706c7cdd52db6e13ce7c375f4048b76c50a5e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-13 16:13:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/50932fc20bc04e7ba5220f5ece5cdf4c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4e638430ccbe4135ba18e8829b6648e8 : SUCCESS in 54m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/98e0b2e443a7470ba66ab398ce78d00a : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8d2ca6528eb244b5876b6b32685e42bb : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7f749f33fade4e379ed12e71b190111f : SUCCESS in 4m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8122478cb31b45a1a5568a31dc31fbce : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c9bfbbd83d44e8db6c7f8d2784fee97 : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56ad231b9b4b413bbb540b851f62df82 : SUCCESS in 7m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/891e53ec32564227a956cbf4c4e7535a : SUCCESS in 7m 31s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/41d21464a45f4697b61fc5e107cc9cc2 : SUCCESS in 1h 06m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/967abc7a26ac4dcfa4322406b33e78fa : SUCCESS in 1h 07m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/525fda9dc91c42bfbafb519dbce86697 : SUCCESS in 1h 17m 43s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3ed361410b3c475aa1d896682ff0b585 : SUCCESS in 25m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c8aa95ab08eb0c64f0ca6e8b519bbb80f7736be0","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-07-15 13:13:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"850cc6bc3e8e7bb4de913ccb714f77ea8f74d325","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 13:23:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"514ae00c5288050aa44cabd5ae2067b9ecc27234","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 14:53:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a666f106ed443dae5503e6071b50c7\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d465653d0bf24d70927734a53b07e10a : SUCCESS in 38m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3149d41a5c0b4ae0a4519f466f283890 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e6017c81484a404ebbc6f7a51a6d6c17 : SUCCESS in 5m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f87e8d4adadd408e844ab20a55a6dd65 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0549c4ff1ba04d3b805cd298c4b4b272 : SUCCESS in 6m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8281ea3c3a8f406d8db9fd4648c57a1b : SUCCESS in 6m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d53400c805a84914b1a0008662b9ee2f : SUCCESS in 8m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ed67feafddad42e0bf36a22dc6e9d7fb : SUCCESS in 1h 29m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b904e6722d4243f2a21dba1e3b336bd6 : SUCCESS in 1h 13m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3492575cc1234aa69e130d0c49c4ba92 : SUCCESS in 29m 25s","accounts_in_message":[],"_revision_number":1},{"id":"a8069a0c25481781dc1266aa81337c91130cbb09","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 14:53:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"adc0c59e6c524de8035ecf05561e07c4717a145b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-15 14:56:37.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6adf36431bba4612bcce6a07519ba096\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9cbd5322a4b4104a64e74c386f7e966 : FAILURE in 1m 00s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"02732c2c3bb4b1713780e2fb081d5d61afdfabd3","revisions":{"02732c2c3bb4b1713780e2fb081d5d61afdfabd3":{"kind":"REWORK","_number":1,"created":"2024-07-10 14:46:36.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/61/923861/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/61/923861/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/923861/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/923861/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/61/923861/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/61/923861/1"}}},"commit":{"parents":[{"commit":"a5ec69ba1c0a92cc9b91b52b4101b81a330ea684","subject":"Add safety check and detection support to FI tool","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a5ec69ba1c0a92cc9b91b52b4101b81a330ea684"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-02 17:39:32.000000000","tz":0},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2024-07-10 14:45:50.000000000","tz":120},"subject":"Fix import job to provide valid disk-formats list to tempest","message":"Fix import job to provide valid disk-formats list to tempest\n\nRecent changes does not allow image conversion to convert an image\nhaving AMI disk format for image. Passing list of disk-formats to\neach import job which will now use qcow2 as default disk-format to\ncreate the image.\n\nChange-Id: I712b304ebe168705ec2793626f5d64135ad6d458\n(cherry picked from commit 244ccb68dd9c1be8b8c6223c50998464a394b688)\n(cherry picked from commit f98e9ae4a80ab206a8214572df22ac6134908d14)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02732c2c3bb4b1713780e2fb081d5d61afdfabd3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/02732c2c3bb4b1713780e2fb081d5d61afdfabd3"}]},"branch":"refs/heads/unmaintained/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
