)]}'
{"id":"openstack%2Fglance~923911","triplet_id":"openstack%2Fglance~master~I3350fd45adbd5ba878ca588d28cb0057fc1465e9","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"3031":{"account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"last_update":"2024-07-12 03:06:48.000000000","reason":"\u003cGERRIT_ACCOUNT_3031\u003e replied on the change","reason_account":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"}}},"hashtags":[],"change_id":"I3350fd45adbd5ba878ca588d28cb0057fc1465e9","subject":"Fix /healthcheck url typo in glance-api-paste","status":"MERGED","created":"2024-07-11 00:58:44.000000000","updated":"2024-07-12 12:29:11.000000000","submitted":"2024-07-12 07:32:22.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"923911","meta_rev_id":"5ae6b5ffa91fe17f3ae417e8bda47de873a5be7b","_number":923911,"virtual_id_number":923911,"owner":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-12 07:32:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-11 11:51:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-11 11:51:28.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-11 01:02:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-11 02:59:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-11 11:51:28.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-12 12:29:11.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"b38b6f500c96331500f483d5b96acd49e71d1b70","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2024-07-11 00:58:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8871233338a44d87fde501469f8960918485ccc","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 01:02:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c3a900c00104432bc9a5e4b94e45d1e\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/145299a9d4ca42a29e4a91c39dc5c151 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d914bd69adf140368706930aef06dc4a : FAILURE in 2m 47s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"31b5089cac4025e6b6cf01f53ace19ad55dd1eae","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 02:59:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/60a715ba2a1c4077845bb241376ea152\n\n- grenade https://zuul.opendev.org/t/openstack/build/214e7ed94e1c480f8786870411df6e8b : SUCCESS in 45m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8a0598eaba7a45b3a9e93617d02b8432 : SUCCESS in 1h 11m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d1f62594e84d471a9929815efc4ec229 : SUCCESS in 57m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b039c0d4685e47bfb467236d566f0653 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3b0fabae00ec4da2a7f182e5326f0a76 : SUCCESS in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e82f0f00d7564c14987d5d24c2166a4c : SUCCESS in 5m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2618ada480314929acaf35379de371cf : SUCCESS in 9m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0c1fc0d994804b61b2a862114ca697c2 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38-fips https://zuul.opendev.org/t/openstack/build/d0c113283de84175aa7270cd5fc3768b : RETRY_LIMIT in 1m 40s (non-voting)\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/107ab7bb277545c9ac965d74bde37c7a : SUCCESS in 8m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/29d0c7fe5b644933a39d3b5187041582 : SUCCESS in 7m 47s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d0f934bfd5ae4850909fe121fbb193ca : SUCCESS in 11m 39s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/6dc99651edd74caf8aa1ded9f104383a : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/12c24c85e1c94681bccbd36e9b208c56 : SUCCESS in 1h 42m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b83ab389d1eb4c75a04a1d1718bbde38 : SUCCESS in 1h 53m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28dbf77770bc498585aeca7613594dc5 : SUCCESS in 2h 00m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ee8c778e58d344a6b1da3c777144e699 : SUCCESS in 1h 25m 32s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/56d66711f72f4ed5adc529ec4c0811c3 : SUCCESS in 1h 50m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/68b5f12c28f84438835ab852497e6b5b : SUCCESS in 38m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e21dcabd5d8248efbe005ba59ab60343 : SUCCESS in 1h 31m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/5f66d7b1eac340f78bd1a077013c5f64 : SUCCESS in 1h 30m 08s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a443e5dd8fab49bcb1fe59f47e0232dc : SUCCESS in 1h 17m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d44e6316f8a347abbf5af3d4cb3a79cf : SUCCESS in 33m 46s","accounts_in_message":[],"_revision_number":1},{"id":"7d6617f88f3adc66296223b4a3d7347f231dd10c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2024-07-11 10:45:39.000000000","message":"Patch Set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"93b41b87716043f95f6faaf22e9b41db18ae3495","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 11:02:14.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c53e10572cf42b39be31f5239325269\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e819dc6ffb484fa7bca13daa0e79d85c : SUCCESS in 14m 35s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d398a013200e4542b59f1d588778aa3b : SUCCESS in 12m 07s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f328292236cd407795abe7d0f8bd533cf66d18d2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-11 11:51:28.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c74ce6ed61039193f183ee0b675a974ab9f53fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 12:53:47.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2bda10cbd914214b9d1caaec77bd2e5\n\n- grenade https://zuul.opendev.org/t/openstack/build/15bc0bc4a6b64c2686a7ba1eeba3d6fc : SUCCESS in 1h 17m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9bcec7012a87427f94dcbc9baae78374 : SUCCESS in 1h 32m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/bd2725b916264d40a2ee9740607dcbfe : SUCCESS in 58m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/91cb6711381c49eaadef2b65ec00fc87 : SUCCESS in 4m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d936ffadbfe54166859f20492e844287 : SUCCESS in 4m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2989c1346b7d4b91b25482fc5c476599 : SUCCESS in 5m 23s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/26aed02ecb8f42e495a24946d1e33f11 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/465a41c2fffd4c08a606a59dfd4fc2e1 : SUCCESS in 5m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/47ed365b863549b58b67cf28ff3d6d60 : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95f7094ca027427c9a395532de250936 : SUCCESS in 8m 09s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a661377e3ae2403fbc48ad76f76b9d91 : SUCCESS in 11m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d0269ed7954c4b97b28e02cb1cf7301b : SUCCESS in 8m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7eb2053315394b5baeaccbf581ad4359 : SUCCESS in 1h 50m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5bcab8d6b5b347a59394c2b0edae2934 : SUCCESS in 1h 39m 51s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0e23cf3c32a04ca69d29865ef80d0a07 : SUCCESS in 2h 02m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/35a5ca79bbee4f47879500347594f6f4 : SUCCESS in 1h 42m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/af9ec0c3fa984903922daa6851baf762 : SUCCESS in 1h 52m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d587fe634d9d43f4a46526ac78647fbc : SUCCESS in 52m 25s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6b7a30bbfedd40edba4c3371930c1793 : SUCCESS in 1h 36m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3c7379fd965f4d6f849df96418e5a15c : SUCCESS in 1h 56m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cb8d47cff3614a0280e468cbbc96c801 : SUCCESS in 57m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c22a527de40045ef89725afa6abfbc13 : SUCCESS in 25m 54s","accounts_in_message":[],"_revision_number":2},{"id":"f9e9af392f615617be4d80829e6ef1b889b4bd50","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 12:54:32.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"2d154c3e465a548801fa06ba85452286b2c43d35","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 17:32:00.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/24beed577df24c78849bfa587a471891\n\n- grenade https://zuul.opendev.org/t/openstack/build/0b6f62a68b394d0cbfd7b95625759032 : SUCCESS in 51m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ef62a808ec284ae0a85a8e8adda43b16 : SUCCESS in 1h 17m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/caf71d83ba0342cfac1cd238ceba9378 : SUCCESS in 1h 07m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/143230ac33c949308eb3f433d952b94a : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d86e8280a57e4911932d44e4bcf4a81c : SUCCESS in 7m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4420c240200404b858d662851721930 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8c5e27edfe44887b6b005b404010627 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/96f871ac2ffb4c47ab8a593a72d85522 : SUCCESS in 9m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f4152859b49b4ce68956d785eb5de8b1 : SUCCESS in 8m 24s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/bfdd05f5a01849d8bda2340eee903319 : SUCCESS in 12m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/95c360dd2ba04b9bb55f26dc49f4112a : SUCCESS in 1h 56m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7917cba404be4e1fb07d47f93ff61201 : SUCCESS in 1h 35m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1f1be1e8a99c4ed381b376aab2b71930 : SUCCESS in 1h 53m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/01f9a75ee8204692b9226481376094f8 : SUCCESS in 41m 38s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/83cacabdaaf446ef91f7ae0e8b6aff35 : FAILURE in 1h 56m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/33625c677c7846d9a6bc3e70b8e598e3 : SUCCESS in 22m 39s","accounts_in_message":[],"_revision_number":2},{"id":"45624b2fe809c59d6bd4d3bd949959d7b39a40c3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-11 17:36:24.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1d2a22807f631586643610fd44de1ab68de039fb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 17:50:40.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/08ecbfc594b244c8989b4fe457548563\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/1983feae791441899d0e5bae4595cfa6 : FAILURE in 2m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/21c53bfdd85d43d386d841cf7174c58c : SUCCESS in 10m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"53480c86693798195ad59733a9e68fe571a270c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-11 19:48:50.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/09f4fca72b664b138e5780f5a8c72234\n\n- grenade https://zuul.opendev.org/t/openstack/build/d44ef18226f54ec0a999d7fa877c8055 : SUCCESS in 1h 03m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7aaa85ba56e642a38865c655992fd353 : SUCCESS in 1h 38m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e1a32197283046a4a4b5bbf803788d71 : SUCCESS in 1h 01m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e1a9c79e5862431b8c22828d4309fa2e : SUCCESS in 5m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5c1e4c24205488c94f5afea0b1cd60e : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4f4bf380d18d4e048fbcee596dd83b0d : SUCCESS in 7m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ae3c25aa22164e3793b9fd0ff0ecbeb6 : SUCCESS in 8m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2e8a535587fa4d3c928bfe7a2c13178d : SUCCESS in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2044d557dc15453c88c8721381e2ffa5 : SUCCESS in 9m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99fe36fbc47c4ce689b6ae7e5781ac1c : SUCCESS in 8m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b3bf84a5a5774262bff9259eaa494b00 : SUCCESS in 11m 29s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/428ca8e473a740d7a565a9b32680de57 : SUCCESS in 8m 14s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f2efa23f6e284cba85a627f86990fdf7 : SUCCESS in 1h 50m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d4f77fb5b6624658a065a1c9dce95dec : SUCCESS in 1h 44m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/553fc6d97ce14417821f62147b8a537e : SUCCESS in 1h 43m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f3591f9908c1479f9d893d4fc50e0988 : SUCCESS in 2h 01m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/92215cbd20084f5b905d8e4cb5aa21d2 : FAILURE in 2h 03m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4ff637f48de0461bbe0cb06720cb517e : SUCCESS in 43m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a13a4c71f26f4690af05c8177f64c125 : SUCCESS in 1h 42m 20s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/521b8474a7bb48a9bf78e5c212346876 : SUCCESS in 1h 45m 09s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bdc8c9d728df45b6b804320b037c7696 : SUCCESS in 1h 09m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/0f0dbb1afda74daa8310dbf1b202f8d8 : SUCCESS in 32m 24s","accounts_in_message":[],"_revision_number":2},{"id":"0da00adf63747cb1ba8bb455d0e7bdf490d07746","author":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"date":"2024-07-12 03:06:48.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"118852e8400ba8053de854492c75864c243793ee","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 03:24:42.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9a36a323e304cea9b759e6d44d92c65\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f4fe24ab7c36473087739f943c65a8de : FAILURE in 4m 06s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/60acad2dc01e4772a61829042d70fa28 : SUCCESS in 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"6d13646f0cb6aa8ac75b491c5e3431c9ca33da89","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 05:25:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3e18b792ca6c4c32a5f7212f99dd4347\n\n- grenade https://zuul.opendev.org/t/openstack/build/3824d7e6192241508637e196e767149a : SUCCESS in 1h 04m 43s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/46e41980cfdb466ea0e0b2eb487cf701 : SUCCESS in 1h 30m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/e888659e010d4b4fa0bf11871d1e387b : SUCCESS in 56m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/74c50c49483f4053b396136bf5ce61e6 : SUCCESS in 3m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/51d8ab5c2a9045dcbe8e06e7c748556e : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6ee7f9851934a2681cff14ef2f5f7ce : SUCCESS in 6m 02s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ecc6f4f170914023b0ca7c21e4446876 : SUCCESS in 10m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1131af3419f41e0b26227c4311c3aa8 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9c2b84ddd6b1445aa2c5759000fc9980 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/02651cce499b4c6f8eeaadd0cf4e3fe3 : SUCCESS in 9m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e2b99307f9fe4a7d975eb21586948595 : SUCCESS in 12m 53s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/784cf70eb2014569ae48db9fcd8ed748 : SUCCESS in 8m 11s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3eb2675ea74b4b0d863969e0cf1caaeb : SUCCESS in 1h 40m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9960981327894ae4b591cf8e93d65167 : SUCCESS in 1h 41m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/70983997cbe7423cb07bf3c0ddaab912 : SUCCESS in 1h 18m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/781553237d034094abd5ead211f712e2 : SUCCESS in 2h 08m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/919dc7a1b12140b983ea9e3b10be95f7 : SUCCESS in 1h 54m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/26f7b73afc3e43ad80002384cde07493 : SUCCESS in 50m 45s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a4217ae27d8b4eb9b9a7ea061df35e99 : SUCCESS in 1h 44m 40s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/eabd9ab3c2034a2fac815f7bfae1b79e : SUCCESS in 1h 48m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cc52029cdb054c46b201af796d4251d7 : SUCCESS in 52m 08s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f51cba9ec90f4a38b63cf87a53190c2e : SUCCESS in 24m 30s","accounts_in_message":[],"_revision_number":2},{"id":"28e66cb4c8043467d603ed88437610e950395893","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 05:25:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"afb70c1f6c553a5ec004a11e0dd13f664d69c6a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:32:22.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/650ac36d8d8e48de9fa1c18b618132fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/4497ce1aea9744d1935377c66f84075b : SUCCESS in 51m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b31b12bb9a0640558d966d268e15904a : SUCCESS in 1h 06m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/13c85d7aaebb4ab6b37200268e6a992f : SUCCESS in 42m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9f3a1df57014a8b95576c453b87cd3e : SUCCESS in 3m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e1c9fdf0a3544b5aa2a432f6824f9e2 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d77b57124c246b181813678f780e095 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/466bd65208634d6daa0cfab4d0c37f48 : SUCCESS in 4m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35ba1d439d724b389ac92f334a6ee0e9 : SUCCESS in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9ff8d5fc0bf3488ba514058f81624e9c : SUCCESS in 8m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/08fd24919e2240b0ac446b2e1ef2bc89 : SUCCESS in 10m 28s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/150c3a0f32c74c00925203802590c94a : SUCCESS in 1h 50m 00s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a8c265111a264aed8b1826908b85f952 : SUCCESS in 1h 42m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a96c374c44704246b24c2a2be674fa5e : SUCCESS in 2h 01m 26s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d56d7f07224562bfaec4727aaa310a : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e567691ebdc944eb8410412b2287699c : SUCCESS in 1h 43m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/53016f8f01c94e589416080a9676218b : SUCCESS in 31m 17s","accounts_in_message":[],"_revision_number":2},{"id":"b8c7407c5c938317acfa3fc52211a4d01a97a736","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:32:22.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9075c4b585db9c198ef5fc72e06a66ed8b3989b3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-12 07:33:54.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3a0834c515614156bcf640662da1514d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11cd1af461894abba8c39ae195f39609 : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":2},{"id":"5ae6b5ffa91fe17f3ae417e8bda47de873a5be7b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-07-12 12:29:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"629ea06a5029143136eea23774891e0b75b9084f","revisions":{"ad645dce07c06e2411424ba6592f05dc5694431a":{"kind":"REWORK","_number":1,"created":"2024-07-11 00:58:44.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/11/923911/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/923911/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/923911/1"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2024-07-11 00:58:18.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2024-07-11 00:58:40.000000000","tz":600},"subject":"Fix typo in glance-api-paste","message":"Fix typo in glance-api-paste\n\nChange-Id: I3350fd45adbd5ba878ca588d28cb0057fc1465e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad645dce07c06e2411424ba6592f05dc5694431a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ad645dce07c06e2411424ba6592f05dc5694431a"}]},"branch":"refs/heads/master"},"629ea06a5029143136eea23774891e0b75b9084f":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-07-11 10:45:39.000000000","uploader":{"_account_id":3031,"name":"Sam Morrison","email":"sorrison@gmail.com","username":"sorrison"},"ref":"refs/changes/11/923911/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/11/923911/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/11/923911/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/11/923911/2"}}},"commit":{"parents":[{"commit":"2e65391744a82421bc6f026ee8f1f3550038f175","subject":"Merge \"Add releasenote for CVE-2024-32498 fix\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2e65391744a82421bc6f026ee8f1f3550038f175"}]}],"author":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2024-07-11 00:58:18.000000000","tz":600},"committer":{"name":"Sam Morrison","email":"sorrison@gmail.com","date":"2024-07-11 10:45:39.000000000","tz":0},"subject":"Fix /healthcheck url typo in glance-api-paste","message":"Fix /healthcheck url typo in glance-api-paste\n\nChange-Id: I3350fd45adbd5ba878ca588d28cb0057fc1465e9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/629ea06a5029143136eea23774891e0b75b9084f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/629ea06a5029143136eea23774891e0b75b9084f"}]},"branch":"refs/heads/master","description":"Edit commit message"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
