)]}'
{"id":"openstack%2Fglance~924388","triplet_id":"openstack%2Fglance~master~Id2548fad3e90144b47964b82bb4f702ad6407ae2","project":"openstack/glance","branch":"master","topic":"fix/disk-formats-tempes","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-18 13:35:17.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-18 09:47:47.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-07-18 13:35:17.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id2548fad3e90144b47964b82bb4f702ad6407ae2","subject":"Revert \"Fix import job to provide valid disk-formats list to tempest\"","status":"MERGED","created":"2024-07-18 09:07:53.000000000","updated":"2024-07-18 13:37:38.000000000","submitted":"2024-07-18 13:35:17.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"revert_of":923321,"submission_id":"924388-fix/disk-formats-tempes","meta_rev_id":"07a0f034b9376e122bd608a52a6100e5d2eff904","_number":924388,"virtual_id_number":924388,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-18 13:35:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-18 09:47:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-18 09:47:47.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-18 09:07:53.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-18 09:07:53.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-07-18 09:07:53.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"e8046c65f0e9b4e41d85039b32d52ecc49525d1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-18 09:07:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6152ee8946909305feb29f0f98e9868bcf62db8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 09:15:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2c179ffaa31943c99190b42c05888acc\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8e5b4cefc20b4b05a15d383f961023a7 : SUCCESS in 5m 58s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/52ffb7b816e24c4fa7d51e10938da627 : SUCCESS in 4m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07e5019521842da136d39b63e8dcfce0696fb581","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-18 09:47:47.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7d8c369ab710a5d6a8a80d319d7ff6ec47f12cd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 11:09:17.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12779fb148bc4c80b9e1375962315752\n\n- grenade https://zuul.opendev.org/t/openstack/build/10ab2039c2d64502a2a65c447325901e : SUCCESS in 1h 03m 15s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2ec5b2b12cb14faa8f14c5fd28b821ad : SUCCESS in 1h 28m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3eb3d75a4e04357acda90662db193fa : SUCCESS in 52m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba6b076fadfb415e8785b50e9e669d2a : SUCCESS in 4m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/01c2951bbea942af83219a0845b78bf1 : SUCCESS in 6m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3b1861a090241269c681ba833878770 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c98d8c7c55c340b49d33d766a14efbbb : SUCCESS in 10m 09s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb7f1da10614e7fbc610f704201c844 : SUCCESS in 6m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dfe81754366d4738908285016a6df598 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a129d52c5e254d10a6a1fe006153ae52 : SUCCESS in 8m 30s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/742234d44e244e7681fb18656df8753e : SUCCESS in 12m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1ae78a282bc44f8f9567a166deb7c6db : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8961e4a2f68c43aa830fd2217e1431db : SUCCESS in 1h 47m 42s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c08f7dcb2114b0192f21a95af0cdbe0 : SUCCESS in 1h 55m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9a86a6e76c594665af64b2ba0d7d53a9 : SUCCESS in 1h 50m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c1f6a551ca584de18be81d1c25eebe8b : SUCCESS in 1h 40m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/74c24def54ba4e6a82b457e6c115b58f : SUCCESS in 1h 55m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d2cddcd76c084adcaeb101f814184524 : SUCCESS in 52m 59s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b53bdb6de53443a88b7599ef6062279 : SUCCESS in 1h 35m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/870ff3e4e0c64d66bccaa84c648bbd76 : SUCCESS in 1h 53m 04s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6ef06ccbc3b94a3c90e31a6c6d7a5711 : SUCCESS in 50m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/fdeb79f7cee748289f268700707926bb : SUCCESS in 32m 57s","accounts_in_message":[],"_revision_number":1},{"id":"d95acf92904dc7b5ac87a75a9766d87ad4ecf45e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 11:10:13.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7f82598dafd89d211915d27a457a9214c07be0dd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 13:35:17.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a28d790af01449bea836f529a157b59b\n\n- grenade https://zuul.opendev.org/t/openstack/build/51abd6d82410409b92920fdc66618555 : SUCCESS in 1h 06m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e4417f36acba41c0a7b32a5e2ce95d82 : SUCCESS in 1h 13m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d52aa8b5457d4ddea0aef189053c8151 : SUCCESS in 56m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/872b4fa5471b40de8325387798a79525 : SUCCESS in 4m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db3e507d846042cbbae7212057541492 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/47050ac37bbc42fa920d6bc936f46e67 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4a66c6bb87b4c838c2970239c125ba9 : SUCCESS in 6m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3ce168afaffe4bf9b9c2d53c47d265b6 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/419caadedff644efa4fdba6d29fb1948 : SUCCESS in 9m 04s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e8cedb53773841088c2f15833dc11278 : SUCCESS in 12m 47s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/aa4d5b9089764bdc854eac4c2e1b830b : SUCCESS in 1h 58m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/28465a3004b145dc9183ce1991bb9f00 : SUCCESS in 2h 19m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a0886d7e832244f2b68891d9aa237ce0 : SUCCESS in 1h 38m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ab1887f3bb3b42de8c02dcb2be33f2d6 : SUCCESS in 1h 02m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/52253149d30b4bdf93a6572e8117fa9d : SUCCESS in 1h 35m 29s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/74f3e29f413b45218aa348b7eaa0e92d : SUCCESS in 34m 49s","accounts_in_message":[],"_revision_number":1},{"id":"42be0af4b9ef235e151d87d157da23e4f7932a5f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 13:35:17.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"07a0f034b9376e122bd608a52a6100e5d2eff904","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-18 13:37:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedc0661f47c4c818d86b86d3e26a484\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9273a80e8e984aa6a0d48b1fa25398e0 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"57f0c93d25ee3cfa0bb173f0064252ee8265df31","revisions":{"57f0c93d25ee3cfa0bb173f0064252ee8265df31":{"kind":"REWORK","_number":1,"created":"2024-07-18 09:07:53.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/88/924388/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/88/924388/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/924388/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/924388/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/88/924388/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/88/924388/1"}}},"commit":{"parents":[{"commit":"244ccb68dd9c1be8b8c6223c50998464a394b688","subject":"Fix import job to provide valid disk-formats list to tempest","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/244ccb68dd9c1be8b8c6223c50998464a394b688"}]}],"author":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-07-18 09:07:53.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-07-18 09:07:53.000000000","tz":0},"subject":"Revert \"Fix import job to provide valid disk-formats list to tempest\"","message":"Revert \"Fix import job to provide valid disk-formats list to tempest\"\n\nThis reverts commit 244ccb68dd9c1be8b8c6223c50998464a394b688.\n\nReason for revert: disk-format list not required to be passed to the import jobs after the tempest change [1].\n\n[1]: https://review.opendev.org/c/openstack/tempest/+/924252\n\nChange-Id: Id2548fad3e90144b47964b82bb4f702ad6407ae2\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57f0c93d25ee3cfa0bb173f0064252ee8265df31"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/57f0c93d25ee3cfa0bb173f0064252ee8265df31"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
