)]}'
{"id":"openstack%2Fglance~924824","triplet_id":"openstack%2Fglance~master~I0409982ae27b662e60dd2363ba2f7863d0722fea","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-24 17:15:07.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I0409982ae27b662e60dd2363ba2f7863d0722fea","subject":"Do not set_acls if store is not associated to glance node","status":"MERGED","created":"2024-07-24 07:34:57.000000000","updated":"2024-07-24 17:16:21.000000000","submitted":"2024-07-24 17:15:07.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924824","meta_rev_id":"e8162a35f8c2ac3e378e6afdeef8a42d4179f501","_number":924824,"virtual_id_number":924824,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-24 17:15:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-24 10:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-24 10:01:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 07:40:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-24 09:57:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-24 10:01:20.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"fbaa1981f608d79e22e5dabb6a7530a467f924da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-24 07:34:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"93fafd0cbc017040d3f147ee32f2600e7c20d7e5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 07:40:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/700abb9151c54d2f94c7d7874b78eab2\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/08d351cd62d943c7a8e08d8432ecb64a : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3b6bcdb92a27435a92b2f9159ac6d935 : SUCCESS in 4m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f5400ec8048600d91d90d6a38902a5d068a67588","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 09:57:51.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f31d1fd51ef4127ba57db7ea57b86f6\n\n- grenade https://zuul.opendev.org/t/openstack/build/e103942c64ee4afe97eba91f427d82db : SUCCESS in 1h 06m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/162bae3f186046f9a6ee2ad860a1b1b7 : SUCCESS in 1h 37m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0422611357c34c2a88632730ebbaf3c6 : SUCCESS in 47m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da543612463346c6bdfdd7dcdca81af2 : SUCCESS in 3m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cda0f92ebba4dbcba87e1756b714e69 : SUCCESS in 5m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1457cb333d34b168728cc3f90c2d19d : SUCCESS in 4m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/34b29b6f59794f7395d66c8f477e57f9 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4e33feb1e5043caa88f66b098fb517e : SUCCESS in 6m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da70a05fbf9c48e2a562f0e34c186499 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac0cad2cd73d41c9b6d1fce95321a527 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a626dae2fa5340b5bcbe77be4ca99ab6 : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/05712c0b67da41cc9bd2f4d6fa7a1430 : SUCCESS in 14m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/510a3fe78bd04bd6aad5dab620ff018f : SUCCESS in 9m 29s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6146b7662c0446a0bcbd8107dad56589 : SUCCESS in 1h 45m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87be03a5e6974894a985c36c4860a200 : SUCCESS in 1h 53m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4477a3f1dc1d4eb6981e9ce5b44504c7 : SUCCESS in 1h 43m 12s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f7b83d55990c4645b49325a2b2ac5fdf : SUCCESS in 1h 47m 48s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8dc7eb5b0cd940dd97ad313f9af3edba : SUCCESS in 2h 15m 38s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/767e8e13255649f0ae098b8dbef01a17 : SUCCESS in 54m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/897cb9e7663047f68860151223fb590d : SUCCESS in 1h 49m 58s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ec2e2807aa6c4bb8b91c7cd212eee1ac : SUCCESS in 2h 13m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/32379eb98de44f11a356d2539e0db911 : SUCCESS in 1h 06m 05s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5fef026aa6ca4b1a98ef9814bf06a528 : SUCCESS in 33m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8b66ce9eb0c1a2c1cb5b2a6ed6310b389a435f10","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-24 10:01:20.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d1c21717b153147ee5e8490228215b68ee12204d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 10:01:56.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2f9db9613e9e8a46a3cd1971fbcdfca94f3ed6aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 11:55:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9adbc234961d42c9ba27e4945522f0f5\n\n- grenade https://zuul.opendev.org/t/openstack/build/cfd9e803d14c4c538ecc4883d3d9ae8a : FAILURE in 34m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/51916a2d6e81496eb4f8d9fda5381e84 : SUCCESS in 1h 30m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/746f99cdfb324044a13d2178cbfe4257 : SUCCESS in 1h 00m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fb28b072c22c4748b578c6571aa2ae18 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5c9d53751a8c4888b6da5b54077f5103 : SUCCESS in 5m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45badb18f0d748fa9cb5cbf78ebbc40e : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1cd2fd64145747fa823dc95a064da38a : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f99a9fd202b4f08b853e7657bc28ae6 : SUCCESS in 6m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/56f3535e26bc4999a93c49c27fa6f6db : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/aa301261576447dba64cda702be57330 : SUCCESS in 7m 50s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/79b07dc7e88b44ee8d77f1016b2265ba : SUCCESS in 12m 44s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/850ec7fd2b14465e9b99770720d44a1d : SUCCESS in 1h 40m 16s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d4defcfa2f7241e98a8299814597a724 : SUCCESS in 1h 52m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/31e0ebfa07eb41ef83dfdb3927f3b92a : SUCCESS in 1h 45m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d47f06c7f3bd4f588cace6948728a3a2 : SUCCESS in 48m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/375109d141bd44a3882d9694773573b6 : SUCCESS in 1h 39m 09s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cb32bdc62abd4859852e27f20faab10e : SUCCESS in 28m 31s","accounts_in_message":[],"_revision_number":1},{"id":"b023724931ec4b5a66a42a7d493a6d330c4d7057","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-24 12:54:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"be0abbe169d82e466dcbf121d3519cb067aae356","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 13:01:20.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/022c791bb0af4db38d4d3d1049828f07\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/453773e43b34463590f939ffc97cc1bd : SUCCESS in 5m 50s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1eb8dbb0ce59453e97a4890ef03a4f16 : SUCCESS in 4m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ff78a1c6f3222c67ea148d4ac702c789c10aae3f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:58:11.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629553a726ae4439a390e953606df933\n\n- grenade https://zuul.opendev.org/t/openstack/build/f822c8e7aa0948e1b5738e26341c4312 : SUCCESS in 1h 10m 48s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/37d23466abe24a5aa912e22f97a92bb6 : SUCCESS in 1h 43m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dddf0a08e6e14371aeab7fd424a47042 : SUCCESS in 1h 00m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ff03f509b4624369aaf330c2000ab48b : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e408f0fe340b40bdb84aec7e28b58861 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c52ca32579074ce6934b9d2337433c80 : SUCCESS in 4m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1b81880df4ad4fdbbc92c24dcd8fd257 : SUCCESS in 7m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8c337d7fd4d49dd8217de5ef6b19659 : SUCCESS in 5m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e864e3eeb094e0ea6e7d98cf04ebfa9 : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/336e4d687b994a2bab5f50b91bafe277 : SUCCESS in 9m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/715d16f0fcec45d181a9859d08d397c1 : SUCCESS in 8m 35s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92b37421153546a2975c1394acf866c1 : SUCCESS in 11m 43s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c541c5835457445290157d6abb974b89 : SUCCESS in 9m 19s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/316ea0fa70a24b0d978e2a20aa17f98a : SUCCESS in 1h 46m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/edc9a742fd334410980aee5ab9f33476 : SUCCESS in 1h 38m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/40ac53701e904283adc5400fd0252136 : SUCCESS in 1h 42m 02s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/8228a59cb64741e6bc19da27227e263e : SUCCESS in 1h 48m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9dcceef8a1e64551bdffeb58fbca42de : SUCCESS in 1h 58m 00s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f39c0a25e9d246c696033a773d19a5f0 : SUCCESS in 50m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2389ffe361814ae79532b2e0805ec197 : SUCCESS in 1h 54m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/0c4db5adc613451485806424f773dd7e : FAILURE in 1h 40m 58s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2aeae760aa654d88a4478df0bb0f9dcc : SUCCESS in 1h 06m 41s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/16b5ff0ae084435db7ac860fd61d68d1 : SUCCESS in 34m 47s","accounts_in_message":[],"_revision_number":1},{"id":"051e2891970b6fbd8547ce28422881383b86002a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 14:58:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6901af8a0830d2cf3d51d554671f09a95602f55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:15:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ff409b5c8324ee2b9274f03fa0d3a3a\n\n- grenade https://zuul.opendev.org/t/openstack/build/9a0e819871da4944a055e03e20a376e7 : SUCCESS in 1h 03m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0dc0deee47154f34a8db06a5aa291457 : SUCCESS in 1h 36m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b37d40ac9b93472c9ce8945a01da6da4 : SUCCESS in 58m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef13d5c06e6b45d1826da0c9335d3ba1 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed392add3d5d41619d2d3999ece1fe6f : SUCCESS in 6m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5871446f7b5448ca82d852f2ae74a930 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a346f14b68d3472d8913a69968a18451 : SUCCESS in 7m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60efb650ef8b48b0baa93b4342a561e1 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3c5a244fdcd94d6fbf077fcb4c0bca3e : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a89a9c5afcc348b28ebf341e81637939 : SUCCESS in 8m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/42fd5ee1b58045778c4074007c3a60b9 : SUCCESS in 12m 30s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bb5d846b0a394b369e8df2fac2bc99cc : SUCCESS in 1h 34m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e395e32371ef4f58a25e32d194d2dd8a : SUCCESS in 2h 09m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/74e8caf8350f4e4c99b46ac196264dff : SUCCESS in 1h 54m 07s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4503a049729d41a0b6d95ee63e5b2972 : SUCCESS in 1h 00m 58s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ec7ac4ca6c89441fb31785608003dda8 : SUCCESS in 1h 32m 12s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8262fec19fad44c69c5a0ed78a013954 : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"9fc38c6633feff4e0823f963fdb199cfe71756df","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:15:07.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e8162a35f8c2ac3e378e6afdeef8a42d4179f501","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:16:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/362eab2eeecc49b2ba62ca000310acfc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/edff62bc95e34951b0dcd9b087d2e567 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc7c62aa5c7343d2b5f2c9eda12546a6 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ee7e96f06af741bb34bedac18fa2c4616fcc3905","revisions":{"ee7e96f06af741bb34bedac18fa2c4616fcc3905":{"kind":"REWORK","_number":1,"created":"2024-07-24 07:34:57.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/24/924824/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/24/924824/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/924824/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/924824/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/24/924824/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/24/924824/1"}}},"commit":{"parents":[{"commit":"38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce","subject":"Imported Translations from Zanata","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/38a2828e8f25d1cc43f300eaa0bb2cb8144d4dce"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-24 07:31:46.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-24 07:34:30.000000000","tz":0},"subject":"Do not set_acls if store is not associated to glance node","message":"Do not set_acls if store is not associated to glance node\n\nIn case of glance multiple stores (mostly for ceph) nova initiates\ncopy-image functionality if image, from which the sever is being\ncreated, is not present in the referring ceph store. This can fail if\nimage location which is already present in image but not available for\nthat glance edge node. This scenario can only be reproducible\nwith EDGE deployment.\n\nIn order to fix this, if the store is defined on that glance node\nthen only call set_acls method, else ignore it.\n\nCloses-Bug: #2073945\nChange-Id: I0409982ae27b662e60dd2363ba2f7863d0722fea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee7e96f06af741bb34bedac18fa2c4616fcc3905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee7e96f06af741bb34bedac18fa2c4616fcc3905"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
