)]}'
{"id":"openstack%2Fglance~924871","triplet_id":"openstack%2Fglance~stable%2F2024.1~I0409982ae27b662e60dd2363ba2f7863d0722fea","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-07-24 19:41:24.000000000","reason":"\u003cGERRIT_ACCOUNT_27615\u003e replied on the change","reason_account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}}},"hashtags":[],"change_id":"I0409982ae27b662e60dd2363ba2f7863d0722fea","subject":"Do not set_acls if store is not associated to glance node","status":"MERGED","created":"2024-07-24 17:20:24.000000000","updated":"2024-07-25 00:09:18.000000000","submitted":"2024-07-25 00:08:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924871","cherry_pick_of_change":924824,"cherry_pick_of_patch_set":1,"meta_rev_id":"2d6f24e88a7256796ff9debcdbf0752fffe3ba61","_number":924871,"virtual_id_number":924871,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-25 00:08:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-07-24 17:25:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-07-24 17:25:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-24 17:25:42.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-24 17:26:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-24 19:31:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"94aa33d7f1b75daeb2da8097f8f0e29dc9d6d791","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-24 17:20:24.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"cf943065d3884312f8042bfd08dc231ac4475078","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-24 17:25:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ce3f05ea54e67c7a7c5cb67083ec3aeba74f179c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 17:26:50.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8b94c71397848a88493b4a9dbca4860\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9ed2ebdf018046c78d6e31b7bc05fa92 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7b60e6a146bf4236b04c104beb76b298 : SUCCESS in 5m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"29c0b418f15bd86d03cf716c25c892ae51cc926d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 19:31:50.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/40956859ed54440e8fb656b859a70043\n\n- grenade https://zuul.opendev.org/t/openstack/build/299b515b5c8b41c3abaffcd4761171f2 : POST_FAILURE in 45m 32s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/98d188fcabc740abbe411589dc8c5d6a : FAILURE in 58m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/682707d0e16f43038b6345b8dd3c0365 : FAILURE in 51m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/46a222481131494c9bd87009431c585d : FAILURE in 1h 13m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/adbc5d082e6c4b20a788a16f39777696 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa402563ee7941a4ab032f0db723d0bc : SUCCESS in 7m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/405474eeda8b4483869d234c15cc3cee : SUCCESS in 6m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eb4e555b0b94bb5a26457f64753bb50 : SUCCESS in 8m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97dbd2a4166047d5ae73e158e6881770 : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db15e4d59c4d4671aa2a18dbf4e6d651 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/19ec70218cd94020bc840c8a4b51e814 : SUCCESS in 10m 12s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/45392b8d55e44c7c873d21a6dcd28f49 : SUCCESS in 8m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/fe665928e00945b0aec7cb97c8e51925 : SUCCESS in 2h 01m 13s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/797a7df4d30549279d0e2bce641300b7 : SUCCESS in 1h 51m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/11b79329ba104be0ab54a4500f80c256 : SUCCESS in 2h 09m 00s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/34db3332f7404a448cfb0b8a2d3fab5a : SUCCESS in 1h 51m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/963636dcac39455797a418a0dce74428 : SUCCESS in 2h 10m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61cc8a7b31b041e5a6df8f327f0b3f43 : SUCCESS in 1h 01m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4198e97819524712ad5ca886ccf8debd : SUCCESS in 1h 54m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/7410d7facf3e4ce1a3ae5a0137523307 : SUCCESS in 1h 56m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/86eeeb56c1d4467db54cb78c20db3a85 : SUCCESS in 1h 11m 49s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6fcadb504011420c8414733690afe2d6 : SUCCESS in 38m 02s","accounts_in_message":[],"_revision_number":1},{"id":"a9f68ed59ccdd1190620bab8ce5ddfa8d5ee25e8","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-07-24 19:41:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5df13d6c0bb018e8c19e8cace2ac33b67fc9c8b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 19:56:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7bf7c5db3ec1451e8391c9d931e17cda\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/fc8f901b65e84b0fb4ef9080d0766ceb : SUCCESS in 12m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d06d3a43f8b8410fb1c0e767132e28f6 : SUCCESS in 8m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"16f843d2e17c49d8526355ca14d2bec531b60377","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 21:50:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc24a42195b54aca99e4a66914d17fba\n\n- grenade https://zuul.opendev.org/t/openstack/build/245d85ece64f41938be15b3da1f0af90 : SUCCESS in 56m 13s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0f86443fceb144a29d8a802f7bbe87eb : SUCCESS in 42m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/687b109e39794277be1e4230e054d1b0 : SUCCESS in 1h 34m 14s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/eb9c9792c10545a88e18dea0ded8ff2c : SUCCESS in 1h 02m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebec3dcb16fe4df6a8797069cc34eb78 : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/196d91a188884afc8d6145f446e8b1ac : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ba743a6aaff048ce849923f851220904 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25da401f5409456d84d6e6216bb565f1 : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86f7a5d42851435eae1844318bb70c0d : SUCCESS in 6m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/aec552d3f84743359950be0944409da7 : SUCCESS in 8m 43s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ffd8531d7b594295aaecab697682f811 : SUCCESS in 8m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fa7b5c507f1843c2bdb52d6086398145 : SUCCESS in 8m 24s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c9cc6dedb8734b3481976c5e60e13373 : SUCCESS in 2h 03m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/750da074953f48d5b82e9c149ef49b89 : SUCCESS in 1h 35m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/19aa48800e444eaeae1ebfaf9c31b5af : SUCCESS in 1h 41m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d988ced5c651438e847ca79cb365e3a1 : SUCCESS in 1h 20m 21s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/b5a487f78859491aab8f81934ce39b10 : SUCCESS in 2h 02m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/9895249aacca40a78cbbc355c3bfecea : SUCCESS in 59m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c9ccf990deab4b46bb885c516ba78471 : SUCCESS in 1h 41m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/99970f818a1242169824048f732ca7c8 : SUCCESS in 1h 16m 44s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3bb1ac42d1f341a8958520ca2d6e574f : SUCCESS in 1h 10m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d051a6b8164b427488478e52d0291ee3 : SUCCESS in 35m 45s","accounts_in_message":[],"_revision_number":1},{"id":"68f2e981e594362103a224d096d0c703c339cf25","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-24 21:51:58.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6e56d5d87d4d56f3178c0c2290aa258aaff552ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 00:08:07.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5600a1c8a004b93b5137efdd779f7b1\n\n- grenade https://zuul.opendev.org/t/openstack/build/f19eb27d840247f4b74cc06056afe556 : SUCCESS in 42m 00s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5686141151dc4f638cfef92f8881e1b5 : SUCCESS in 40m 06s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/414b589cc14441eeae0ff455e02177f6 : SUCCESS in 57m 05s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6f2147c053241ed8fd08687b8bc8ffc : SUCCESS in 41m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/64b8787b62144c1297cb5a24c4eb25d4 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5137972ecdff4ed785b7c77eb8cee153 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cc1fb3f247434be3a354d4db560a9983 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6a704319ddc4966ac20a802d379ccfd : SUCCESS in 6m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d6872d163ebc4b08bd7b95e48e55d911 : SUCCESS in 5m 01s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/54f691ddf7a7468abe4b7a691251d770 : SUCCESS in 8m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1afd816cf7bc4a0283b6f5686f1acac1 : SUCCESS in 8m 34s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c4cb7b6941f946ca9f9e501d6a0e8222 : SUCCESS in 1h 25m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bb8a9beaafe1441e8b41a278d29d7452 : SUCCESS in 1h 47m 56s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/566f6b55f8ea42eaa4ae817a55b303ed : SUCCESS in 2h 10m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/1a2b800a16fb48439db81eea48f94bb7 : SUCCESS in 43m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf0e7fb46adb4b2694dc4b5e63e6d004 : SUCCESS in 1h 42m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db78a9bedc654f01ad171594c83072e0 : SUCCESS in 35m 19s","accounts_in_message":[],"_revision_number":1},{"id":"778d60637eb5c4e803dc6403637c30df100825b1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 00:08:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2d6f24e88a7256796ff9debcdbf0752fffe3ba61","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-25 00:09:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/324b9c9c7add4bf1964d07e9eed99b99\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b404d34b430d49a199303994de3863ae : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ce4dd1858a2a4af7b9ac04357961f24e : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"aec654ff2cbe75ef682eafa1d83906f18fd48c93","revisions":{"aec654ff2cbe75ef682eafa1d83906f18fd48c93":{"kind":"REWORK","_number":1,"created":"2024-07-24 17:20:24.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/71/924871/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/71/924871/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/924871/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/924871/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/71/924871/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/71/924871/1"}}},"commit":{"parents":[{"commit":"b06fbcc509795e96eee1159c7399c3a2babd6a4b","subject":"Fix /healthcheck url typo in glance-api-paste","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b06fbcc509795e96eee1159c7399c3a2babd6a4b"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-24 07:31:46.000000000","tz":0},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-24 17:20:24.000000000","tz":0},"subject":"Do not set_acls if store is not associated to glance node","message":"Do not set_acls if store is not associated to glance node\n\nIn case of glance multiple stores (mostly for ceph) nova initiates\ncopy-image functionality if image, from which the sever is being\ncreated, is not present in the referring ceph store. This can fail if\nimage location which is already present in image but not available for\nthat glance edge node. This scenario can only be reproducible\nwith EDGE deployment.\n\nIn order to fix this, if the store is defined on that glance node\nthen only call set_acls method, else ignore it.\n\nCloses-Bug: #2073945\nChange-Id: I0409982ae27b662e60dd2363ba2f7863d0722fea\n(cherry picked from commit ee7e96f06af741bb34bedac18fa2c4616fcc3905)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aec654ff2cbe75ef682eafa1d83906f18fd48c93"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/aec654ff2cbe75ef682eafa1d83906f18fd48c93"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
