)]}'
{"id":"openstack%2Fglance~924991","triplet_id":"openstack%2Fglance~master~I7b760d2b28a8b289a303a0a9c1d91a9de0c7138a","project":"openstack/glance","branch":"master","topic":"remove-location-strategy","attention_set":{},"removed_from_attention_set":{"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-07-31 18:50:00.000000000","reason":"Change was submitted"},"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-07-26 07:33:57.000000000","reason":"\u003cGERRIT_ACCOUNT_19138\u003e replied on the change","reason_account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-07-26 07:36:29.000000000","reason":"\u003cGERRIT_ACCOUNT_33765\u003e replied on the change","reason_account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-07-31 18:50:00.000000000","reason":"Change was submitted"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2024-07-31 16:44:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I7b760d2b28a8b289a303a0a9c1d91a9de0c7138a","subject":"Remove location_strategy functionality","status":"MERGED","created":"2024-07-26 06:34:33.000000000","updated":"2024-07-31 18:51:08.000000000","submitted":"2024-07-31 18:50:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"924991-remove-location-strategy","meta_rev_id":"009b97d851d1d641a4b491425e6be62b14d04fca","_number":924991,"virtual_id_number":924991,"owner":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-07-31 18:49:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"recommended":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"all":[{"value":1,"date":"2024-07-26 07:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-07-26 07:36:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":2,"date":"2024-07-31 16:44:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-07-31 16:44:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"recommended":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":1,"date":"2024-07-26 07:15:36.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-26 06:35:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"},{"updated":"2024-07-26 06:35:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2024-07-26 06:35:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-07-26 06:35:17.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-07-26 06:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-26 09:15:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"4feea2c6dd92f84ecdbd9a33dacb54517ed19b2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-26 06:34:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"37c22a25d07f7fb7b7f177985edf588d0b68016a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-26 06:35:17.000000000","message":"Patch Set 1: Review-Priority+1","accounts_in_message":[],"_revision_number":1},{"id":"9a4625a60787ddbc15b9386b2773dcc5fa5a103c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 06:40:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/00a8c04cbf404d8f9ef938783b316bac\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d01846099b6842c3b30e38d692c06fc8 : SUCCESS in 4m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/64261cad872d4dd998533069b0972973 : SUCCESS in 5m 05s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"778227951f9d6cf2f68ec524069df6f14a1f6e60","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-26 07:02:53.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a9cacca9ad4446bc1904fc4ab6b7f63ffee2f1c7","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-26 07:09:13.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"06b2ad6c340caded5d1a37be44a52e6143e82328","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-07-26 07:15:36.000000000","message":"Uploaded patch set 2.\n\nCopied Votes:\n* Review-Priority+1 (copy condition: \"changekind:NO_CODE_CHANGE OR **is:ANY**\")\n","accounts_in_message":[],"_revision_number":2},{"id":"4d9862d04088ccfcf1b49cf3436e76e37c2f9558","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 07:21:41.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/59dbe7d4f7a647b6b057d0ae1933d19a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/76355021b9564013bc95e2ff94835046 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/ab3b2ec814e342d9a2f4de46ff5ee246 : SUCCESS in 5m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"39563809194f90f269a6d1a5f2ba7791cfe61a6a","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-26 07:33:57.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"fb9fbdd3254a53527e932f34c5586ab3fea98fe6","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-07-26 07:36:29.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"28617702e9ef0d08109cdb4c3e2ed5e5d18c809e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-26 09:15:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cb421ae4ec04b6a89bdb23b937ad9bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2180c70e8cab45fd9507cd417ee3e205 : SUCCESS in 1m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/f4b070b9f6304111a6be95fe8eadb66b : SUCCESS in 41m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ca9628aeb9a490abb0ad4ef9dfae535 : SUCCESS in 1h 18m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b5aa1664c834425b52b7ed912cd29e2 : SUCCESS in 42m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9232ec84bc0f4ea1b8ca16a0bcf0f35b : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480b3125f2d541d2940b0b67030a3ae9 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea90e0bb094b411a994d0af75438b7b0 : SUCCESS in 5m 30s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f08c48f49b246f3b2878ecd6d0ad4fd : SUCCESS in 9m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44f2c78bb8e1486a97310d3b778334a8 : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6fdab32444e421aa27dcf28eb20abed : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ea7edc005604958b85c2f66173bdd26 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/33b1a68e310b422fb1b5ef5099dca733 : SUCCESS in 8m 18s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/690dcf5b39ce4f2a9d452d97f94e2ef6 : SUCCESS in 13m 25s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/bb1f1b55c93c4a10abad9f5a8639a43d : SUCCESS in 7m 48s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ac2709f5fbcd4af189df992851bfa73b : SUCCESS in 1h 36m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/070f12aa98f0450ebf2dfa0371f2cece : SUCCESS in 1h 39m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/894e8f1befe04d448ff89e642e3ab5a8 : SUCCESS in 1h 46m 30s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/73edca88c84b445eb932a720233406cf : SUCCESS in 1h 47m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a36c8c82f8b84c5189b0685871c29d07 : SUCCESS in 1h 59m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/af0c11cab1714655ae0b8237bce213d4 : SUCCESS in 34m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ae747f0fd984be08d929ce5644fc0fc : SUCCESS in 1h 48m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/67c41c6299a4443a861df123df77037a : SUCCESS in 1h 39m 53s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/6f7d76bcaf0c4d2682fb5a6027e0a571 : SUCCESS in 1h 11m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f0c6475b4415473299a368f7d69358f5 : SUCCESS in 37m 14s","accounts_in_message":[],"_revision_number":2},{"id":"accef93420eb978088f2a62950b45b7375581b40","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-07-31 16:44:12.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"182c358b5f3a45a00db160fb4a9d2c6a985f913f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 16:44:55.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"711511b01334af0958b2dd0bc38ea5ecf647352b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 18:49:59.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1eb2ee21b5a4e69879f16464cac4953\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/72d0214d7b334de5bda59e1f179ac61d : SUCCESS in 2m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/01b158ff2df9489e937a98fc67a82b5e : SUCCESS in 58m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/648582aa96d643d1a5f8ff96c507cc40 : SUCCESS in 1h 22m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c9fe023c602a478fb191393301b10898 : SUCCESS in 55m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb27e20f14c14c6e94ec672d6e49c73a : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dd0c3432d51f4e4e8b0f9f7a8a4ae16f : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/29f3b11169e74e2e95f59461ddc07675 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91abe05bd8cd4d79803706ab56097dba : SUCCESS in 6m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba56d935a11b403d819466e1b08ea85c : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6e42887c8cc54e0bb71872c9e268efe8 : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/70eafad0d497489c8ead76c6ed9f816a : SUCCESS in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8db98b659c10431fa73b04a5cb9d1056 : SUCCESS in 13m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3d98b133af9f4174a851a76ee2c40e9a : SUCCESS in 1h 53m 31s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cca730ba79874b4fa5bcae0b1c133155 : SUCCESS in 1h 59m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e18acf0b17e84348bb76e2fe1b2c9ae7 : SUCCESS in 1h 59m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6338f7dd956242d7957afca763c17508 : SUCCESS in 1h 02m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/409ab234dc1849bfacbf14874434fff1 : SUCCESS in 1h 41m 05s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/996c3e79a53540e78f0594ce6f4ccfb6 : SUCCESS in 35m 06s","accounts_in_message":[],"_revision_number":2},{"id":"c08007046ceac5a3877c483c06152b451deba0c1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 18:50:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"009b97d851d1d641a4b491425e6be62b14d04fca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-31 18:51:08.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/02107750cf3c4050ad402623dcf81101\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ca3747e6759467f8ba562cfbbd068af : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03726f0e3f8446e59aae3afcfdc30029 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5d5f570f46d2265fa350a19e182b3079df54db7f","revisions":{"493571650d7fac3829548309e277dea66ecd917e":{"kind":"REWORK","_number":1,"created":"2024-07-26 06:34:33.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/924991/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/924991/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/924991/1"}}},"commit":{"parents":[{"commit":"ee7e96f06af741bb34bedac18fa2c4616fcc3905","subject":"Do not set_acls if store is not associated to glance node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee7e96f06af741bb34bedac18fa2c4616fcc3905"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-26 06:19:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-26 06:33:20.000000000","tz":0},"subject":"Remove location_strategy functionality","message":"Remove location_strategy functionality\n\nRemoving deprecated location_strategy functionality which was used\nto prioritize location order for downloading the image.\n\nThe weighing mechanism introduced in the Bobcat development cycle\nnow should be used by operators who would like to prioritize\ncertain stores over others.\n\nChange-Id: I7b760d2b28a8b289a303a0a9c1d91a9de0c7138a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/493571650d7fac3829548309e277dea66ecd917e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/493571650d7fac3829548309e277dea66ecd917e"}]},"branch":"refs/heads/master"},"5d5f570f46d2265fa350a19e182b3079df54db7f":{"kind":"REWORK","_number":2,"created":"2024-07-26 07:15:36.000000000","uploader":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"ref":"refs/changes/91/924991/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/91/924991/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/91/924991/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/91/924991/2"}}},"commit":{"parents":[{"commit":"ee7e96f06af741bb34bedac18fa2c4616fcc3905","subject":"Do not set_acls if store is not associated to glance node","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee7e96f06af741bb34bedac18fa2c4616fcc3905"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-26 06:19:16.000000000","tz":0},"committer":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-26 07:14:19.000000000","tz":0},"subject":"Remove location_strategy functionality","message":"Remove location_strategy functionality\n\nRemoving deprecated location_strategy functionality which was used\nto prioritize location order for downloading the image.\n\nThe weighing mechanism introduced in the Bobcat development cycle\nnow should be used by operators who would like to prioritize\ncertain stores over others.\n\nChange-Id: I7b760d2b28a8b289a303a0a9c1d91a9de0c7138a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d5f570f46d2265fa350a19e182b3079df54db7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d5f570f46d2265fa350a19e182b3079df54db7f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
