)]}'
{"id":"openstack%2Fglance~925055","triplet_id":"openstack%2Fglance~stable%2F2023.1~I0409982ae27b662e60dd2363ba2f7863d0722fea","project":"openstack/glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-01 11:32:08.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-08-01 09:34:14.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I0409982ae27b662e60dd2363ba2f7863d0722fea","subject":"Do not set_acls if store is not associated to glance node","status":"MERGED","created":"2024-07-27 06:26:22.000000000","updated":"2024-08-01 11:33:19.000000000","submitted":"2024-08-01 11:32:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925055","cherry_pick_of_change":924905,"cherry_pick_of_patch_set":1,"meta_rev_id":"86530eac4f14b02e23d6083f4a8f579addbf1911","_number":925055,"virtual_id_number":925055,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 11:32:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-07-30 08:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-08-01 09:34:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-08-01 09:34:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-07-27 06:26:22.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2024-07-27 06:31:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-07-27 08:19:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-07-29 11:23:11.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-07-30 08:02:17.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"621dcf80d93dbcbd42c7d54fd5d96caa76c840aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-07-27 06:26:22.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"b8d49acb3ec21c46209456e2a794f0db9f88af04","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 06:31:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5cae92247f8471b8f19069747c7eb60\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3588ad1cf29144be9d7e99cc73232dcd : SUCCESS in 4m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6bc1d07602abc663cf8ce1d13a5f4ef0fd484820","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-07-27 08:19:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc91f413d4f849e2904a44d35c453b45\n\n- grenade https://zuul.opendev.org/t/openstack/build/378d080ed0224734a5a0e771dd395759 : SUCCESS in 37m 56s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/32301ed9768542079a03ebaa5832de8d : SUCCESS in 39m 14s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/86952235169e4715893fb7a36616a564 : SUCCESS in 55m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c91887832db44c8a27972c66a67012b : SUCCESS in 38m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81a26960e90c47f2925bfe42a8a056a8 : SUCCESS in 3m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4278025f631e44d5a92bd99141516658 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8c973a9532a45818a31c1b3a0286cef : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6e71f9eeffaa48d68d18e8a686fd1d12 : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f47e5e320be496480cbd70fce5760f4 : SUCCESS in 7m 52s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/658a40e45fef48b4b1fdd6ffe96d272f : SUCCESS in 8m 58s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c0fa8e747bb84ae2ac33ffdd56ad1a35 : SUCCESS in 1h 46m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/240a61ac59094ec1b541f2ddc786e1e8 : SUCCESS in 1h 10m 34s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90dd2d78626b4410a60d62b89e5535d8 : SUCCESS in 1h 23m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5b63762fe5f844a894543d3b78b5685f : SUCCESS in 1h 28m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3a187962330649dbb5484ad27fb53a62 : SUCCESS in 1h 45m 50s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94fdd784be8f435dbbe9b4e9373e7922 : SUCCESS in 52m 14s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/38cf11a89b8747e0917b7e886118fede : SUCCESS in 1h 33m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/37e7970a17d245fb8c192a474c2f1733 : SUCCESS in 38m 08s","accounts_in_message":[],"_revision_number":1},{"id":"37e6f517f4ad38729de2702416967a07a4931e7b","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-07-30 08:02:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"642328c186c3cc7489b8a488ac9e1a3425f8548b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-01 09:34:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7830c47fede9e7061c6b47587c336fb8f03f3285","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 09:34:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"28f7b87f23e67d32a0f5d3405896f30a16e7ba6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:32:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/97c5f5c7761f41df85fe3e154801268a\n\n- grenade https://zuul.opendev.org/t/openstack/build/a7c88e4a4d09467aa61b4adf72cce8b7 : SUCCESS in 50m 55s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/4cbeca1529934143b0a8f49c2ac012d6 : SUCCESS in 1h 02m 09s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fae5e3b0fa124f31be83aa7c2b5e5a8f : SUCCESS in 1h 20m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f9693901eb9a4612a5b1f23f82e1aa9d : SUCCESS in 1h 01m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0652c724db9e4d55a95c040c11a6f54b : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1a54923c6c214ec9b98998cf65e931fe : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7db2d4d18c44ff5a4f267806e8e5ede : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2102b7eaf31e48e9accaa5e1ea321be0 : SUCCESS in 6m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/39244bcdd755472493f39b5991c60df6 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/395cbe658c5f41f3ba6754edf7d95a78 : SUCCESS in 8m 18s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7144c5a66ea8449b9b1ec41faf15bd1f : SUCCESS in 1h 15m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b287ae18aaa74e718ae2ed5db2c1e010 : SUCCESS in 1h 55m 36s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1ce5ffd69a5b4deeb1b60bfa0bf3af26 : SUCCESS in 1h 56m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bd6870736114e21bf6f4f788df41aa0 : SUCCESS in 48m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07c7c0d41d1844e6804ad78c7ca1244b : SUCCESS in 1h 37m 24s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ead7f76082e547b59f425745045f9ed7 : SUCCESS in 24m 17s","accounts_in_message":[],"_revision_number":1},{"id":"e1c4a485a2aa9cc873e11c5f84b977a7e1a41cfd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:32:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"86530eac4f14b02e23d6083f4a8f579addbf1911","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:33:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/305f115305bf43d380c7ae77c2c01ef8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/278e57dffb82439ba83deb6841f70198 : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/531ad3f31cfb4428ab3ba1e1921c76be : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fbe4984ec1e34e95c6eda8d6874b7e465b7f070","revisions":{"6fbe4984ec1e34e95c6eda8d6874b7e465b7f070":{"kind":"REWORK","_number":1,"created":"2024-07-27 06:26:22.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/55/925055/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/55/925055/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/925055/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/925055/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/55/925055/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/55/925055/1"}}},"commit":{"parents":[{"commit":"edd75b70a4541c5dd89715a75ad565692dca2efa","subject":"Add releasenote for CVE-2024-32498 fix","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/edd75b70a4541c5dd89715a75ad565692dca2efa"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-24 07:31:46.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-07-27 06:26:22.000000000","tz":0},"subject":"Do not set_acls if store is not associated to glance node","message":"Do not set_acls if store is not associated to glance node\n\nIn case of glance multiple stores (mostly for ceph) nova initiates\ncopy-image functionality if image, from which the sever is being\ncreated, is not present in the referring ceph store. This can fail if\nimage location which is already present in image but not available for\nthat glance edge node. This scenario can only be reproducible\nwith EDGE deployment.\n\nIn order to fix this, if the store is defined on that glance node\nthen only call set_acls method, else ignore it.\n\nCloses-Bug: #2073945\nChange-Id: I0409982ae27b662e60dd2363ba2f7863d0722fea\n(cherry picked from commit ee7e96f06af741bb34bedac18fa2c4616fcc3905)\n(cherry picked from commit aec654ff2cbe75ef682eafa1d83906f18fd48c93)\n(cherry picked from commit 9e28810cc57a5a38bc5793a7855ef34bd5e3120c)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fbe4984ec1e34e95c6eda8d6874b7e465b7f070"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6fbe4984ec1e34e95c6eda8d6874b7e465b7f070"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
