)]}'
{"id":"openstack%2Fglance~925436","triplet_id":"openstack%2Fglance~master~Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e","project":"openstack/glance","branch":"master","attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2026-06-25 01:58:20.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"37206":{"account":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"last_update":"2024-08-21 14:44:23.000000000","reason":"A robot voted negatively on a label"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-08-21 12:31:39.000000000","reason":"Vote got outdated and was removed: Code-Review-1"},"26286":{"account":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"last_update":"2024-08-01 07:09:49.000000000","reason":"\u003cGERRIT_ACCOUNT_26286\u003e replied on the change","reason_account":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e","subject":"Fix image name computation when deleting an image","status":"NEW","created":"2024-08-01 03:17:11.000000000","updated":"2026-06-25 04:15:01.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"8d4324b5e3610562dc045036cfcbc2255fa363b6","_number":925436,"virtual_id_number":925436,"owner":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-25 04:15:01.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 03:33:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 04:59:07.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 07:09:49.000000000","updated_by":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"reviewer":{"_account_id":26286,"name":"huanhongda","email":"hongda.xun@easystack.cn","username":"huanhongda"},"state":"REVIEWER"},{"updated":"2024-08-07 05:36:59.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-07 05:40:02.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-27 13:04:20.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"REVIEWER"}],"messages":[{"id":"648a3a2480feee8624116c69e004d9533e8fbb9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-01 03:17:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"23729589476399411a4b992344285398507f9fb9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 03:33:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16c7424110c24f84824d0a014ead87dd\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/b964029a2bb0402996abe199f0c0b359 : SUCCESS in 13m 33s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/54e042850e5e41018668b1507199e0a7 : SUCCESS in 13m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"cd54ae9c8f6fc46f3f8628dea91eeea2e119f478","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 04:59:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c5766512e3d844668afeb9a85b20c448\n\n- grenade https://zuul.opendev.org/t/openstack/build/95c3836573144050875832931320cef8 : SUCCESS in 53m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/357cb0800ef0442298b80c53b8bdd211 : SUCCESS in 1h 18m 55s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5a13cab202d4f42b3797ae08679bf8b : SUCCESS in 48m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b57d0777382447678a9eb1659b6035b5 : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/091fbf0d7fb645cc8e188bed18ee71b2 : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4da0d632086f42acb0aab11dc084ee72 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/347d75752f0a4a5f8783f5743e9ff274 : SUCCESS in 9m 13s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3696ada0bb1463eb3970faef3f90fe4 : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fbdad75e9ca94049b392ce1ed4ba91ac : SUCCESS in 9m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/652717e61fe94226957de1665cf67f63 : SUCCESS in 8m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/72a507425ff74e9bbd8c149493baf278 : SUCCESS in 13m 10s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/12b9fe4591844d9dac7d27c81bb21135 : SUCCESS in 9m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0eea5591368b40478b336a0c416f0a83 : SUCCESS in 1h 27m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/94d5b8f098f84a98a72ae2d993d73f78 : SUCCESS in 1h 31m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/792d68ace61e4428b73a227270b4deab : SUCCESS in 1h 26m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/3e438f637e0743b5b575ebfc45a02e84 : SUCCESS in 1h 38m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6fc8c342320f4ad9bd253ca28dcee052 : SUCCESS in 1h 35m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d1f3e52a20e542be95047238595304e2 : SUCCESS in 59m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e2b7245ebecd46ccb5fa4502590c06f2 : SUCCESS in 1h 34m 31s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8cb33da54e2245b299758d979c5d5028 : SUCCESS in 1h 27m 02s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2184ac5185d842d281dbd9cb779b0942 : SUCCESS in 1h 07m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a86bee78ed3d4a5488f24c3d22971892 : SUCCESS in 31m 44s","accounts_in_message":[],"_revision_number":1},{"id":"dae9543c2256319139a6d9e5df852ee055072f1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-01 05:58:06.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"890e60a345f38c364e6fc30500f3217a44f35422","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 06:09:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5a80256b72aa4a7db6b5a5f371820c36\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4747bbd254dd41a498e924abc91bc7a5 : SUCCESS in 9m 27s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6baa4a611e0462aaac7bc7b6cc1fea5 : SUCCESS in 9m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"21db437d2a2ddaa3b890407b81a3c1a82eb4e92e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 08:03:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10093afcea5246bfa45e4f6affce0a07\n\n- grenade https://zuul.opendev.org/t/openstack/build/ca3f8904a76f4d2fb95f5ef958a10e82 : SUCCESS in 49m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1e416fdc350749a6a20fc3208f5d98d0 : SUCCESS in 1h 05m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/965566431f0c489a9040af4b30fc5051 : SUCCESS in 49m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02cbea3ac4e24e1ab8f25cf1fd0ded12 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a778bc9f3ce40c599464c3b99f09110 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d624e2cbe791423e81f1949146d3980e : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ce36085a423d4623ae6e575567abbd19 : SUCCESS in 9m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47417fecea844f42b8b5cb36cacb5ba0 : SUCCESS in 5m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f3f72413fe18407aab14b39582b43082 : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4654fa2115ea4e00a2d66d6febe8454d : SUCCESS in 8m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d647fa4b9b3046409215fc44cdf9bccd : SUCCESS in 10m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/da05685c6369417ab7d617e5a82d5431 : SUCCESS in 8m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9d74bc9484004970a20fb6886f8e21ec : SUCCESS in 1h 41m 54s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/250be5e075a54db5842c27cb1ea99866 : SUCCESS in 1h 46m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/874606ce263a40b48f5b9cef0030a955 : SUCCESS in 1h 42m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c3c67be925e43f8b494801badbcf367 : SUCCESS in 1h 53m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a308e4fbe99d453ba57534f2958188ef : SUCCESS in 1h 59m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fc5fbad04256416084acbe1cb6954bcf : SUCCESS in 1h 04m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f82ca3fcc834ce683d456b27484004c : SUCCESS in 1h 19m 29s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6c6cfb12117a4552a48bb37145bb8865 : SUCCESS in 1h 36m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9ced956b5a7848f29127aea30116e076 : SUCCESS in 1h 07m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a18747597d0544fe8f0c6738409ea00c : SUCCESS in 20m 55s","accounts_in_message":[],"_revision_number":2},{"id":"50745f016e7c1d72a0909eace49aa3cf0b5e2fa2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-07 05:36:59.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f39d77361f82049c536bc4b361fdbe752d86b3e6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-07 05:40:02.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5413aee3141226b07ca1933d4068f398fdf8b689","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-21 12:31:39.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"c60babcd839c40c7a0b8b7142617b3774f028e81","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-21 12:33:27.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"3fb2028c17943ae11982a4d07189569f44f53156","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 12:58:59.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1abf64390bf14f3bbf485dbab6dfff09\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/8d2937a679e64455b83f107df02961bc : FAILURE in 8m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/0d9fc150646647738bebd8a146923ed0 : FAILURE in 7m 55s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"fcf24fd4d4ef9d65090f2e73a74c1b62d5e997d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 14:44:23.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a46acae00cd7425a8ff40368b1bb3f27\n\n- grenade https://zuul.opendev.org/t/openstack/build/3f605f76d8234806ae7e2ff47ecf5367 : SUCCESS in 1h 01m 58s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/60169b496b63441b875c86e3c50cf98e : SUCCESS in 1h 45m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a7dc5ff94094388bdb75db494fffca6 : SUCCESS in 1h 08m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e0c8f78aa9b84f4b8c18b0045ee3b0d0 : SUCCESS in 5m 12s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e0f63d468ceb432bae271351772e888d : FAILURE in 5m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e76c6d5649d6434ba7b95878846ceea7 : FAILURE in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/575aeb88dddc4ae98be68ab49c95680f : FAILURE in 6m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0826266b7dbb44f1b66859478f45e426 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cee01e64218742ffb3f33da3b0809736 : SUCCESS in 10m 22s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/99dd7972f2af477e9dd477f47a228f0d : SUCCESS in 11m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92a774313dcd437d92a54faa7eb48b34 : SUCCESS in 12m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/965b5178897044478ae3a00cd53eb470 : SUCCESS in 8m 09s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3dd245808755431f969c0f153cd1409d : FAILURE in 1h 55m 34s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7ff672caa4974eee818d35d263e0fc71 : SUCCESS in 2h 02m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/014508a392344de9bb0179defdea3b15 : SUCCESS in 1h 59m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ec94e7a10d56407dab3389db7be98a58 : SUCCESS in 1h 58m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/28407a6461404255ac6e1fef2bd35512 : SUCCESS in 2h 10m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/39cd6953e604489bbde4d25fa9412d1f : SUCCESS in 53m 39s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/115a778bad524dbaa50ceec823380aac : SUCCESS in 1h 58m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/e6be3d9bf95845f6abd8f5506e4a91b0 : SUCCESS in 1h 53m 13s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b3ab6664581c4ad0a3a5288b2659c4fe : SUCCESS in 59m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7e3b4564758149b68135179192fcff92 : SUCCESS in 26m 00s","accounts_in_message":[],"_revision_number":3},{"id":"ba3a1365815f031c00ffa2a3e5a0b4255f6bccba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-22 03:09:27.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"a4ec5746d053e2a62004d090ed275b33ab6c258c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 04:08:14.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/41611d94935f4345b0d48f8a274e2252\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ee4d5d041399467aad2d33039fa37490 : SUCCESS in 9m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aba9ed1d65054d4a992a9f9e912f7311 : SUCCESS in 9m 24s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"469ae5626347a798285944bd9137fdf1ae445209","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 05:09:25.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/163523826a3a479c81655c4fdc174da3\n\n- grenade https://zuul.opendev.org/t/openstack/build/b1f9fd5b813c4e4b985d002366a1dab9 : SUCCESS in 1h 02m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8968928db1e842d19c6066ef3e9314fa : SUCCESS in 1h 25m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7c1261e697df499d9f88d3f60c3fa0cb : SUCCESS in 47m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4c04a44c30940228c1bb31cefcccb4e : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/218e73c860384b96a74ebd981c21fa30 : SUCCESS in 5m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8e2add5ee3914e80ada590d297041618 : SUCCESS in 5m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2041c225f7324264ab5f1a86dd8efee7 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ac494a00ff5432aaad3b90af4c310b4 : SUCCESS in 5m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/53e0f1fadea1478f996b62284125e878 : SUCCESS in 11m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/abf5653232094d938e031210c702ae66 : SUCCESS in 7m 43s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0cd56306b4c84df8befd2a909a81a7f1 : SUCCESS in 15m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0325a7cbbf714f778462790d344a34ad : SUCCESS in 9m 39s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9f6840bfa5b649bf83c2487de5a81ead : FAILURE in 1h 59m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6cd1f77af4d4421b872fe5e2e6bebdd5 : SUCCESS in 1h 36m 59s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6f6539fb590c4d3ba9f53fd7db08165f : FAILURE in 1h 32m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/920477862e8f4d84a36ffc55ed645ade : SUCCESS in 1h 42m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e3c2efb3add7481e9c64ff203630ad16 : SUCCESS in 1h 54m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6d18c80c8d444a868c4bae0e9c0e6cfd : SUCCESS in 58m 57s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f14d17563d8f4a95b21bc170cf36f693 : SUCCESS in 1h 41m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b87dcd74716d41648698ea6638743f12 : SUCCESS in 1h 41m 12s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/de537b9166d74fcda2c3471e66b6a1c9 : SUCCESS in 1h 02m 23s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ec51e1c56c6f4ac1a6c8a02717d1a92a : SUCCESS in 24m 52s","accounts_in_message":[],"_revision_number":4},{"id":"903dc88cef8e7c680617d780043768daa7066b64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"date":"2024-08-22 09:19:47.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"9d04ee689e11b14629858448f812d330dbda016d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 12:01:55.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f086336ae23467fa00cd4939df8c760\n\n- grenade https://zuul.opendev.org/t/openstack/build/c2bfb321780a49f1b28e1d58a26b5b0b : SUCCESS in 53m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/de9b31b36d374ae8ab734db02c90c1ba : SUCCESS in 1h 21m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9c8f64c5f2584df9a82dd9ea48349666 : SUCCESS in 58m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6291ac0e0dc040d78572aafdc6952d42 : SUCCESS in 4m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/06d89fdfab264cfcb9f0c9e93f3fb7e0 : FAILURE in 4m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ce316b06d7cc4bd095812e774542eca3 : FAILURE in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1466b29a4104d428b22b2ce4f3f26ae : FAILURE in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df6877ca84b7492ab2f60f4cdb30030d : FAILURE in 4m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c474c4d14fcc43dd9fee2466e8a9a29b : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/767300bf31f84aa38f4145ecb5504759 : SUCCESS in 10m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/3d71d4d38f7c413ebcae266ebc4632ac : SUCCESS in 12m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9a4d77f5465e434d94188b1b78e82be7 : SUCCESS in 8m 46s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/76ac73ecc7c74dacbbfd06c3a5f3dbf5 : TIMED_OUT in 2h 03m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/240ffa2b9f8545e282f12034e4a6622e : SUCCESS in 1h 40m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/81221cc5333b4875969ceb5490ae9396 : SUCCESS in 1h 42m 32s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d2bcc17ca5fc431c83d5706faa53ccf5 : SUCCESS in 1h 51m 37s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8324636373eb4d8d9f2310fe53365aa3 : SUCCESS in 2h 31m 04s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7154bc858d5c4e10a1a2f80867b3f1ea : SUCCESS in 53m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c62bcedbc09640bb9b2e1937823c0b47 : SUCCESS in 1h 33m 59s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/801420373d354f97b1581e3495c1fc3d : SUCCESS in 2h 00m 04s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4cb3fd6b1822457383e63fa19e0c2713 : SUCCESS in 1h 04m 58s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/86cf3afba7a1439eac699b34306d8992 : SUCCESS in 23m 55s","accounts_in_message":[],"_revision_number":5},{"id":"8485fef0a20a5966adcc49b7405d9a3232194eb4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-22 13:03:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c20b85206d6244a4915b843b59ef778d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0b2f646ec4824724967ee8d907496268 : FAILURE in 12m 38s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/719dd8bd249245d7b8e564e74041a37f : FAILURE in 16m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"feb7e94cc54819cb172b9bfd482cac5aa6d9ec38","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-08-27 13:04:20.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7bf9ce33357dc90922f6aab26c1ddc528b3ccc2f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-25 01:58:20.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":6},{"id":"94eaf1a9ea6585d727c81a7d04f2cf7f2f395ca1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 02:09:58.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8edec7b4701543af94705d2c9b25cba6\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/70602230d4eb432489efb192dba81a08 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d52bd6260b784bf982c8bb2a671a4668 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/c55d00f860044be6bba5a6f752158094 : SUCCESS in 8m 54s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8d4324b5e3610562dc045036cfcbc2255fa363b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-25 04:15:01.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ee4bcf4d5ec3440497a6391488e1be6d\n\n- grenade https://zuul.opendev.org/t/openstack/build/eb799647c6544351874972e928b13ae0 : SUCCESS in 54m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/699f240648564ef7b7109c8d14e0f155 : SUCCESS in 1h 04m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1b6f945649fd4cbebcebda35850611df : SUCCESS in 1h 45m 34s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/28fc3c2ce4104895a6ae50396052f209 : SUCCESS in 1h 11m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19e0434b18349c2afcfc74e11dc5a51 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d26a0540f69f44caaf12d286c0018bc6 : SUCCESS in 5m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3b5706a6e1e64ed5b945fee432a6b761 : SUCCESS in 5m 27s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/a91d01614d9e4d0da47565361e12f4ae : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0db1edce7ae54ff88af5652d2e25291b : SUCCESS in 7m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6c26895cb22f40a692d9c6167cf9a667 : SUCCESS in 7m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8b1c048685604d4ca9fe5a43e66baa7c : SUCCESS in 8m 44s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/e48a4682798b4ff3b9d3d80ef3c34551 : SUCCESS in 4m 05s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ce819d2c8fec4e439023be37c461394b : SUCCESS in 1h 07m 16s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3903260f5b8f4a329a60654803e50922 : SUCCESS in 1h 07m 42s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/45c8a09f747e46cf890ae1dffda62b41 : SUCCESS in 2h 10m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f54a082e9bcd4fcd8f0b8ce1e9d67a17 : SUCCESS in 1h 15m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/61a80231254c461a9bf4b62313c157d7 : SUCCESS in 1h 00m 37s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/718b087bcfe743febfb43772ae17506c : SUCCESS in 1h 34m 18s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2a77ba5f15a84d2e95b66f8acb9aeebc : FAILURE in 17m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/42e5fb458fb440078307ce1ec98c58c5 : SUCCESS in 1h 59m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c4814912af4f4eccb67e441e1a35ad9c : SUCCESS in 38m 04s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/9234809151574ef8816135bd0fc71a03 : SUCCESS in 22m 54s (non-voting)","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"b7a9fbc4eb3860e3d187f505347ae8ae02feff1a","revisions":{"490fa736fdbeea97ae426d7ab5f6e3f9a63e8360":{"kind":"REWORK","_number":1,"created":"2024-08-01 03:17:11.000000000","uploader":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"ref":"refs/changes/36/925436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/1"}}},"commit":{"parents":[{"commit":"536a34b5d79bfd2179d66cda7c2abd827ebb681e","subject":"Merge \"Remove location_strategy functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/536a34b5d79bfd2179d66cda7c2abd827ebb681e"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"subject":"Delete image in stage when using the image conversion plugin","message":"Delete image in stage when using the image conversion plugin\n\nWhen use the image conversion plugin, the image\nif not raw format, the file path will\nend with .output format.\n\nWhen delete the image in importing status,\nthe file path obtained by glance will not be\nthe converted file, it will cause the image\nfile to remain in the stage.\n\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/490fa736fdbeea97ae426d7ab5f6e3f9a63e8360"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/490fa736fdbeea97ae426d7ab5f6e3f9a63e8360"}]},"branch":"refs/heads/master"},"682081b1f1f3f293268e0fd40376c2e5b096461a":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-08-01 05:58:06.000000000","uploader":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"ref":"refs/changes/36/925436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/2"}}},"commit":{"parents":[{"commit":"536a34b5d79bfd2179d66cda7c2abd827ebb681e","subject":"Merge \"Remove location_strategy functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/536a34b5d79bfd2179d66cda7c2abd827ebb681e"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 05:57:35.000000000","tz":480},"subject":"Delete image in stage when using the image conversion plugin","message":"Delete image in stage when using the image conversion plugin\n\nWhen use the image conversion plugin, the image\nif not raw format, the file path will\nend with .output format.\n\nWhen delete the image in importing status,\nthe file path obtained by glance will not be\nthe converted file, it will cause the image\nfile to remain in the stage.\n\nRelated-Bug: #2075489\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/682081b1f1f3f293268e0fd40376c2e5b096461a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/682081b1f1f3f293268e0fd40376c2e5b096461a"}]},"branch":"refs/heads/master"},"594b5dc90223977bab2d53ee39d948e8f9e4844d":{"kind":"REWORK","_number":3,"created":"2024-08-21 12:31:39.000000000","uploader":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"ref":"refs/changes/36/925436/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/3"}}},"commit":{"parents":[{"commit":"536a34b5d79bfd2179d66cda7c2abd827ebb681e","subject":"Merge \"Remove location_strategy functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/536a34b5d79bfd2179d66cda7c2abd827ebb681e"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-21 12:31:11.000000000","tz":480},"subject":"Delete image in stage when using the image conversion plugin","message":"Delete image in stage when using the image conversion plugin\n\nWhen use the image conversion plugin, the image\nif not raw format, the file path will\nend with .output format.\n\nWhen delete the image in importing status,\nthe file path obtained by glance will not be\nthe converted file, it will cause the image\nfile to remain in the stage.\n\nRelated-Bug: #2075489\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/594b5dc90223977bab2d53ee39d948e8f9e4844d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/594b5dc90223977bab2d53ee39d948e8f9e4844d"}]},"branch":"refs/heads/master"},"cdea07e2548e2e47f7fa3e73e7a2f2b1b7794592":{"kind":"REWORK","_number":4,"created":"2024-08-22 03:09:27.000000000","uploader":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"ref":"refs/changes/36/925436/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/4"}}},"commit":{"parents":[{"commit":"536a34b5d79bfd2179d66cda7c2abd827ebb681e","subject":"Merge \"Remove location_strategy functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/536a34b5d79bfd2179d66cda7c2abd827ebb681e"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-22 03:09:09.000000000","tz":480},"subject":"Delete image in stage when using the image conversion plugin","message":"Delete image in stage when using the image conversion plugin\n\nWhen use the image conversion plugin, the image\nif not raw format, the file path will\nend with .output format.\n\nWhen delete the image in importing status,\nthe file path obtained by glance will not be\nthe converted file, it will cause the image\nfile to remain in the stage.\n\nRelated-Bug: #2075489\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cdea07e2548e2e47f7fa3e73e7a2f2b1b7794592"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/cdea07e2548e2e47f7fa3e73e7a2f2b1b7794592"}]},"branch":"refs/heads/master"},"ed3a5758dbe121106448e8d5c420415b6d507e7b":{"kind":"REWORK","_number":5,"created":"2024-08-22 09:19:47.000000000","uploader":{"_account_id":37206,"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","username":"shisen.zhang"},"ref":"refs/changes/36/925436/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/5"}}},"commit":{"parents":[{"commit":"536a34b5d79bfd2179d66cda7c2abd827ebb681e","subject":"Merge \"Remove location_strategy functionality\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/536a34b5d79bfd2179d66cda7c2abd827ebb681e"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-22 09:19:33.000000000","tz":480},"subject":"Delete image in stage when using the image conversion plugin","message":"Delete image in stage when using the image conversion plugin\n\nWhen use the image conversion plugin, the image\nif not raw format, the file path will\nend with .output format.\n\nWhen delete the image in importing status,\nthe file path obtained by glance will not be\nthe converted file, it will cause the image\nfile to remain in the stage.\n\nRelated-Bug: #2075489\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed3a5758dbe121106448e8d5c420415b6d507e7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ed3a5758dbe121106448e8d5c420415b6d507e7b"}]},"branch":"refs/heads/master"},"b7a9fbc4eb3860e3d187f505347ae8ae02feff1a":{"kind":"REWORK","_number":6,"created":"2026-06-25 01:58:20.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/36/925436/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/36/925436/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/36/925436/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/36/925436/6"}}},"commit":{"parents":[{"commit":"fa9a03f0a5fdda81607376fde6a542168049c2e0","subject":"Merge \"Remove unnecessary direct dependency on Pygments\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/fa9a03f0a5fdda81607376fde6a542168049c2e0"}]}],"author":{"name":"shisen.zhang","email":"shisen.zhang@easystack.cn","date":"2024-08-01 03:12:44.000000000","tz":480},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-06-25 01:58:15.000000000","tz":120},"subject":"Fix image name computation when deleting an image","message":"Fix image name computation when deleting an image\n\nWhen using the image conversion plugin, properly compute the filename in\nthe staging area before attempting to delete it.\n\nCloses-Bug: #2075489\nChange-Id: Ie5cd832001b4e49581fbcd3d3281f6d4fd36948e\nCo-authored-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\nAssisted-by: Claude Sonnet 4.5 \u003cnoreply@anthropic.com\u003e\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7a9fbc4eb3860e3d187f505347ae8ae02feff1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b7a9fbc4eb3860e3d187f505347ae8ae02feff1a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
