)]}'
{"id":"openstack%2Fglance~925466","triplet_id":"openstack%2Fglance~master~I50907c90686366cab12be072b53c65b048b7c510","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-01 15:03:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50907c90686366cab12be072b53c65b048b7c510","subject":"Auto-detect qemu-img qed support for unit tests","status":"MERGED","created":"2024-08-01 11:16:43.000000000","updated":"2024-08-01 15:04:58.000000000","submitted":"2024-08-01 15:03:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925466","meta_rev_id":"789408a8bb287b4ead17df505bac911c65b0f221","_number":925466,"virtual_id_number":925466,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-01 15:03:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-01 12:59:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-01 12:59:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-01 11:36:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-01 12:12:26.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-01 12:59:14.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"4d3a3704e4a74f55b0ba8651702f4e36bd7c508d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-01 11:16:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7477ba68eba476d3b48384022b032217e500431b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 11:36:39.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ac31cbb30a7241ebaa7e977d474a79eb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/493499d49c5c4d69a84d4ba837dec96c : SUCCESS in 10m 01s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/807d5df8b0854da19fdc772d5f39b54f : SUCCESS in 9m 46s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5580ac8014e38f5c88d85dcbe654dbb61b730d26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 12:12:26.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4f97cba1cc241009fade98df58f2d11\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6dbce93af7f944a4ab76adfb7da2357c : SUCCESS in 54m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef5374677a8746c7a2f177bd13d0d6fd : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b359bc0328d7440f9839d7ca7131339a : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e871e1acb414367a32b9d125e327fae : SUCCESS in 4m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/57b892da883f4a39a86d1c61014664f4 : SUCCESS in 8m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e65c4110e55346338ca4663344872f9b : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/74681c9c505948738bcb5091b081ce7c : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67b46c1a92294a609756fb5cd583686e : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f9e1853ea63a454b8f300e3fa8c2565c : SUCCESS in 14m 31s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/78b1682c4a0a49ac9fa1045f5bb10cea : SUCCESS in 8m 15s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/32b5045ab1e444b3b30d46bdd12d0dc8 : SUCCESS in 32m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8a13ef5f4859e5bb77baef9ae9c48787ef00aa57","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-01 12:59:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f095cc1f9e064c9ff1f521d3eb22813695e5505d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 12:59:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d6b37468c48b27d0e9b5e0e1286813198a6af93b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:03:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/61f622bef6284bf291233403f2817a89\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6f01d358d7644c309cbf5fb6706fcceb : SUCCESS in 55m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4a62e466d9744b2a920c1c19cea0fc5 : SUCCESS in 4m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/27fbfad16d5d4c09af9acf56d81db5d8 : SUCCESS in 5m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/41dab93b9b154d44b5b7847e08f6424a : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8a51b1cd57a443a9969e6af1476aa841 : SUCCESS in 5m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/365038eb0daa4c2d80e837a3830fb66f : SUCCESS in 8m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/44927c8a4c554c3081e8d575a4ad03ab : SUCCESS in 8m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/31c05270437a405a962e5328fa954805 : SUCCESS in 16m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6e67785ed4704ca7a178d7ea299e44f3 : SUCCESS in 34m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2427111b3e0e5df2033aa8724f7fc29fbfcfe518","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:03:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"789408a8bb287b4ead17df505bac911c65b0f221","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-01 15:04:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ceaa2be80a6c4c4a9ff909958f074c95\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cbe186a2bb54193bfc44e4932da9a25 : SUCCESS in 58s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"93ae58c61ccf816d2c6b824e78022ee72bffea24","revisions":{"93ae58c61ccf816d2c6b824e78022ee72bffea24":{"kind":"REWORK","_number":1,"created":"2024-08-01 11:16:43.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/66/925466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/66/925466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/925466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/925466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/66/925466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/66/925466/1"}}},"commit":{"parents":[{"commit":"7e75dc9e9a7405b501bfb731cbceb4ab302cd273","subject":"Merge \"Make separate schema for new location API\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e75dc9e9a7405b501bfb731cbceb4ab302cd273"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-01 11:09:02.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-01 11:14:32.000000000","tz":-240},"subject":"Auto-detect qemu-img qed support for unit tests","message":"Auto-detect qemu-img qed support for unit tests\n\nThis will let downstream distros with qemu-img versions\nthat don\u0027t support the qed format pass unit tests without\nhaving to patch this test out.\n\nChange-Id: I50907c90686366cab12be072b53c65b048b7c510\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93ae58c61ccf816d2c6b824e78022ee72bffea24"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/93ae58c61ccf816d2c6b824e78022ee72bffea24"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
