)]}'
{"id":"openstack%2Fglance~925730","triplet_id":"openstack%2Fglance~stable%2F2023.2~I373dde3a07332184c43d9605bad7a59c70241a71","project":"openstack/glance","branch":"stable/2023.2","topic":"bug/2072483-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-06 08:39:00.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-08-06 06:47:22.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I373dde3a07332184c43d9605bad7a59c70241a71","subject":"Revert image state to queued if conversion fails","status":"MERGED","created":"2024-08-06 04:34:17.000000000","updated":"2024-08-06 08:40:32.000000000","submitted":"2024-08-06 08:39:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925730-bug/2072483-stable/2024.1-stable/2023.2","cherry_pick_of_change":925687,"cherry_pick_of_patch_set":1,"meta_rev_id":"874fbffa18dd43465504ad43617386259d016ce2","_number":925730,"virtual_id_number":925730,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-06 08:39:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-06 06:47:22.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-06 06:47:22.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 04:34:17.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-06 04:34:55.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-06 04:51:02.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-06 06:38:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"5bac895586585e9c30a53e9ed3df41c9a7170234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-08-06 04:34:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"0f9c72c85f514a74f642c870cb5c533343438fb5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 04:51:02.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9594fd3da5964555bb8140a6edb138d7\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f0902d087bf3421087fcd2ed8290a38a : SUCCESS in 9m 39s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/67f7a2620693493b8c98406bc6f274c0 : SUCCESS in 14m 28s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/a8e10b364987411ca3541097d19ea8de : SUCCESS in 13m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/d6b58a6ffec54018bd82e13e06b7dfe7 : SUCCESS in 12m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e50f6dcd2f4fe0f0f1a89822ca4eb94a1415fc97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 06:38:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a7af18b6c9b04a2f9fc6de49041034b5\n\n- grenade https://zuul.opendev.org/t/openstack/build/84873e1599694064bbe654a887e3ee9e : SUCCESS in 42m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4ba810f4b1e04f42ad9e8e326e626836 : SUCCESS in 58m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b75009b63ac84516ac6a11cca6f5b8d5 : SUCCESS in 41m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7179ac61a95441da80500bdf939661d0 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4cf516d6b6264586b48f41fd236d633f : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3ce0ee7a0dee4c8e9e00a90b6a52a43c : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23ed9bd8e85347b0ba4792ef2cceef71 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a308252c67f443a97ae3b67f7dd8280 : SUCCESS in 4m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d364af3457644849da6dd58c5a80d9b : SUCCESS in 5m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4289a1f083584127bdea0b6b6bfb26e0 : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6c7b65e088ab4f36878555110a02812a : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a68260cc163241d6a44abf2773e398ba : SUCCESS in 7m 51s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4fd3c1f276aa4a00bc89fcf19a4cf9ec : TIMED_OUT in 2h 03m 15s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/bc7de84d8e6c4701b2ec986e00dba504 : SUCCESS in 1h 13m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a83dbcabf06044b69bad1f6ea0d38531 : SUCCESS in 1h 19m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/50072d1de0b24b868a121b5a08a16ca4 : SUCCESS in 1h 48m 33s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/406b2b7da5da45739caf391242700f1f : SUCCESS in 1h 26m 19s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/cc0a4760d5c14c60b315c603f55bb761 : SUCCESS in 37m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c0dbbca35d1e4a64a2bd5a7fb8cc8a9b : SUCCESS in 1h 54m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/cbce34a671e64dcfa5b6126fe6a91156 : SUCCESS in 22m 19s","accounts_in_message":[],"_revision_number":1},{"id":"b7c6a3730e89a28d3f180bb4e5bf3028557b11c9","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-06 06:47:22.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0064747c68a63ff009851fcbc7e9d0f0c1e5e6f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 06:47:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4316567fbbebe6e901131be5249fd41e67d0a95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:39:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e941756371684bedaa66280b1c17203d\n\n- grenade https://zuul.opendev.org/t/openstack/build/5ba35ac1216f4505aba47615d7daf3d9 : SUCCESS in 40m 18s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/381b66d302bc410a98c4454998f8b56a : SUCCESS in 1h 02m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/31afddd008a6443cacdad0424d52ca33 : SUCCESS in 44m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7b3fe7ae4e994848ba3ff412de7be434 : SUCCESS in 4m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6ecdc8d67f4f42648fe5801b7da05cf4 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf0ebe7f58804add805e324948ab5852 : SUCCESS in 5m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6e45e30852fe4293acfd08669160a784 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c7b8abf4d7849e3939450d27d491799 : SUCCESS in 5m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/59fdb5e8d29c4ea3b190a1cd92eb9a87 : SUCCESS in 8m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8ea54b303a4423eb60a4227e20d37f3 : SUCCESS in 7m 55s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b42348d6e59843c284ffce41af0e82d7 : SUCCESS in 1h 20m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1d3300d391014f7eb04204c2b7d430f4 : SUCCESS in 1h 15m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/64bfb4ba60ba40abb09523abd66665ba : SUCCESS in 1h 44m 28s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5bdecd89a2e4443e803c053028fd0d3b : SUCCESS in 1h 02m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b67f69ed05614a30bba6a38c76d8c9f4 : SUCCESS in 1h 49m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d2a5813452a34fbf868ea837e1cf90ce : SUCCESS in 26m 34s","accounts_in_message":[],"_revision_number":1},{"id":"12ff00c29410253730ac288c2d811d05cdab2259","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:39:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"874fbffa18dd43465504ad43617386259d016ce2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 08:40:32.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/44abcdc6386940ec94a1d16700212255\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0546c805e24e407f94458e7ea22046e4 : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"3b3141baa38a8dfbf37e7633cf9bd62898a545da","revisions":{"3b3141baa38a8dfbf37e7633cf9bd62898a545da":{"kind":"REWORK","_number":1,"created":"2024-08-06 04:34:17.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/30/925730/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/30/925730/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/925730/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/925730/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/30/925730/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/30/925730/1"}}},"commit":{"parents":[{"commit":"d9f33251035283329019f3ca4091f537ed723cbe","subject":"Merge \"Fix /healthcheck url typo in glance-api-paste\" into stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d9f33251035283329019f3ca4091f537ed723cbe"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 09:49:55.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-08-06 04:34:17.000000000","tz":0},"subject":"Revert image state to queued if conversion fails","message":"Revert image state to queued if conversion fails\n\nMade changes to revert image state to `queued` and deleting image data\nfrom staging area if image conversion fails. If image is importing to\nmultiple stores at a time then resetting the image properties\n`os_glance_importing_to_stores` and `os_glance_failed_imports` to\nreflect the actual result of the operation.\n\nCloses-Bug: 2072483\nChange-Id: I373dde3a07332184c43d9605bad7a59c70241a71\n(cherry picked from commit ea131dd1442861cb5884f99b6bb9e47e397605ce)\n(cherry picked from commit 633e85c9c07f85de36677137ea15bc3a7f4feb04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b3141baa38a8dfbf37e7633cf9bd62898a545da"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b3141baa38a8dfbf37e7633cf9bd62898a545da"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
