)]}'
{"id":"openstack%2Fglance~925769","triplet_id":"openstack%2Fglance~stable%2F2024.1~I50907c90686366cab12be072b53c65b048b7c510","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"4523":{"account":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"last_update":"2024-08-06 16:37:54.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I50907c90686366cab12be072b53c65b048b7c510","subject":"Auto-detect qemu-img qed support for unit tests","status":"MERGED","created":"2024-08-06 13:42:49.000000000","updated":"2024-08-06 16:40:12.000000000","submitted":"2024-08-06 16:37:54.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925769","cherry_pick_of_change":925466,"cherry_pick_of_patch_set":1,"meta_rev_id":"95bef60ceafe62cda87b77d3f3d26b11f2c0707b","_number":925769,"virtual_id_number":925769,"owner":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-06 16:37:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-06 14:07:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-06 14:07:51.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-06 13:53:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-06 14:07:51.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-06 14:52:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"ba1f14bf1d0ab63d160361dfff8b4dc622c625d2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"date":"2024-08-06 13:42:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"ad5dbf1355d0f9138c12e26a0e291c05e0f00f63","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 13:53:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cbaf3978032e465980dc3c7bac956775\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3b2316888c3f4e25b7eb50831ddadc5b : SUCCESS in 7m 42s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/aade2ccac3a04d068a4efb31ee7b795f : SUCCESS in 8m 52s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c42794903cc2914e986d54086d33282978ffb15a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-06 14:07:51.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"44b173c5b6aad1fc5390c832c2e387fba6b16d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 14:52:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2551a08c7642d3b846aba0fdfed94a\n\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/01b430d5f96e4ca68fb5a66de7478ee8 : SUCCESS in 1h 02m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1d02fa94245848f3a8659f5f6e2023da : SUCCESS in 1h 04m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ae9bf997968e464b9d488a11537c8fff : SUCCESS in 4m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4a2e565726b4fe0a03335e8702ac6ab : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b3614ad1b5604b08b14174c3c6ca85b3 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ca0db282c0b40f4ab813a44169b73bf : SUCCESS in 5m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dc242fb428754ac3b1a76d53d550dba8 : SUCCESS in 9m 15s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3b3ef293601f4a1293e48803be9d86bc : SUCCESS in 7m 57s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/080143008c814fe2b55bf27e89be1b9e : SUCCESS in 8m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3261888933614021a4b56242dab2fa2b : SUCCESS in 32m 40s","accounts_in_message":[],"_revision_number":1},{"id":"981b62601c9e77bb551396486169e10826b42c2a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 14:53:32.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c1650b1b33d119143cb1f67674d8f3f90fae3ebb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 16:37:54.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f42ce3ab7bd844b2a67480b4eaff0948\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/69d085c879ed48d2b58e5dcdfb334090 : SUCCESS in 1h 00m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5a2cb6fae19248dd9b80e4d383871a3e : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/84c6b67d0d404e96bdd6db875e765459 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/911c02598fb2446e81301fa9c1488baa : SUCCESS in 6m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/beebd7ec9914490aa0f92cfcce472d5f : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1e6acb9fd2dc4be6961c39eccabbf026 : SUCCESS in 10m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a28ba93e037143b797a9f0f966fe3f67 : SUCCESS in 9m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c6e8c0c6f35e48beaaf62c105eeb4bdf : SUCCESS in 31m 08s","accounts_in_message":[],"_revision_number":1},{"id":"4163b4aeaedd0d6beaa2414d64998aa2142713fb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 16:37:54.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"95bef60ceafe62cda87b77d3f3d26b11f2c0707b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-06 16:40:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6eb883672d42699bc9ea1d0684e7fd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd822bfb4ecf4a6299991c58253115ab : SUCCESS in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"584accc55816234c9189e9a3ed5b8fb316b1fa6f","revisions":{"584accc55816234c9189e9a3ed5b8fb316b1fa6f":{"kind":"REWORK","_number":1,"created":"2024-08-06 13:42:49.000000000","uploader":{"_account_id":4523,"name":"Eric Harney","email":"eharney@redhat.com","username":"eharney"},"ref":"refs/changes/69/925769/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/69/925769/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/925769/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/925769/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/69/925769/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/69/925769/1"}}},"commit":{"parents":[{"commit":"066fa00474b4bf027cd3c21d2d6c4267deea2b14","subject":"Merge \"Revert image state to queued if conversion fails\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/066fa00474b4bf027cd3c21d2d6c4267deea2b14"}]}],"author":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-01 11:09:02.000000000","tz":-240},"committer":{"name":"Eric Harney","email":"eharney@redhat.com","date":"2024-08-06 13:42:49.000000000","tz":0},"subject":"Auto-detect qemu-img qed support for unit tests","message":"Auto-detect qemu-img qed support for unit tests\n\nThis will let downstream distros with qemu-img versions\nthat don\u0027t support the qed format pass unit tests without\nhaving to patch this test out.\n\nChange-Id: I50907c90686366cab12be072b53c65b048b7c510\n(cherry picked from commit 93ae58c61ccf816d2c6b824e78022ee72bffea24)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/584accc55816234c9189e9a3ed5b8fb316b1fa6f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/584accc55816234c9189e9a3ed5b8fb316b1fa6f"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
