)]}'
{"id":"openstack%2Fglance~926025","triplet_id":"openstack%2Fglance~stable%2F2023.2~I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c","project":"openstack/glance","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-10 02:23:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c","subject":"Fix LP#2054575","status":"MERGED","created":"2024-08-09 05:36:18.000000000","updated":"2024-08-10 02:25:05.000000000","submitted":"2024-08-10 02:23:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926025","cherry_pick_of_change":924949,"cherry_pick_of_patch_set":1,"meta_rev_id":"3b93d13b61e072110ffd0b42e7667a441fe6f4f4","_number":926025,"virtual_id_number":926025,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-10 02:23:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"all":[{"value":2,"date":"2024-08-09 20:02:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":2,"date":"2024-08-09 05:39:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":1,"date":"2024-08-09 05:39:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-09 05:39:38.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-09 05:49:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 07:44:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-09 12:09:18.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"CC"},{"updated":"2024-08-09 20:02:07.000000000","updated_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"}],"messages":[{"id":"188ecff35a89c7c720aa459e51fb46e8e41ef4a7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-09 05:36:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"c21e1af4f551548eb91c2d028347183c6496d4df","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-09 05:39:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ae5e73510028fdd6e221608c0d12d18275c38637","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 05:49:33.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9d7a23a0580c45bd9498d31162294930\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c6e68e2d9320463fbff6d9624ea339f1 : SUCCESS in 9m 05s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca201ba611534400a2599c3a93a0c0c2 : SUCCESS in 9m 08s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4739882096bb4a8db624f338b20e77a6 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/a70f33f869f841b9afc149d811abb64e : SUCCESS in 10m 29s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9ce3f61f3814561fa4d51b37c41d5857e6c6baed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 07:44:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/1d635065f5504773afd7e8c9ad0d52d4\n\n- grenade https://zuul.opendev.org/t/openstack/build/399c0feea8be4cc8ac6a95083f779d0f : SUCCESS in 53m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b32f0962b966471584eb08ac79a8e493 : FAILURE in 1h 20m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8ecf727112b047ed95cce96f3eed572e : SUCCESS in 1h 04m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a43c91158ce94d548c5b0fff81843997 : SUCCESS in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c3fc6ed03430470dbcb59c710cd6e121 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f6fa84d6ee524acf855ccffb45478b9e : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa012c0e6f424051954e0d06554d45b5 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c4af8144df724a1194d2f66d3bc25456 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0e1a03e64f24b9496cc7e774b99d9e7 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fefb312c581d414aa31ebf4187bd4e7b : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea3ba0256e12451baf80ecbdd57b6a41 : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1f89b4c3c90a400781ef0d02fa6a23ae : SUCCESS in 8m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b17d1bf1b34e4da68575c9bd110a5579 : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e53084c60ea841b9a96d5a26cf69099c : SUCCESS in 1h 55m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/96ec8d82ffac4d4a8aaaa5ca02c142fb : SUCCESS in 1h 38m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e7bec9b3bcc4472786ee3575d2a50028 : SUCCESS in 1h 48m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/79f09d5cd7d54debb4a85b221e172b62 : SUCCESS in 1h 34m 20s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/8781d776249b4d2589a75e5f7dbff52c : SUCCESS in 1h 59m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d8d48b451b894d5b89ec86bdf052c39c : SUCCESS in 54m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/62d09cc68e2143c1a52c7359ea96a487 : SUCCESS in 2h 01m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/59b5a6aecbfb4195b3a7745c0c22fb3b : SUCCESS in 36m 07s","accounts_in_message":[],"_revision_number":1},{"id":"2ed4337d556da151d667f1817ae5b88950d0ffb5","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-09 12:09:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7817f976a4840470902c0cdeac33f19b66966446","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 12:42:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/76477c2b042d48528ee68f589a477573\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/74354f55de5248a49b0effe6e89bd634 : SUCCESS in 11m 32s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/655f30e1421b4bb7b70bbb4105d35fe6 : SUCCESS in 10m 47s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c268e9791a364fc986b134121fb7e27d : SUCCESS in 10m 53s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f1f4ff35a15c4901a45b832ac28f71a2 : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"73fa7f7c0fc197bbb1b8f1628b3cc418b77927eb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 14:16:00.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee6093c38cdb4b1db791aa911dfbb4ab\n\n- grenade https://zuul.opendev.org/t/openstack/build/461cd5203c73437d9c90d2c97d218a0e : SUCCESS in 41m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/921048efa32e4453a3132080f8174a53 : SUCCESS in 1h 03m 31s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ed0e36aab66f4f249fe0f14177a69ef4 : SUCCESS in 42m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e48cde127bb49f897e698dd1b37d90d : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02bf28a6907643639981f334b4d4a45e : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/888d98e6e7294fe1bffd11a93510dc57 : SUCCESS in 4m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/80156f5860c741ce8b4719d6c1d02fb8 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/de0f0c767c6445a896b8de03b9ffd9b0 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b12be676c45c4e60b2e2cd3cc66d623c : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69862146cc7c40cf91bb1b738a62b53f : SUCCESS in 6m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29d49a974ffa48669bf60059f56cddd2 : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2ff4dfa3c4b246ca866fb76e02cd0186 : SUCCESS in 7m 06s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/d44c88a7d6074de197310e00faee5eeb : SUCCESS in 8m 44s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0471c02cf8af4039937781438309c8f2 : SUCCESS in 2h 01m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/af044c9fb520459a8f6e209330047833 : SUCCESS in 1h 37m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2eadd537b8fd4dfeb7d3f75c0ea5c410 : SUCCESS in 1h 49m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/44ecabb99a044980ac505e36d5cfbc25 : SUCCESS in 1h 43m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1ce625305b854267b7b61a5417568486 : FAILURE in 2h 06m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/711cf3256779488193762bb6dea9cb6d : SUCCESS in 59m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5f584820610646a3994d7192045d8b7a : SUCCESS in 1h 47m 10s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08c596ba984f44b693b4979e6e0ba62a : SUCCESS in 33m 19s","accounts_in_message":[],"_revision_number":1},{"id":"e1517bf111e5cfd7db6254d8d32e202aaf0ee0cf","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-09 20:02:07.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"df425b4c925af3b70c385aa864beaa9a49aae57c","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-09 20:03:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"dac18fd510dd07be33a80aa883b64ddabf5ba7ea","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:35:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dc271b3cb6449b494f48fbb70cb4824\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e0ce9630b2c948348381f07399ea9980 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cf60fb90b8064a8a9088e18f9f13b36c : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/30be701c40354c588f60dc0ed860b476 : SUCCESS in 10m 56s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4584423f05e64ecb8a1d1228b21d9a4d : SUCCESS in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"088f47aac580085428654e1889a66bb0f81543b7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 21:49:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbc6ffdec3bd479b82e76270d86bb72d\n\n- grenade https://zuul.opendev.org/t/openstack/build/df653720c26a4847a06b17ac0202a77d : SUCCESS in 38m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b15aa6f44d904585bb546cd507e0a93c : FAILURE in 1h 18m 22s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/42b5f94db228416daad67cdd97b162f9 : SUCCESS in 44m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f44b039df7214ec88708423219f4cd55 : SUCCESS in 3m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4bcd5f7384734cd7b58da035dd3957f2 : SUCCESS in 5m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d6e4ef35121642c6bcc8e6b252954b35 : SUCCESS in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4ed51974e83c434fbfa6b7085eecb7a1 : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/678863b433004f04926c16b2850bda2d : SUCCESS in 4m 22s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/baa0b7ef54e64d9391d283fd95044ddb : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a149fd1dfa9b46999f391cc484504cae : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c4817a37bd694a28ae990f4dbbf53a28 : SUCCESS in 8m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fb21d8d1d32a498b94a3d94e193b0369 : SUCCESS in 7m 24s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a13f1dc9d8d449bba11c035d117469e9 : SUCCESS in 7m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/211acee90e9b4dd998f42270d754f591 : SUCCESS in 1h 33m 24s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/099af5b91bf54a269b7517205cabcec9 : SUCCESS in 1h 30m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9670f5ba837c4274a056b204289ab65e : SUCCESS in 1h 15m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a35af94f0dd44c0ebc7a89eeaa357cb9 : SUCCESS in 1h 37m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e5759475b25c4f4084c4b0e30079475c : SUCCESS in 1h 45m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a9a1ae4ee73e4c8ca79a284b6c661db9 : SUCCESS in 38m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3e2673c107124ba0aa85059a34149a1b : FAILURE in 1h 32m 26s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b8198149d41849bbbf8ae2a7bd433330 : SUCCESS in 22m 44s","accounts_in_message":[],"_revision_number":1},{"id":"3d217e4051d27853efe9abb8f588440a64f1a237","author":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"date":"2024-08-09 22:08:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b9590eb4022b4c6c5c4a8d5693c8321b10af16e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 22:34:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6611dad5586743d2952c1de8ce8baaa5\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/c4f57ea3c17542a1ad4b07a5db8c60c6 : SUCCESS in 13m 43s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/bd8e1d0377ff4ca98f0f87fc729d509b : SUCCESS in 10m 27s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/769abbec05534a59b67e88d8e2b0d411 : SUCCESS in 9m 04s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/10a3efc7d28b444f90c0b5e659fb29a1 : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4ff31bd9bd17d352704de1384a6166e0a92206dd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 00:04:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88e3ffdafa384f55a1a1f529bb8cd86f\n\n- grenade https://zuul.opendev.org/t/openstack/build/1a5a50d3e723415d8c3990f11d757722 : SUCCESS in 38m 11s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df89accd95164ad291d34e7d55565a27 : SUCCESS in 1h 28m 27s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8bf2c66d5aeb428aad0d5129155e1759 : SUCCESS in 1h 10m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93e588247e324d178c0c18a3a3f29606 : SUCCESS in 3m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f99f9cb3b704084985dfa40c31843ca : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/836bd3b251fe4470b50b3e907ce01e88 : SUCCESS in 4m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e4b9c13906a1487fb2acd938f8f23e9f : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e806e6e1bd74456e8a2e70a77b13a7f0 : SUCCESS in 5m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a378ea7c9c4654938cc9d7c9d7893d : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3c8ae31864574ebda27d7d1b6b7ec941 : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/784f64540b02426c99b4e2f940886ce7 : SUCCESS in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3619713fee8d4011a5ead71c174cc79f : SUCCESS in 8m 14s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dbce7df77ddc41a3b26f6f6a8dca7589 : SUCCESS in 6m 42s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4b3ed54467a64544bf315a1b8c5adfcd : SUCCESS in 1h 50m 20s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e141df8d9514e6e956ecc6eefe59cd7 : SUCCESS in 1h 40m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ad44c9d1e4004fe88dcb8d7e95d09c50 : SUCCESS in 1h 43m 33s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/829016c6535e43e79c4f6030597e4a70 : SUCCESS in 1h 36m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/201b29438a7949ffbb4d318c90977d10 : SUCCESS in 1h 39m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/35e5ba11497c4949b81030029089c011 : SUCCESS in 39m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/47f8108bf0504661b601130abbea38df : SUCCESS in 1h 22m 03s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/502c47bdd68a421185b5c4f65d0dd355 : SUCCESS in 27m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9a9fa489aabc87da0f9880e22ec7785ca2bbf342","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 00:04:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"634de63d145be1abbfa18a13ee41889038461845","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 02:23:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff6b93197474cf1a5d5e97df3ada893\n\n- grenade https://zuul.opendev.org/t/openstack/build/e286ec331a114fbfb4ed69e8a088dc2e : SUCCESS in 39m 37s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f885621571b44d0fa1f3c1456ff68432 : SUCCESS in 56m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7e6a9a5bb1f94047856bc21189f608e4 : SUCCESS in 58m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c776c8bdca5541bfb2724e6e4c2dcdc0 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95d32472853a4970988bae077193a54a : SUCCESS in 4m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60e4c54a6e83437abd01c6507c173161 : SUCCESS in 5m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b822ab8fe6c84b3b990b1a22607d2330 : SUCCESS in 4m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bd6ab45fe2c495faf779be39a5601de : SUCCESS in 5m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01ddfcacec64485aa4e969e86db5429d : SUCCESS in 6m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf5b243dee3d4261816c017b282ffdff : SUCCESS in 8m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/67049da81148440aa63632009a2120c6 : SUCCESS in 6m 57s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/70d6ca7163174b10ad8665b428493831 : SUCCESS in 1h 10m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/814690d85cc74c97a323b233a29633ab : SUCCESS in 2h 07m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c9803c5de7b14754a0dee5eda89e5dd8 : SUCCESS in 1h 05m 53s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/211c007a873648f7a1f7705c67acedc2 : SUCCESS in 59m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6954eebdbe4849b79148ee8722eb1536 : SUCCESS in 1h 07m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/12a4c9e4cc754cdea959e3b4625d6d5b : SUCCESS in 22m 02s","accounts_in_message":[],"_revision_number":1},{"id":"0ae942e91faea8dcd23480f4e033329d0035e2c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 02:23:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3b93d13b61e072110ffd0b42e7667a441fe6f4f4","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-10 02:25:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e52f45c33854a88a9da974bb16b4ca5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c17000333419409b81656006f06088b0 : SUCCESS in 1m 18s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/119088d437104e00893da7213177c7cc : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8540ffc6ee34976285209e9f3c7c728bf649e0ff","revisions":{"8540ffc6ee34976285209e9f3c7c728bf649e0ff":{"kind":"REWORK","_number":1,"created":"2024-08-09 05:36:18.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/25/926025/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/25/926025/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/926025/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/926025/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/25/926025/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/25/926025/1"}}},"commit":{"parents":[{"commit":"3b3141baa38a8dfbf37e7633cf9bd62898a545da","subject":"Revert image state to queued if conversion fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3b3141baa38a8dfbf37e7633cf9bd62898a545da"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-25 15:46:27.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-08-09 05:36:18.000000000","tz":0},"subject":"Fix LP#2054575","message":"Fix LP#2054575\n\nThis patch contains a squash of two commits:\n\nCOMMIT 1:\n\nMake location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nRelated-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n\nCOMMIT 2:\n\nFix: optimized upload volume in Cinder store\n\nWhen Glance is configured to use Cinder store and we upload\nvolume to Glance in the optimized path, it fails with\nInvalidLocation error.\nThis happens because Cinder is not aware about the store\nin which we will create the image and supplies the old\nformat URL i.e. cinder://\u003cvol-id\u003e whereas Glance expects\nnew location format i.e. cinder://\u003cstore-id\u003e/\u003cvol-id\u003e.\n\nGlance has code to update the format from old location format\nto new location format but it isn\u0027t triggered in case of\nold location APIs.\n\nThis patch adds the context to the update store ID request\nwhich calls the Cinder store to provide the updated location,\nhence fixing the optimized path for upload volume to image.\n\nCloses-Bug: #2054575\nChange-Id: Idd1cb8982b40b85a17821596f76dfa10207f6381\n\nThe commits are squashed together to make backport easier.\n\nChange-Id: I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c\n(cherry picked from commit 53175262f0f8df303176f4e9cd44e36d442c6cea)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8540ffc6ee34976285209e9f3c7c728bf649e0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8540ffc6ee34976285209e9f3c7c728bf649e0ff"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
