)]}'
{"id":"openstack%2Fglance~926087","triplet_id":"openstack%2Fglance~master~Ib9a503783df6fb40058089319c7c8b7e9d89d3b8","project":"openstack/glance","branch":"master","topic":"secure-rbac","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-08-13 18:17:42.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2024-09-05 01:22:20.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib9a503783df6fb40058089319c7c8b7e9d89d3b8","subject":"Remove default override for RBAC config options","status":"MERGED","created":"2024-08-09 18:36:09.000000000","updated":"2024-09-05 01:24:27.000000000","submitted":"2024-09-05 01:22:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926087-secure-rbac","meta_rev_id":"88157102b69e985a3980bd3c9d29c1ff42fa52ba","_number":926087,"virtual_id_number":926087,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-05 01:22:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"all":[{"value":2,"date":"2024-08-13 18:17:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":2,"date":"2024-09-04 23:04:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":1,"date":"2024-09-04 23:04:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-09 20:10:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-09 20:54:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 17:57:43.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-09-04 23:04:48.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"2024bfb42d45d8c54cbebd6f3b97fdfaf4e9ced9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-09 18:36:09.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"34214712abf495f684a3f680a8f979a80bd1663d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:10:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/20a13e64d89d437f8a94e65776d5a34b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/d9727e98ad144a2bbf381cecbd1b6108 : SUCCESS in 9m 36s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/786c232224364156a357b3ef6d02aed4 : SUCCESS in 9m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7408fe2a1b77e37b9e117561a63379585686855c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-09 20:54:56.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4f906b4b076747de8b84f8746684ff4c\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14653d8da9e74c80ad7f201786012038 : SUCCESS in 2m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/6739dad45f924564b2e76f2276ea06ce : SUCCESS in 1h 01m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9fa8b9c266e149e89e80dc365066abea : SUCCESS in 1h 25m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6669ff85d9e24887b8ece0dade08d850 : SUCCESS in 55m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/646c64ccaf88435289a852bf479934c0 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2443a8dd1b694b0e804f732bcdfb0458 : SUCCESS in 5m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/da9385226f134262b0a3ae3c06bbd3d2 : SUCCESS in 5m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/33839c3e3a5445e9a643b3cf09f2fe05 : SUCCESS in 8m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebd9bb0f77e34c61a3db27c2cf6431e4 : SUCCESS in 7m 15s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/291566b7fc9a43f4906e80156c6161f2 : SUCCESS in 11m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe8a34021e564e9b8f9d52a637e6daa0 : SUCCESS in 7m 52s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2ea1acb2231d47fb9b6a7f07c5eb93b8 : SUCCESS in 12m 16s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4c0c83f0a89b48bc80c51e4442a043c2 : SUCCESS in 8m 57s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f208f5967b954aafa2a9b6df8f6940db : SUCCESS in 1h 25m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9e2e8de4c588468fb4d34ad4b13e70e3 : SUCCESS in 1h 40m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/645b466c5358475c9409775a20ea18ef : SUCCESS in 1h 32m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/43e268c4930c484c9266ea143e4dbc87 : SUCCESS in 1h 43m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/100d70e7cd734756bddc3c975c046018 : SUCCESS in 2h 13m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a786ac6af56345aba7c50c196f9611a0 : SUCCESS in 55m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/39935a1adacd46f2a7dc7ad769ac7bcd : SUCCESS in 1h 38m 22s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9244068fbe554eee974e3c5c006e3ff7 : SUCCESS in 1h 33m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/2e3e205ac162433baa7a1330f2096c52 : SUCCESS in 1h 05m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08d5b7a336ab492daba6d40ac27ad40d : SUCCESS in 30m 59s","accounts_in_message":[],"_revision_number":1},{"id":"2621de8bfec0d495d0f919427ff33743a48c9037","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 17:57:43.000000000","message":"Patch Set 1: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"5edf2d8df6d38edcb8ef3de6d6593d21dd80f7d8","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-13 18:01:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1139a2228c33402ebf55b38b6570d22cbc9e80d6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-08-13 18:15:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"9ad20328b68874b19c63d18c4fc18de5a541de04","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-13 18:17:42.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"3e5b6a8a9a74eb9445a81f3076a8d8425caba62c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 18:28:57.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4347a347edb643c3b7a0d23734594ec7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/44930aa4517b480b86693b8adf27f4e7 : SUCCESS in 9m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/8f25e18317de43b3a242fc72cb747c76 : SUCCESS in 11m 11s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f9395acafcb9f49d6516e507b8be1f5cfcabf9c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 20:38:27.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/bbb0b448337d450aba0f36bf1da45a6e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/7317d0e815214696b2cbd63be6e5b001 : SUCCESS in 2m 42s\n- grenade https://zuul.opendev.org/t/openstack/build/9dc5ae6953f640c1899e8aba2f05ac27 : SUCCESS in 1h 05m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/66d58a2187b94503bcbb01c7992c61a7 : SUCCESS in 1h 27m 48s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18faeedfe73c4bcdb034b371ff2c1a84 : SUCCESS in 56m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18145f20798f470fb341a1aacc77fbef : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed031b3e91114880a63e1dacc63c02b7 : SUCCESS in 6m 30s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e385e13f7ad14f7bbfb6767089250f45 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a7f25e9ed9f34c6995b371488b28c781 : SUCCESS in 10m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a4d7272a4194fd8b0521a3f4741a77b : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/148c54f7b0404a138ee9e27e9ee36288 : SUCCESS in 9m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6635fcc3d3164262b840de21a7b2db16 : SUCCESS in 9m 22s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/21c30b67072a4c67a19519ed92b856b2 : SUCCESS in 12m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d772c0b2c62444c2acfd89b41ea340f7 : FAILURE in 1h 47m 26s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d27f699609984878a85cbdacbea9ae97 : SUCCESS in 1h 53m 09s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4d8404275fcd4102b9180c37c9f2b232 : SUCCESS in 1h 52m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0f6efc40c77c4e7db6bd1ce176950216 : SUCCESS in 1h 32m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3edd4e82a8a543979a07e236131973b7 : SUCCESS in 2h 17m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/93537974c0cd4c848bb0b9f41612b5c5 : SUCCESS in 41m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0682cdce44d9418294fb054c9fbb118b : FAILURE in 1h 43m 38s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/52521d1a9d634fbdb4b4fd4753d84b98 : SUCCESS in 1h 32m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bf4a5e59a35e4897915752488d3e4f4c : SUCCESS in 55m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/707475c5eb3041fe8696c9def90cf601 : SUCCESS in 21m 10s","accounts_in_message":[],"_revision_number":2},{"id":"85eb671c8235c4d8e508d452bd9f7c5558b08138","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2024-09-04 00:28:22.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c03d1c22845c796fc680e5b5b4c0913cc9211d38","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 02:48:17.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e83eee0099104a1d9ecaa14391193e6a\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3cf94d8745e84839880213b4e5744764 : SUCCESS in 2m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/6b0a3689dc58447982661b6f085f05ca : SUCCESS in 53m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0fce6e148fa45aea14317f42cb1dda7 : SUCCESS in 1h 31m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/93bafd16399643abbf7b0b40fd548412 : SUCCESS in 1h 00m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/20ee27aa32404caf8c48d98dcb72ad79 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aa4e838842404f42932475273245cbc8 : SUCCESS in 5m 09s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/125f6de4807f4108af50ea130c07a33d : SUCCESS in 6m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ed9a3520460442ed9e2a87c6552d30a5 : SUCCESS in 6m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e95cd7520bb48b5bb254917c2cfd91f : SUCCESS in 6m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/10453742d0284996a8d20d4dfeaeb5d3 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d941e4b1a85e44c485dfe6d59d98301d : SUCCESS in 9m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/2fd990b99d6d452aad1378b9f29d167d : SUCCESS in 14m 00s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/094c6cb7f3b4483f9a30e7d7a3898e76 : FAILURE in 1h 30m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/deb4931ae0f0491091c07917d323ac9a : SUCCESS in 1h 48m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/90a77de32c2f451cb014548c0576c67d : SUCCESS in 1h 55m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/330c3c25ed9e468c94b884852b41a1a2 : SUCCESS in 1h 58m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5600041e84ae425b9be2b2464c792cf5 : SUCCESS in 1h 57m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8ee1ad33dae04c6d8de7d80d5ecfd50b : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b2bd44c0460f46d29497c37ef90bad48 : SUCCESS in 1h 37m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b69e265899c6412c96639aa34894e617 : SUCCESS in 2h 03m 45s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d6cfd5367e8241718371d9332822aa87 : SUCCESS in 1h 10m 09s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9e8ecea2e6d542bebe4ca5f5e049b14e : SUCCESS in 33m 35s","accounts_in_message":[],"_revision_number":2},{"id":"f87706b721ee6d113f2ddbc4c4a441f50bcac061","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 03:20:18.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e84d646ac01a468bbb014091c4e308b0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/38b68001f8184a26ba87dd5d44b186f6 : SUCCESS in 13m 28s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/54ea7bf6cd70427cad3366e7af4cf84c : SUCCESS in 11m 20s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2db3bd47f9a23b8ec1eac7dd93cebab160b3d7cd","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-04 23:04:48.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"3e8404cd560864ed4167d5e748b8a54bd514ee95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-04 23:05:34.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"09cfb8d62e523a1f4cc7abc164c10d6b72dac406","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 01:22:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/611041f7c0014ccf961e3853dc7fd4ed\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/f5d83eb887d34022bb86f76bed9f8f28 : SUCCESS in 2m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/22b46a7d71b549438720c4e0d5159ad4 : SUCCESS in 1h 05m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6d187ded9c484fbf93bb47beacdd5873 : SUCCESS in 1h 37m 09s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2fe5266ff1d14b6ca71fb9fa31dccd39 : SUCCESS in 1h 00m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad21829dad84ee4803f8869b97ce4ac : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8f4e1b68eaf4367a66a4d786824e3ce : SUCCESS in 5m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bd8cd777859a4c319340824690b82e8b : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd9efe01287747289e232b315d09e83e : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3511f0a71679425caeb0629557c2e0c5 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cc11823cedf949b98addccbb36f59d72 : SUCCESS in 8m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4866f3da7cde41ebbae0943f88169307 : SUCCESS in 12m 52s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0e0b4d308240402a9bf37fe73627caf4 : SUCCESS in 2h 05m 03s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62c8b302ba1e4c2691eb1e02a3b937db : SUCCESS in 1h 43m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f1b23e1c710d47529855f49df8a2d1db : SUCCESS in 1h 45m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a372c7a63fac48919472b2b716225706 : SUCCESS in 1h 01m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/07478cef71074072a7abbdb1d6fbe762 : SUCCESS in 1h 18m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/762b20c422634aa5ada06013ee4f4838 : SUCCESS in 37m 15s","accounts_in_message":[],"_revision_number":2},{"id":"5442fb721a8795cbd15458c7a13458b3a9b7dc1d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 01:22:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"88157102b69e985a3980bd3c9d29c1ff42fa52ba","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-05 01:24:27.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a18ceb9b336744ceab0f9f8220295b32\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d60b597033c40118671182a975ab888 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"273825d01dd2ecc403346975ecef66ae11f5d233","revisions":{"84146aca16b55bf27c9502ebca44498ec97dccc9":{"kind":"REWORK","_number":1,"created":"2024-08-09 18:36:09.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/926087/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/926087/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/926087/1"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-09 18:35:29.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-09 18:36:07.000000000","tz":-420},"subject":"Remove default override for RBAC config options","message":"Remove default override for RBAC config options\n\nRBAC config options enforce_scope and enforce_new_defaults\nwere disabled by default in oslo.policy and Glance had to override\nthe default value to enable those by default. Now oslo.policy\n(4.4.0 onwards[1]) changed the default values[2] and enabled\nby default for all the services. OpenStack service does not need\nto override the default anymore.\n\nNOTE: There is no change in behaviour here, oslo.policy provides the\nsame configuration that Glance has overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/925032\n[2] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nChange-Id: Ib9a503783df6fb40058089319c7c8b7e9d89d3b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84146aca16b55bf27c9502ebca44498ec97dccc9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/84146aca16b55bf27c9502ebca44498ec97dccc9"}]},"branch":"refs/heads/master"},"273825d01dd2ecc403346975ecef66ae11f5d233":{"kind":"REWORK","_number":2,"created":"2024-08-13 18:15:08.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/87/926087/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/87/926087/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/87/926087/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/87/926087/2"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-09 18:35:29.000000000","tz":-420},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2024-08-13 18:14:59.000000000","tz":-420},"subject":"Remove default override for RBAC config options","message":"Remove default override for RBAC config options\n\nRBAC config options enforce_scope and enforce_new_defaults\nwere disabled by default in oslo.policy and Glance had to override\nthe default value to enable those by default. Now oslo.policy\n(4.4.0 onwards[1]) changed the default values[2] and enabled\nby default for all the services. OpenStack service does not need\nto override the default anymore.\n\nAs every job run with glance new defaults:\n- Removing glance-tox-functional-py39-rbac-defaults\n- Add new periodic job to test old defaults\n  - tempest-integrated-storage-rbac-old-defaults\n\nNOTE: There is no change in behaviour here, oslo.policy provides the\nsame configuration that Glance has overridden till now.\n\n[1] https://review.opendev.org/c/openstack/releases/+/925032\n[2] https://review.opendev.org/c/openstack/oslo.policy/+/924283\n\nDepends-On: https://review.opendev.org/c/openstack/requirements/+/925464\nChange-Id: Ib9a503783df6fb40058089319c7c8b7e9d89d3b8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/273825d01dd2ecc403346975ecef66ae11f5d233"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/273825d01dd2ecc403346975ecef66ae11f5d233"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
