)]}'
{"id":"openstack%2Fglance~926112","triplet_id":"openstack%2Fglance~stable%2F2023.1~I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c","project":"openstack/glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2024-08-13 08:34:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c","subject":"Fix LP#2054575","status":"MERGED","created":"2024-08-11 19:35:20.000000000","updated":"2024-08-13 08:35:37.000000000","submitted":"2024-08-13 08:34:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"926112","cherry_pick_of_change":926025,"cherry_pick_of_patch_set":1,"meta_rev_id":"74ff56b8273b84ab76247b7001ddca8057f9a507","_number":926112,"virtual_id_number":926112,"owner":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-13 08:34:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-13 07:09:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-13 07:09:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-11 19:45:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-11 21:36:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-13 07:09:52.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"b7c3ad233be599517dac49a80078dea3df7732a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2024-08-11 19:35:20.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"2294bac87196d99779d2763a188402b3972fd28f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-11 19:45:44.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/033873cf9ed64d059c069f45d5be6be1\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/04a79f000e064fba9ca917b923474697 : SUCCESS in 8m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0f2f0b75134a69232ae4ec571932056438ae45bd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-11 21:36:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54b034b8308c42e68a87f187f18fddd2\n\n- grenade https://zuul.opendev.org/t/openstack/build/299d8b0e893347e49cb751e6e3ca4233 : SUCCESS in 37m 39s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/27ac3434f432488ea227720a4634a7f3 : SUCCESS in 37m 01s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5f16143b7a15434693aac9529f0f2a43 : SUCCESS in 52m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8d4d861e54cd4289bfa912b7a2df9bbd : SUCCESS in 36m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/865a5277647543069ca86c9cbb0f6e43 : SUCCESS in 4m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/046fe220ed7c454d8f51ee4b7cd39ecc : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8eff9af1c2e4427e8c2b138a494aff98 : SUCCESS in 6m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cd3ea9f327364a96a19d9ac0cfd14ba0 : SUCCESS in 6m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b2a30123009e425ea7078c08c5d56619 : SUCCESS in 7m 42s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/e24b0062755b453a83dca1ba9978f3d9 : SUCCESS in 8m 06s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a2ecc4ca68eb4773a0884426f8968608 : SUCCESS in 1h 35m 35s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5efdd902502747fabc81ce22466aba77 : SUCCESS in 1h 21m 17s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee7520215e8e4d558dfaa01cad438a3f : SUCCESS in 1h 27m 10s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/afa4c051d26f4195b29ec6df5373db83 : SUCCESS in 1h 52m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4d8f6dc0329e4844b8ce2a0648de83e8 : SUCCESS in 1h 45m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/822586fb9605482aa6a3f58aff14e91b : SUCCESS in 35m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/af804a34da0e4de8bec2c88c34cc8aa3 : SUCCESS in 1h 36m 41s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/af3f45cd3d4d49f186a3459150b9e805 : SUCCESS in 21m 01s","accounts_in_message":[],"_revision_number":1},{"id":"9f28ad98773a6abf3b39168ed8456ad9054fadcb","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-13 07:09:52.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"406dd3f916558831a2ece6f53f89637752890396","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 07:10:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cb9bd1698703efd2bd00d91e474278f18e89306b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:34:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c088d8266b9849f48add4b408300c968\n\n- grenade https://zuul.opendev.org/t/openstack/build/74fd20099e9f417eb54e282380ebfb99 : SUCCESS in 1h 02m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/da3af2d534b740beb1913456505d778a : SUCCESS in 1h 00m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/17e40c4ca72c45048cf3b4b072a25c2b : SUCCESS in 1h 19m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/768837ef719c41cca7204688af0e9a30 : SUCCESS in 56m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/125d549fc71949bc8f33d9ce51cf446c : SUCCESS in 3m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/02e47f86b28f429fb6a01d26845d4a8a : SUCCESS in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2211f01c3f8466dba0e1aae01aeca49 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a98b1740c694daa8de8d6a378cac02c : SUCCESS in 5m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/70f00b36f8fb4fa5b244ad7c2e23a235 : SUCCESS in 8m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27580d010f354d7a9c90e283fda37225 : SUCCESS in 7m 29s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/78ccf2044a2c454a800492fd2a9cc081 : SUCCESS in 58m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e8d376abddda4cbcaf36fe05ef6dcbfa : SUCCESS in 1h 13m 48s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/adf6c419744c4442b244531c6338c57f : SUCCESS in 1h 13m 13s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ea4505913c664d40a32f71e586094d19 : SUCCESS in 34m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/17143f7298d04e459887442390623c11 : SUCCESS in 1h 23m 00s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/a9a8c2d418a3485081c2fc4464e55186 : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"ab86646c7dddfc42f5dbaa1dfdc2a15ec51fa380","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:34:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"74ff56b8273b84ab76247b7001ddca8057f9a507","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-13 08:35:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6285f62e6f84441e8b49f2d29bb5cec1\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/95ddcea390274124a6292ad07138adb5 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a949c30b2e274cdbaad6148a3b253e7d : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d","revisions":{"6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d":{"kind":"REWORK","_number":1,"created":"2024-08-11 19:35:20.000000000","uploader":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"ref":"refs/changes/12/926112/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/12/926112/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/926112/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/926112/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/12/926112/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/12/926112/1"}}},"commit":{"parents":[{"commit":"3f7f01bb0584289c933a0fae3f915afec41093c8","subject":"Revert image state to queued if conversion fails","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/3f7f01bb0584289c933a0fae3f915afec41093c8"}]}],"author":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-07-25 15:46:27.000000000","tz":330},"committer":{"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","date":"2024-08-11 19:35:20.000000000","tz":0},"subject":"Fix LP#2054575","message":"Fix LP#2054575\n\nThis patch contains a squash of two commits:\n\nCOMMIT 1:\n\nMake location URL compatible with cinder backend\n\nWhile adding location to an image, cinder sends location url as\n`cinder://volume_id` for single as well as multistore which is\nincompatible with glance multistore and store throws\nInvalidLoctation error.\n\nModifying the location url to be compatible with multistore as\n`cinder://store_id/volume_id` to avoid Invalid Location error.\n\nRelated-Bug: #2054575\nChange-Id: I5f791c58ae857f6c553276dd9808799c3db3aa4f\n\nCOMMIT 2:\n\nFix: optimized upload volume in Cinder store\n\nWhen Glance is configured to use Cinder store and we upload\nvolume to Glance in the optimized path, it fails with\nInvalidLocation error.\nThis happens because Cinder is not aware about the store\nin which we will create the image and supplies the old\nformat URL i.e. cinder://\u003cvol-id\u003e whereas Glance expects\nnew location format i.e. cinder://\u003cstore-id\u003e/\u003cvol-id\u003e.\n\nGlance has code to update the format from old location format\nto new location format but it isn\u0027t triggered in case of\nold location APIs.\n\nThis patch adds the context to the update store ID request\nwhich calls the Cinder store to provide the updated location,\nhence fixing the optimized path for upload volume to image.\n\nCloses-Bug: #2054575\nChange-Id: Idd1cb8982b40b85a17821596f76dfa10207f6381\n\nThe commits are squashed together to make backport easier.\n\nChange-Id: I9ecdfe08b63c00446dc3e24195e3b8e59b82f55c\n(cherry picked from commit 53175262f0f8df303176f4e9cd44e36d442c6cea)\n(cherry picked from commit 8540ffc6ee34976285209e9f3c7c728bf649e0ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
