)]}'
{"id":"openstack%2Fglance~926295","triplet_id":"openstack%2Fglance~master~I3a030b1e044653a6470377f58eeed5baf779b7c3","project":"openstack/glance","branch":"master","topic":"LUKS-image-encryption","attention_set":{"27665":{"account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"last_update":"2026-07-15 15:56:40.000000000","reason":"A robot voted negatively on a label"},"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-06-18 14:01:30.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}},"4690":{"account":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"last_update":"2025-12-16 10:30:37.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}},"5314":{"account":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"last_update":"2024-11-08 10:22:14.000000000","reason":"\u003cGERRIT_ACCOUNT_28271\u003e replied on the change","reason_account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}},"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2026-07-15 13:41:19.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}},"5263":{"account":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"last_update":"2025-09-04 20:01:52.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}},"28271":{"account":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"last_update":"2025-06-25 13:50:25.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"27615":{"account":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"last_update":"2025-04-09 09:06:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27665\u003e replied on the change","reason_account":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"}}},"removed_from_attention_set":{"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-05-13 14:37:09.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-06-25 14:48:12.000000000","reason":"\u003cGERRIT_ACCOUNT_8122\u003e replied on the change","reason_account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"}}},"hashtags":[],"change_id":"I3a030b1e044653a6470377f58eeed5baf779b7c3","subject":"Standardization of encrypted images","status":"NEW","created":"2024-08-14 15:57:10.000000000","updated":"2026-07-15 15:56:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":166,"unresolved_comment_count":6,"has_review_started":true,"meta_rev_id":"07f5300c6c006b576debbf2f3d9da48f659ee2f5","_number":926295,"virtual_id_number":926295,"owner":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-07-15 15:56:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"}],"CC":[{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-14 16:52:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-14 18:22:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-29 07:07:54.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2024-08-29 07:09:41.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-08-29 07:09:41.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-29 08:10:10.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-08-29 14:23:30.000000000","updated_by":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"reviewer":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"state":"REVIEWER"},{"updated":"2024-09-26 15:30:02.000000000","updated_by":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"reviewer":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"state":"CC"},{"updated":"2024-11-08 10:22:14.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"state":"REVIEWER"},{"updated":"2024-11-08 10:22:14.000000000","updated_by":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"reviewer":{"_account_id":5314,"name":"Brian Rosmaita","email":"rosmaita.fossdev@gmail.com","username":"brian-rosmaita"},"state":"REVIEWER"},{"updated":"2025-02-28 21:45:06.000000000","updated_by":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"reviewer":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"state":"CC"},{"updated":"2025-03-03 14:47:49.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2025-03-11 01:45:14.000000000","updated_by":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"reviewer":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"state":"CC"},{"updated":"2025-07-08 16:26:37.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"cb2b1a10c4bd227e6c8379a3e924ad679d6cfdec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:57:10.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d06e86b1e907733b6873aaaed2dde220a2bf5034","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-14 15:57:24.000000000","message":"Topic set to LUKS-image-encryption","accounts_in_message":[],"_revision_number":1},{"id":"f8e897530171dcd92dfa94303a4ef469cdc2c496","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 16:52:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e07146371ec74081aa7381c466244c82\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/95cfe871ff4c43b5b96fd1aee7902834 : SUCCESS in 11m 54s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f96ec22aa1ab4ba2a8d5d67fdcad18ec : SUCCESS in 11m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"39a84ab5dbf393dd99179f6dcccf099c0c74b270","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-14 18:22:46.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c38be70246d54f589e432d3738d1bf8e\n\n- grenade https://zuul.opendev.org/t/openstack/build/074352bf932c48e698469f2a4375ca56 : SUCCESS in 1h 06m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2d9a75e9abde4e01a0d3ef5c90d59ae8 : SUCCESS in 1h 45m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b19e0cd344dc432784354c574291950a : SUCCESS in 53m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5cc4fe2e1bc247d2b8b043abef099728 : SUCCESS in 5m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/21d66d7e214f4df382e7ac5f6e6f410b : SUCCESS in 6m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d404da3b5dcf4deb8f26e7825b673674 : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1990a9f907744de1a6f6c7529657d449 : SUCCESS in 9m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/706047d3d03c418787483f77f128e2cd : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/881948d28ed2496eb5007f54f41f4467 : FAILURE in 10m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14e7f7d70a514411a71c94ce11b0ea5d : FAILURE in 9m 03s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9d67e065d25f462a97da254c27586f5d : FAILURE in 13m 01s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/c3be6b7d7c054299bbb72a38a7272f42 : FAILURE in 9m 50s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/08c0d0c0b2c440f3b5ce17307102e3ff : FAILURE in 1h 58m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f3263b62a1294ea79a9953c6aab36463 : SUCCESS in 1h 53m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee459e3d665244f2bee5c60fdbfc6032 : SUCCESS in 2h 03m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0c27eb19cfe745378bf9bbb076f96d96 : SUCCESS in 2h 01m 14s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ccecbea5581748289b80376a13b62554 : SUCCESS in 2h 24m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/b795336fb72746c4a057dd0348136440 : SUCCESS in 49m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9df38ee3c829495eb18f64388e773c4d : FAILURE in 1h 56m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/fdfdf993a7f541109cdf890ccb2e70ca : SUCCESS in 1h 57m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4a5257215795434b9f312239e6ddfbed : SUCCESS in 1h 07m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/209ced94a25446d5846b770769cff75c : SUCCESS in 29m 51s","accounts_in_message":[],"_revision_number":1},{"id":"c690252996f3dc45f78d517c9ff70f8c761856dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-20 13:22:12.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ebcff3d913d96391fb084b62c6576239306915a5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:00:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e9fd1e95a5142ad91c2c44803ad92a5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/745ccafe70894fedb1564db7a3783414 : SUCCESS in 11m 21s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bfec2f873f0d4ed4a80b00892a7ef89a : SUCCESS in 10m 18s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d5d05df35120560e934dac9ca684e90c6f9acfa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 15:31:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/12d10c140e8842bf9659d10c168201d7\n\n- grenade https://zuul.opendev.org/t/openstack/build/d70c08e889034734b6d425918e38825b : SUCCESS in 55m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0f11b9204a824a048d10c56ef3d6d5bd : SUCCESS in 1h 21m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4a16e123128f4cdbb18dcc7b3bd2c9c1 : SUCCESS in 1h 06m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f0016501ba04b088344ffc31494ae41 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b4daec62bfd2403ba8952541d1e6bad2 : SUCCESS in 6m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3fbdb60204ce4279a8252210ee6c08f2 : SUCCESS in 6m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c96929d48054b5e8a366b3dd1fd2bb9 : SUCCESS in 6m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33303347d6d14aa6a3ab82843c0b3d69 : SUCCESS in 5m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1d5b8557ce5b4cf88342fbfe61c9d458 : FAILURE in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57cc6c7be617494db2fd10b113ecf6ac : FAILURE in 9m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9a641571baa14908b6cb4360f7e221ea : FAILURE in 14m 47s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8ced9fb7ffb44be0a31c7d85d1f772ef : FAILURE in 11m 01s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bf303b2171da482391aeb39671fcd627 : FAILURE in 1h 56m 18s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6faa5f3132574c63b9fb51ffbb60e7bf : SUCCESS in 1h 43m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/005ddd7c8b4346d494b0efca8f8213c7 : SUCCESS in 2h 03m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2e3816924fb14739bebdd5e3d8687b22 : SUCCESS in 2h 07m 00s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d5ef341ecd874658be81df01eaca75e1 : FAILURE in 1h 54m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/65ad240d9c584054a269c5ba16d848c0 : SUCCESS in 56m 43s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2a6460f9852046eeb5954b649f32977f : SUCCESS in 1h 56m 55s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/2585ad5fb64d44ac95412be0585b4edf : SUCCESS in 1h 56m 19s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a401b5a3291a42d6b764facc30d9a98c : SUCCESS in 1h 00m 18s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2b5e2434277d44fdb4d01acf08aac7eb : SUCCESS in 25m 15s","accounts_in_message":[],"_revision_number":2},{"id":"8aaad0e5b8733cad999afc1e4a75d320a3aa69c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-21 12:44:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"afacc74deaa529f0c6129d5c0c88c6c8a7e1e2e1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 15:52:14.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/78c5d1f02b7942af9e8e7b36c58262a6\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8153b9b200c4dd582c5a31f5f283faf : SUCCESS in 56m 07s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a2691ba101cd4ef4b9c5a20c5b2d8255 : SUCCESS in 1h 44m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6b7ac9892d9a4e1e818a315998b71e49 : SUCCESS in 1h 05m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd6f17a1e9c54f46bcc27acb16cf95d6 : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3da3bdbceb724123a3b6c57194be8f5e : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65bb554bf5de4860b45604799c43de4c : SUCCESS in 5m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6ff8ebab8b2c46bd86e6718fdfee9c45 : SUCCESS in 6m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f293ad5e80124b39b4394d1ba8609334 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1f30a3aaa0da4410b758910d62147acc : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36f182d33bc541669015d68e6bad4332 : SUCCESS in 7m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5a4c5ff8a3cf48f9b6bdbc25cb2bcea7 : SUCCESS in 12m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/049a9521fd7c47ce93e45234f29f0636 : SUCCESS in 8m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9bc47a32df08407eaa684955f32896a8 : FAILURE in 2h 01m 30s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/703860dd7b624c22a56e967cec574b78 : SUCCESS in 1h 48m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6547c55796894510a66003e9856dccc5 : SUCCESS in 1h 59m 49s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d304dd8562a441997890447a01e2859 : SUCCESS in 1h 46m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ee0fddec86e54c28b4f0aa496e44b970 : SUCCESS in 3h 03m 27s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8b000e8093d94867a91121ca2f6662fa : SUCCESS in 46m 23s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e46ff1696c0042e2b23fe4c33174917e : SUCCESS in 1h 36m 45s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/a2fbe3b085b14cb1b79b9b752b18ed22 : SUCCESS in 1h 51m 50s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7b2c1d9498fe466c926edbd57d42c4bf : SUCCESS in 1h 02m 35s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/524733cf97f94ab48537fceb2f1d3a3d : SUCCESS in 25m 07s","accounts_in_message":[],"_revision_number":3},{"id":"3c1f448261f1d4e354da37d1d38930b6775aec8b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-21 22:15:07.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2013662e2fa7435888260f5516da0033\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ec907e9669bf43f89832f3349368725f : SUCCESS in 7m 16s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/58654d47483b4413a048b78205f1951f : SUCCESS in 12m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"dfeca7325e67962fdb5c9bf993557d716949b479","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-08-29 07:07:54.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"6d987a407000daee2c3de5f901bb2d172b19047e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-08-29 08:10:10.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"ddfe8a739659d47e5b041808c8ccb084af0356b6","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-08-29 11:26:06.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"a10d4c3f32820542d1b73fb03c271b89c286f692","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-08-29 14:23:30.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"39f243be28ae0e20dd14998f6e4461439c393252","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-10-28 14:20:07.000000000","message":"Patch Set 4: Patch Set 3 was rebased\n\nCopied Votes:\n* Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"680daad13c81976169eb598736d4a6f17c9ac451","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 14:32:03.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c647c4b9133048baaf201773eee66186\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/825be88288754eb6b2932cffab230671 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69e1297e4c6f470b9f8bb760c0eed002 : SUCCESS in 9m 31s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"93aaf047e62397763f9ad87a4b8fd4a69d8783a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:18:59.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/66cc50725cde433abc2ab8fd7266bce4\n\n- grenade https://zuul.opendev.org/t/openstack/build/512dc12af0fb42a7bc89e4a0290658ed : SUCCESS in 49m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a89dda605a7b45a49a22163a56187d92 : SUCCESS in 34m 02s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7d48b68309b54051bdab826f274a257d : SUCCESS in 1h 23m 08s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/48f74ea4e7e1492199cd3f6fe93ecddb : SUCCESS in 58m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd1178c82c7940ec9961495e1e9f8c61 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e9da98fe58aa47c3836aa8b514359e08 : SUCCESS in 5m 10s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e343ddefe063460d9813caad9450740c : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ccc6730be7ed4ca98b248a501877a268 : SUCCESS in 5m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a12c98f5ef894c3884c7ced7b5a1ee31 : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0f1143e702714b09b2f2d35562219bf6 : SUCCESS in 7m 27s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/6b4cc07970f644c8bdeecb2c0467759b : SUCCESS in 11m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3d6d7fa2dadf4bc9b55628a669ff5352 : SUCCESS in 1h 38m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d49a6126e8044b9681e45f46fa3453e0 : SUCCESS in 1h 17m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5213ad4275124e75beff0409cc8f03a0 : SUCCESS in 1h 50m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/39251b84ced74c80a61f017a376b6f7b : SUCCESS in 1h 46m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5aafefd5fec74cea8ccda06c78f01794 : FAILURE in 1h 28m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e89b564f6aae42d8ad0b4ac3f22fcf77 : SUCCESS in 40m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/80e05d16dc844bd98131f7f1a6e3ed9b : SUCCESS in 1h 35m 33s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/85301a92f882423fa8a3c6cb085a75d3 : SUCCESS in 1h 34m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/be19e32adc7c4faebb0b4192b4487a1d : SUCCESS in 1h 01m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8bd20eab277f4a9a8062e53909bdcd7e : SUCCESS in 33m 23s","accounts_in_message":[],"_revision_number":4},{"id":"c36cb49ca9d1d0a92eca04d678fe4bab377e979e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-10-28 16:31:48.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"c3899943309b9cb9d3be353b76fa447eac420da5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:48:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed966ae1285542b3bfb46e5396e1af67\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/970510011e484145a903fac9073d5f94 : SUCCESS in 14m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/470cd44ef5894cc98491511506fcddef : SUCCESS in 11m 38s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"b1f016aebe10d5d536bc5021a83d06073417ed58","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 18:29:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/762a84023a794285818f79983e2875cd\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8b339c447c548d3966076bab4355447 : SUCCESS in 1h 04m 17s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/5280cbbb8a3544bdbbca2797d9a1e07f : SUCCESS in 1h 02m 01s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/1111b9a436944cc6903df46d951ebd21 : SUCCESS in 1h 33m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8f1696e631b1459aa92933a039f38f77 : SUCCESS in 42m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8dd236cf0fd74d6f8faadff33edfa8ad : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/65d8afdedb404837894fc33e81383170 : SUCCESS in 4m 59s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/366aa38d2a7f45c593fbdab501fe82e3 : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/166e5991fd294313aecb65cf7ab23055 : SUCCESS in 6m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7caa4fd1f9eb42ee9066e8e9348e30a7 : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/42311f34280d4d9e86cb1c2575f12de5 : SUCCESS in 7m 38s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/e8fca84a821b4432bcc0a8d60e2567de : SUCCESS in 12m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c32ab22c52c24e1185a8690d559984a7 : SUCCESS in 1h 51m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2775c4d26b4648ac9f041084a62f3251 : SUCCESS in 1h 51m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2cb6a82409a048d890e8f027eceed5e9 : SUCCESS in 1h 39m 04s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/b331b74aa9a24418825956305f8b71f0 : SUCCESS in 1h 43m 06s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e313243cd4ba47ed9b82df600a224a61 : FAILURE in 1h 55m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f5d67fb3763f49598d083d4042ab6e2c : SUCCESS in 53m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ea33812c07cf432595615613aa2b1f6b : SUCCESS in 1h 39m 25s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8eb90c9a26e444f584bf7b06d62333a3 : SUCCESS in 1h 47m 44s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c5e2e6c3aeac4454a0c10090d3fb284e : SUCCESS in 1h 05m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/774bde2f237c4ed697a705ff496dd9b7 : SUCCESS in 36m 38s","accounts_in_message":[],"_revision_number":5},{"id":"38ecc283aeee39d61262e40139000ec5ead5a20e","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 12:50:54.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"bb6b13f72d17d9d59f06cedd679d5e71577a4fe7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 13:32:42.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"e2bc2c18c05bd0d17071d608b2b3e14ebdd93108","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 13:43:57.000000000","message":"Patch Set 6:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a25ceb64d354ed1816e2bb19cd422a5\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/4f762ee2661c4d6ba0f0e1b185593652 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/8651bac6aaa94d9297ecb7dd5ce0548b : SUCCESS in 8m 57s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8b6a4fc1dae6d7233b926ce0d54ed2e9386605e8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 14:51:59.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"ff5b088589e4e36530d133a28034d1868a2e3f8c","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 14:53:39.000000000","message":"Patch Set 7:\n\n(3 comments)","accounts_in_message":[],"_revision_number":7},{"id":"943a92f287b64e88fb79d2f056c9bfe7dbeb95f6","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 14:54:33.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"e64e434c40c3a2fb36f9d4b321a0cfc2b620769a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 15:02:38.000000000","message":"Patch Set 7:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/722bc6dc6f014654b20c49c46b331063\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/004ae160e1ec455fbbaa756a2ef9db62 : SUCCESS in 7m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/9fad5570b40c42ee92e103f31000f9ff : SUCCESS in 8m 48s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"79dbed4eb18dcb4611a33202e447bc17f435e59c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2024-11-11 16:27:33.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"9fbebe3848f75d7091b150f8a38653ed18e341b3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 16:37:23.000000000","message":"Patch Set 8:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/152b98b559134ee0b6582736ae69786b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/f23c96f51c5a41b383b4473e4efe2ad7 : FAILURE in 7m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fde582ae0a74f189c08e14c8e9d9faa : FAILURE in 7m 42s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"7853b8688f75aca36018793005bb7511e1d63c20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-11 18:40:27.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/7035c34404b1477990c69309a3670942\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/956e1f127b8f4f5e9beb6a4f242b41f9 : SUCCESS in 2m 52s\n- grenade https://zuul.opendev.org/t/openstack/build/2b4081ff09a74e01be2d137722d77c0a : SUCCESS in 57m 14s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b9cdf6557e034e1d9db79315606f9d3e : SUCCESS in 54m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7db1f318f72a446da0aa021114a543b7 : SUCCESS in 1h 32m 02s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0c432776063e4f078d3f5e4d04cb8f45 : SUCCESS in 1h 09m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b716a34b91d40d49a32dbc2ce2b624e : SUCCESS in 4m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c476b17d151c4d2a9d94a2f35c2122c0 : FAILURE in 6m 12s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9efb7a09e874ad2be688610d61512e8 : FAILURE in 7m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7d0f7abf11604166a08dd95578476e82 : SUCCESS in 5m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23956fede03e45f4be62f7bb11d429a7 : SUCCESS in 5m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a666c8814a5a40438f61d6022473bd2c : SUCCESS in 7m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/858d7495fbaf4dcd9d2ede1c9b90ec00 : SUCCESS in 7m 57s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9d2b89dceac54b9b901f646365fb9f78 : SUCCESS in 12m 31s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d1d182f547e641398302c952fd7f15e9 : SUCCESS in 1h 45m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/77fea8df3fe94c5d8363151a07b3274b : SUCCESS in 1h 36m 06s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/29957e0c3ad445d3b5caf1e2bc7d3d37 : SUCCESS in 1h 56m 29s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/a19713dc4e0d4b43a13f8425b4d5f60f : SUCCESS in 1h 39m 11s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/5baee229820b4486a662b841ad5fd27a : SUCCESS in 1h 51m 30s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/47879c3997634d65b4804e5736cb5fb0 : SUCCESS in 49m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3be8cb44ea5e440bb4633ef84528a9c3 : SUCCESS in 1h 47m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b98610df578c452792e2ea2751b8cd70 : SUCCESS in 2h 06m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/992cf62bddce4f9fb93e4a69a4392a25 : SUCCESS in 51m 16s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/40a337fd7cca4651a1d7a9844b285342 : SUCCESS in 22m 08s","accounts_in_message":[],"_revision_number":8},{"id":"78915a4acc12a002bb6e7de8a0ca7b8689919b1d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-11-21 09:04:36.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"d9a062649ba61f24d5c40344758c938d75ddb836","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 09:14:53.000000000","message":"Patch Set 9:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7c06ac4c2a5c47fc888a29379998b100\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/346a381491904c9bb719a786f3025196 : SUCCESS in 8m 02s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69b8a188cbc6420bb8de92260bf87b8d : SUCCESS in 8m 08s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d2e1bc0632f1d2fae549136e03faa4484b76151","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-21 11:07:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/313d4c857e4742c2af8e1e5b995c37c7\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d30e98c5570e4aa798ea40609325f885 : SUCCESS in 3m 38s\n- grenade https://zuul.opendev.org/t/openstack/build/e19028a643414199a9e42a673c126099 : SUCCESS in 1h 06m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94be326cc9a44822b9dccf6372f0a659 : SUCCESS in 53m 35s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0c10b46d880e4549b61536f558bb10a4 : SUCCESS in 1h 16m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd4995bef4a54000bf4bb7b2b140b24c : SUCCESS in 1h 01m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24f49075934842d3a4af80d8b66f7ce8 : SUCCESS in 4m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/507266b4bd8248359e3bb82594cd3056 : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eab3458a936c412db2380d43031f2c23 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/952ea114fdf743b19ed5e95d38c1c03d : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fac5f2e01d614e00bfd6f1c574458df3 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9fe321358009488488087043ad62404e : SUCCESS in 7m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/152f8a461a9c4828a9b00d9a1c96b1aa : SUCCESS in 8m 11s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/46f4ae3e5054452f891ed0e425ec9171 : SUCCESS in 11m 14s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3a71972ed2c947a6a29fabcd3c03e03b : SUCCESS in 1h 51m 28s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1840da5b8c354402a49b93f9abfdd0af : SUCCESS in 1h 56m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1cfacfca782f46b0bc43586e690a1368 : SUCCESS in 1h 58m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc4d70a61cd54784988f8a372f55147b : SUCCESS in 1h 58m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/bd1421caa953431995be308f0ca9f995 : SUCCESS in 1h 59m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/83a09ecb8ce44a0fa6384efa66a1ffc1 : SUCCESS in 1h 00m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2cebd9271901489e8afb65ab6d2262fd : SUCCESS in 1h 44m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/dea86d1afc724baba6d9b11a319a6f9d : SUCCESS in 1h 48m 59s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7769228476ad4630b9be49d7c2ebbf5b : SUCCESS in 53m 34s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/95aa3c3a347249ab878ffeb6502766c5 : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":9},{"id":"6140f6a24fcc4d15541da9498f8a754b59d6d035","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-22 09:19:31.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"2450a18a0dd1e2eccd45fe5735e55c9f058695e1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-22 09:40:59.000000000","message":"Patch Set 9:\n\n(6 comments)","accounts_in_message":[],"_revision_number":9},{"id":"231cef6fd47b386e8af1ccc60f22128022776bf7","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2024-11-22 12:55:51.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"1f902e0fa2862ee81265ba596d6b8d89e708417d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-22 13:34:34.000000000","message":"Patch Set 9:\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"155d19a3f61ed534acf19ec7c737facb210d416d","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-12-16 19:46:55.000000000","message":"Patch Set 9:\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"662e15f5736f3004b18fadb51e003b9bdb7b476a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-26 14:58:54.000000000","message":"Uploaded patch set 10.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"3bce4e940a40af422328e0aad594987f068f0c46","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-02-26 15:07:01.000000000","message":"Patch Set 9:\n\n(5 comments)","accounts_in_message":[],"_revision_number":9},{"id":"a8957a3cae3d8e348103fa247f953b126cb31dd2","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 15:18:28.000000000","message":"Patch Set 10:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8965bd105794dad9fffa58d320f8d26\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/25cb014b6870451ca6902fa8db00c4ac : FAILURE in 11m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4148f7d988f14e618100feae0f5c97cb : FAILURE in 12m 45s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"aa18f84f99d22359230ecdb2af36f14ad3db6580","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-26 18:11:39.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/94e69ed0fc0e42d5980c70dcb0c8b6f3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5075bebdde494cdb98f7ca25d5f7d935 : SUCCESS in 7m 44s\n- grenade https://zuul.opendev.org/t/openstack/build/745e91fd01b4452d9bcf63fe4c8c5106 : SUCCESS in 1h 02m 49s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a7c26ce5ce64403b9070185f59418463 : SUCCESS in 1h 03m 16s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ebac25169ab847a2b69426bea5501f55 : SUCCESS in 1h 31m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d29195149acf43df86bf34f160cd0ac7 : SUCCESS in 59m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8d00012168e40db9cc3c786c6a4bc79 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b096511fdc04d3b84a6f41a99878567 : FAILURE in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3da3d33f90c446919f6747ef25e3a0d8 : FAILURE in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd71429ad52b4702a0dd1be9f28fa4e7 : SUCCESS in 8m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33efc765ca9b47aaa2f93c452a10c437 : SUCCESS in 5m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e7e60a011b1a4319abc161fe5c708112 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/83a7ac5e72ad4c3787a68d7089572177 : SUCCESS in 9m 06s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4a671c81c7014d4ba495a6da0d26461c : SUCCESS in 12m 15s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c72c17a574dc42798f22c2965671ad15 : SUCCESS in 2h 35m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c49f2e5c202049adbdde3c2b46cdf4b6 : SUCCESS in 2h 04m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4ca5c7f5c67c488eb7bd5edd187e48cf : SUCCESS in 1h 47m 08s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bd6a14eb160644c488f30920fe0653f4 : SUCCESS in 2h 18m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4f722086b11443d9b057bc78f860ba3d : SUCCESS in 1h 58m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/31b88c6f088144ef87c4592407fe5b8f : SUCCESS in 48m 33s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/3123f395ac4e46adbdcf46938f708e78 : SUCCESS in 1h 53m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/972a6c9d4e124eeab5d4116c8648b1ee : SUCCESS in 1h 56m 54s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/c1bf272f0a034b14b6ea8cc145a6762f : SUCCESS in 1h 08m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2b9d21016caf4330bb995a295f7d9e02 : SUCCESS in 1h 14m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5ed4cae9dd01499899581713febf36b5 : SUCCESS in 32m 47s","accounts_in_message":[],"_revision_number":10},{"id":"3a27aa86c0732a3ec0c8606218bb848f1b3e4d19","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-28 21:45:06.000000000","message":"Patch Set 10:\n\n(1 comment)","accounts_in_message":[],"_revision_number":10},{"id":"f59b8664e8786f06d661da0620e042ca75adf148","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-02-28 21:46:50.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":11},{"id":"715fbfa57c76f2708bdcd6f60f470f8a2e1798a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-28 22:00:17.000000000","message":"Patch Set 11:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80d3a99319bc408c8ca3afa097208658\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7d20cbade4264f81a7f1bb9afd32a82b : SUCCESS in 9m 31s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/871ed5900c9d4696a9951bfbdf697c8b : SUCCESS in 11m 25s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"c0520ff40542b4ae827ddea633fc966bf8000f78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-01 00:20:43.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ae4f4119b11402bac86991c05b723a5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/73573489595144c0ace6da399294b1c8 : SUCCESS in 4m 11s\n- grenade https://zuul.opendev.org/t/openstack/build/c86e7400bcfe48bcbb03ea9b3cc74280 : SUCCESS in 1h 01m 04s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/94a80efdd04c4de999aa70287f92fc9c : SUCCESS in 1h 00m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ff268099f55448418e52636fb840c6f4 : SUCCESS in 1h 39m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ea0198696022486286e5f2ce7152cd38 : SUCCESS in 52m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d248da8db59341bf88e5763d9b3c5923 : SUCCESS in 4m 47s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e7dc7ee683e4237be0c5a6994d9abb5 : SUCCESS in 5m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0edf55c8c96741608f216f58d5b51e8e : SUCCESS in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9c7102a707d4c61a52cf2fc9fbeb35f : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e3f6b906c22b4cc08cacde1e44cd5c2a : SUCCESS in 3m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/19e11feec04547f1aad8a8dbcad27007 : SUCCESS in 8m 17s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f73c9028ac06452189dc1bd2126622f3 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b3f5f0bd717c4ce8a3b91c82173ece82 : SUCCESS in 11m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/251ee312e7494554a065bc6459fe96a6 : SUCCESS in 2h 04m 43s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7526a0e19a2d4d2fad6bfe906b22c83d : SUCCESS in 2h 26m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/62d12e9fb886499c981dcddf0b8ca45b : SUCCESS in 1h 58m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/352338de295249b9a6ceac764723c5cb : SUCCESS in 1h 49m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/699e95e2494e40d6bc4677a620bb3020 : SUCCESS in 2h 10m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ba42223e04cd4832aa691a37aaeb68c4 : SUCCESS in 1h 09m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/945e847370cf47d59bbacb85eb109831 : SUCCESS in 2h 06m 47s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/bc9bbc64aea447bab12d5c57092f5465 : SUCCESS in 2h 15m 16s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/f3f1ada25f1545e283724c4fe79143a6 : SUCCESS in 1h 02m 53s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3add9a204f6e4f70a8a3371b1fb542ed : SUCCESS in 1h 58m 51s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2a87f60c5ffd43f5aa80a5b6ee7c1f5a : SUCCESS in 27m 34s","accounts_in_message":[],"_revision_number":11},{"id":"b592f2e4102bc338fec4c45df675ce4a25ccdd83","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-03-03 13:25:12.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"b99ff6cefc4a8da4f8703f054540466f8ef5e095","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-03 14:24:15.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"af782785af1751ad3d77dbf7c3806e2bdf86f21e","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-03-03 14:47:49.000000000","message":"Patch Set 11: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":11},{"id":"2ba0ca5edeecc3c24e9b1be728a5aaf972cd5b5a","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-11 01:45:14.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"1ddead510e9df0443a80a7c0c3d5518dc2e2d3f3","author":{"_account_id":27615,"name":"Rajat Dhasmana","email":"rajatdhasmana@gmail.com","username":"whoami-rajat"},"date":"2025-03-11 01:49:20.000000000","message":"Patch Set 11:\n\n(2 comments)","accounts_in_message":[],"_revision_number":11},{"id":"3befd7ebd164c0f1378851986b63cd21a58e916b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-04-09 09:01:12.000000000","message":"Uploaded patch set 12.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":12},{"id":"020b38e012766efbb20a9a326c1d91b3034ce3bc","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-04-09 09:06:06.000000000","message":"Patch Set 12:\n\n(8 comments)","accounts_in_message":[],"_revision_number":12},{"id":"6124a70a673d544c21d712684365aa17c1e487d8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 09:11:14.000000000","message":"Patch Set 12:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7391565a1eb54314969bf794e76a2273\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/aa39af6fd37244fab9372522f20d774d : SUCCESS in 6m 50s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a3ccede4600340e491c477d73e7553c9 : SUCCESS in 8m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/a6d0b706cfe147ec88abc627d1da7946 : SUCCESS in 9m 38s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"c97955f07d41b19ddce263602ef9e12175e5ddf1","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-04-09 09:36:10.000000000","message":"Patch Set 12:\n\n(2 comments)","accounts_in_message":[],"_revision_number":12},{"id":"39d99e4fe3cfa0b5dbae4b88a074e89d752aa11c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-09 11:09:07.000000000","message":"Patch Set 12: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be32dcd24095486196100102a06e1283\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2b2440214004cfab2feb0143eac16e2 : SUCCESS in 2m 32s\n- grenade https://zuul.opendev.org/t/openstack/build/15478da4b37b454fa5f918612401ac61 : SUCCESS in 32m 44s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/471c2e947a264ef8b38f9644c05d5790 : SUCCESS in 28m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/051500d587b644d392ff955cbf3b9ade : SUCCESS in 1h 32m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/881c3bd6bd61406aaa0a83d7c2736524 : SUCCESS in 37m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aee0a1e8fa4a4d6d9c71478092fe5a7e : FAILURE in 2m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/47c048d05b474bdfa3ffac06fb9cc40f : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0e16cc90e10d424a9622539431085ef4 : SUCCESS in 2m 59s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/fa339bbb544f4d828dd9d09265c12c61 : SUCCESS in 7m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c603752cd99e4117b072883c6742b943 : SUCCESS in 3m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7037ac3dec0d4286aec33b8e6ad9bf68 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2b74ae67fb054e1dbbb8bcbe62490e03 : FAILURE in 5m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0d402471e0f64d39aa0138840236f0fa : SUCCESS in 8m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/d838bf4765a140cdad8e9057d182e489 : SUCCESS in 10m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/52311f8d316340219b7996d2a677fabc : SUCCESS in 1h 28m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f5244a7090ad4847af9eed576206754c : SUCCESS in 1h 36m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/749fd9167b45435abc8558ba3d64e03d : SUCCESS in 1h 51m 43s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/ce7d776e204f4b458920a94d94dba2c1 : SUCCESS in 1h 54m 17s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a3111e5a589b49ddb4ba8c2f28df44b1 : SUCCESS in 2h 06m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ec738b4533144cb0931dd87c918f827e : SUCCESS in 1h 02m 09s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/010e7e96be184fa2a21a6c5912b1e0d1 : SUCCESS in 1h 57m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/05fe39e2ff3047119a2bc6c06506b0fa : SUCCESS in 1h 55m 36s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3162001931d443bab7d3d2f207fff384 : SUCCESS in 59m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/a53736f1b50440049cb2c25395fc8cfe : SUCCESS in 1h 44m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/54aa5564d23e4e9895fc737d008df4b2 : SUCCESS in 25m 00s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cfa0431a76b340f79897a66da4d79e63 : SUCCESS in 1h 01m 54s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"8d7b335173fedda6c22ca69d8c95a1519f20cbf0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-23 15:29:02.000000000","message":"Uploaded patch set 13.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":13},{"id":"79c07f17c8b74e93a2a2a6a27ada8d0225a99146","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-05-23 15:36:09.000000000","message":"Patch Set 13:\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ed1d1257e485c0fa41ab7c86f3ae67afb376ec80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 15:49:53.000000000","message":"Patch Set 13:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e529f526440d440a884fe0927ab8c194\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2507cb3c76484b679069d326b538a26f : FAILURE in 6m 29s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2fce37402cb6408aa25bac0838577648 : FAILURE in 12m 43s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/7b74235995cf4e088e5619daab23dd26 : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"7282d1f368b5199dfd4a96f2c6bef0130c773bfd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-23 17:53:52.000000000","message":"Patch Set 13: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5dd5c1e408149f08ee0cad5af4b803f\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/2291929f837c40e1aba8705d1456f318 : SUCCESS in 6m 07s\n- grenade https://zuul.opendev.org/t/openstack/build/34ab1acfee46455cb353aae4140bff52 : SUCCESS in 1h 05m 31s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c36c6c8857224f7bb50b7c2cb11f5f35 : SUCCESS in 1h 05m 33s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/20c3cc6a41794907b4a88f60f8f95d09 : SUCCESS in 1h 36m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a5602ed4393a46ada99ee0b9d23a8af1 : SUCCESS in 1h 08m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9316c569102843b2ab04ffeb1a3514e2 : FAILURE in 5m 15s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f9b3375b128f42778ae207534d6854af : FAILURE in 4m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/13588f6f62bf4397a0ca8268adb113b9 : FAILURE in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b4a627b52f3b458ea437cdb0cc3bbb76 : FAILURE in 8m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8fc92147e8d7459a9c84bf502e80e262 : SUCCESS in 8m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/982c3a70f3f74b4290edb0ff2047bc11 : SUCCESS in 6m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0a3a2ebbcd9848c58cab9c244fcaa851 : SUCCESS in 7m 07s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/878bc8854ab648569f4e9612df5dc9e4 : SUCCESS in 9m 58s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e6108679c4b5453980f2ec6716fccb89 : SUCCESS in 2h 01m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dfdd9f4ba43045ab88d05a139540c484 : SUCCESS in 1h 51m 02s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/1dd52a69d2a04e8197701d06653badb9 : SUCCESS in 2h 01m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e5c7d0c4451b4caca5bdcdde236207b0 : SUCCESS in 2h 18m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf22c3b4b25f4c30a5cf35441eeba73b : SUCCESS in 59m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/34aacf8bff2344edb373fffbc01ad4f6 : SUCCESS in 1h 55m 14s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/278c309d6db3474b90297747057a0f85 : SUCCESS in 1h 55m 57s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a901be3477da4b3abde892851864eba4 : FAILURE in 35m 07s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/211fc98c031e4903b151b6d0c690a402 : SUCCESS in 1h 48m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6ff4c50c68a24cd09e3defb9c9d1cbe1 : SUCCESS in 28m 30s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/b8fc37a8b61c468cac5c8bbd8c9669dc : SUCCESS in 29m 36s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"586e2b37acd46cba36901445a1e8e426c027d6d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-06-16 13:35:55.000000000","message":"Uploaded patch set 14.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":14},{"id":"b7defab6477eb80c455182ac892193cdbe843ffe","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 13:47:59.000000000","message":"Patch Set 14:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/17ff48fe0ea7446ca518c140538efd40\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/594c0a57ed374c918ea43348b754f0b6 : SUCCESS in 6m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/225dc7559bb84e35b279b5015d7b4e2a : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/02e8740f4ad94a74a812d17786f1c52e : SUCCESS in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"5b61403b7580d154c51329c7838cec3d4f8470c0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-06-16 16:08:24.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/85f88c86480c4577a3773e9563cb6523\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/a975bda05cbb46cd94983a3f03cc8c08 : SUCCESS in 4m 46s\n- grenade https://zuul.opendev.org/t/openstack/build/10d8c5a8654546b2a3625283a5b4f113 : SUCCESS in 59m 43s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/b018a8abff7a4646afc5d2bb2167929b : SUCCESS in 43m 40s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/973c206cc78f4f98a6380d1685cbc4e5 : SUCCESS in 1h 44m 17s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/798beeecb30f4dcfb3fc61eb68f989b1 : SUCCESS in 32m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9b2dfab3928241858be372bbd3d7620c : SUCCESS in 4m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cb1d5d0659064f29861dfd950b1a45a1 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83dea488910d454c9450c1bfdf8701a2 : SUCCESS in 7m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3e5f237418ef4a549a36987e17d5c4cf : SUCCESS in 7m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4579edc9878e429f9c5fdf845985923e : SUCCESS in 5m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bcca6e71619043dfb4f14240b0d2ada1 : SUCCESS in 6m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cda8532f224743ddac5ad6a66f74edbf : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/62d67f6812cb47198c5f2f797edcd7fc : SUCCESS in 9m 38s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f7c95bafa8a543b09aa6497c073f4ec3 : SUCCESS in 1h 47m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5be546c16e764d6ab8f3f4de14448f7d : SUCCESS in 2h 07m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0466a96277bb40ec937e83e3b58d21d6 : SUCCESS in 1h 40m 42s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/32d5b5eece2f4f1892aa228f89b4d83f : SUCCESS in 2h 27m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a75dad1805c34d1494ab9d773dfe8f3b : SUCCESS in 1h 15m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/557c7062137149308aacc5c8afbd3e71 : SUCCESS in 1h 22m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/15e1ad36d1a0485d86e0bfe6503388e2 : SUCCESS in 1h 51m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3af27dddfd4347ecac601d8bc7dbc5b1 : FAILURE in 27m 47s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/83bb8b5d75af41c3bf2b71c932e2748d : SUCCESS in 1h 52m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/da82c0154d2040da891ff4bf15b02716 : SUCCESS in 30m 54s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fe362ab8fa4b4a9d961aa4c48bb88309 : SUCCESS in 53m 06s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"d442317bc1d94942abf25ab42427fd02cb0b1a4b","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-19 16:40:01.000000000","message":"Patch Set 14:\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"8ec7fe7cc3bed41923fc27965f6df587a54938f4","author":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"date":"2025-06-25 13:04:55.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"7f068464944b1724807ee6b33d28cb928c6c0df6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-06-25 13:50:25.000000000","message":"Patch Set 14: Code-Review+1\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"beee4ffd00e0c0933c425f4b0c5884b740664c53","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-07-08 13:33:39.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"63105792bf2e4401d3850e676f9294d058eade60","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2025-07-08 16:26:37.000000000","message":"Patch Set 14: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":14},{"id":"9040542b4f323621b43f96206dc5bb9b88eac908","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-07-10 13:59:12.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"0a6e881c9084f3293b084ec0cd0c54683aa2a431","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-07-10 14:00:21.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"501df8b1dcf0ca66696e17549e75f1e26a3f6587","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-08-28 13:39:28.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"f30834aff89a5ffc3e6e0a48255f660b7598c85d","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-08-28 15:46:16.000000000","message":"Patch Set 14:\n\n(6 comments)","accounts_in_message":[],"_revision_number":14},{"id":"6e769f3149758e0873e7c7e72bb3ab6b0e08ea61","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-02 15:11:48.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"10dca2b03c23b871f382111c90e6abcd6faf0cd8","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-02 19:25:57.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"d6998b3f1729e8383588d545c88a7d14f93b7b55","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-03 09:56:16.000000000","message":"Uploaded patch set 15.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":15},{"id":"f332d9c7f860d3386a9e1d221f00199ea96657c1","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-03 10:00:18.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"7b84d2f49beb35e28efcb0415f17edb3e4691596","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 11:27:55.000000000","message":"Patch Set 15:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2d9d09b6ce04ea0adca8e040f554bc8\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/2724b3556cc5468c87bdeb838a08b24f : SUCCESS in 10m 06s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3868bd93edb541c1a351920503026300 : SUCCESS in 11m 22s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/b77c9fb55a254883a12954458aab6031 : SUCCESS in 15m 56s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"9ef754415da6e506a28dc9134122fcf31e1f89b5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-03 12:12:44.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c4f3298307344ce6b7fdb50669998e9e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/3e89e9048591415484602f03101b8a8e : SUCCESS in 4m 00s\n- grenade https://zuul.opendev.org/t/openstack/build/dffe8a49f4d243d48a39133a79a3a049 : SUCCESS in 58m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/d4bc8caf74ee419e8f004be407eaece2 : SUCCESS in 59m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/773518d24ee441e2b26889db996ce468 : SUCCESS in 1h 34m 18s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5795793e00f946429aa09e0ee4e1591a : SUCCESS in 1h 00m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7eb5a32f8ac64243bc33835744a67730 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b36b7c4549474cf08e9fa960ba37493c : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b9e2472bed1e4a8eb6b40c87f448e622 : SUCCESS in 5m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/353f4600fca247b3b183d6d2ff84cc3c : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc9e9c832cf84ea3b9f8392f48c09c38 : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe1e0e26cc5f4b798377b8ae2c279f39 : SUCCESS in 5m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6cd45bb1e7324795adb662e5c8e22ccd : SUCCESS in 5m 54s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b7b3d34d430449788c1b388989bd59e : SUCCESS in 10m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a8230115929e46a1a056b3cb3ca3fcfa : SUCCESS in 1h 55m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c13d9bbf2d8747648e8a9443303a688f : SUCCESS in 1h 48m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/354d79975a1a4198a463622f3f133f87 : SUCCESS in 1h 57m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c25fe9c0e7c84583900b0dbf6f3145e0 : SUCCESS in 2h 11m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fdce893fd35d4399b3bb70b69af97650 : SUCCESS in 39m 28s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/6a445d15199749fcb8ed81c9f3f614b0 : SUCCESS in 1h 41m 44s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e811929c4bbe4103afb731a5caf67fd0 : FAILURE in 32m 04s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/480ae7ef658f4ece8e0df041bceb9437 : SUCCESS in 1h 46m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/87d5fdb3c693419e9b395a9662c1bd24 : SUCCESS in 27m 57s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/de7f99b9c4ad49ea9f9bad23b7f79ebc : SUCCESS in 36m 04s (non-voting)","accounts_in_message":[],"_revision_number":15},{"id":"de59a2479ef58216f97a7c54466dc6d697cf9f64","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-03 14:10:19.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ef7b8fa4f9c170a2c0874a4ab9bfe067ed62abf9","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-04 08:37:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"00357639d573edcc6147d87a955c38fa3f3b0ccf","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-04 13:35:15.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15},{"id":"4d3d4d2cc3a0875e59e7f6a283ccf4d893f69418","author":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"date":"2025-09-04 19:52:06.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"293a484e19476e0632bf9bb980f18cad74016bf6","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-09-04 20:01:52.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"e5f5bfa83904ed690a85bc84bf7e8c3b466f7034","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-08 15:51:46.000000000","message":"Uploaded patch set 16.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":16},{"id":"7526bad04b41cbecfe4972c5a47a6c565feb4103","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-08 15:54:25.000000000","message":"Patch Set 16:\n\n(1 comment)","accounts_in_message":[],"_revision_number":16},{"id":"935696075d27eda72d3c30ab22b6a0abbb7034e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 18:19:31.000000000","message":"Patch Set 16: Verified-1\n\n(5 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b69e0041a214c8e9d7e35d216f50f71\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8315bd1cc83944e1aad1d8b34b38fcf1 : SUCCESS in 4m 31s\n- grenade https://zuul.opendev.org/t/openstack/build/910d5e8c3b5a4f6e9c064650ade92ed0 : SUCCESS in 1h 03m 02s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/76bb9273ab274bed8ea5ebc607b0e33c : SUCCESS in 51m 13s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8cfbece88a42481d8e821613571ff3bf : SUCCESS in 1h 39m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/77c31bdfe19947668f8d64f3cca163cd : SUCCESS in 1h 05m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcfd267e9e3f479da1f9082f168c9a66 : FAILURE in 4m 17s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/13e3f2b0cbbe435a9c7a2a47577fc025 : SUCCESS in 4m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/83551cdbae7842f6ae2e1853daedd69e : SUCCESS in 5m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/06605217afdc4ceeb3e4cdb73df74dcc : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c0da45742b14d56a0c8bdd0b14470ea : SUCCESS in 6m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37f5a0af1b724fbf9a8d139b46152e6f : SUCCESS in 7m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e7857c17e937451596970f5e6ac3ba82 : SUCCESS in 6m 45s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/5a8f317c39014c99a169c5fd4c2976f6 : SUCCESS in 11m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cee21ad3f7264cd1a915033e0ee338a0 : SUCCESS in 1h 47m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a813305bf755456086f0bd8dc7104cfb : SUCCESS in 1h 13m 28s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/586c31b337fd4364940d8747b0d36a07 : SUCCESS in 2h 04m 49s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/18ac2163d98b42fc803e4afa1b0b6d7b : SUCCESS in 2h 21m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dd2752468352470abd49b34156483243 : SUCCESS in 47m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d4934085267d437091c75230f6ebb992 : SUCCESS in 1h 44m 31s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9ab3c515823d4cd493d88c6d8ecb5a39 : FAILURE in 32m 01s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/7b699399696c4d63866b7dc3e724dc47 : SUCCESS in 2h 03m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f04e7a4bd25443dd877156f650aad3d7 : SUCCESS in 18m 29s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/44438b9c17ee4650a89f8d4ee0df4320 : SUCCESS in 1h 10m 44s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"8bbca0373a0384dfd12d02cadedbf67b74ee7d88","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-08 20:00:15.000000000","message":"Patch Set 16:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f398b64bd5694479af63f8921a40d32a\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/005265941d1641bda0bc4056d12ef914 : SUCCESS in 8m 18s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/1e92eb4b6bb748f9aef1a3cddd732765 : SUCCESS in 12m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/78f4f4a1c5774cde9494dc03b42402f2 : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":16},{"id":"292ee84127225e4e9d3e103f21c4121ba6df0c14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-09 07:34:44.000000000","message":"Uploaded patch set 17.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":17},{"id":"1364a982cc2d8ac25d2523bf3a0d0f24ca32de08","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 07:48:05.000000000","message":"Patch Set 17:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/452e985541a7424ea6a67b605cc23651\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7b8760eb0c24218a78a272e6bea1421 : SUCCESS in 8m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/cd9b1d8e6b9a4c98b523a5462ef05953 : SUCCESS in 6m 26s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d18982b2030b41f6954e1a7d8599765f : SUCCESS in 11m 50s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"69da104323c0e8cc224e6e416536fe2d56dd9b01","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-09-09 09:47:37.000000000","message":"Patch Set 17: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ba624a8f04f74c03a7266e3a2108d653\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/6f3612e4df0547239e7c4c79ba58094f : SUCCESS in 4m 30s\n- grenade https://zuul.opendev.org/t/openstack/build/33863d170d1d4da8acd7dd446cbabcf3 : SUCCESS in 57m 39s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9c1cf13132149cbaf3aaa1825a01d18 : SUCCESS in 1h 01m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/94110ae93d854ebcab00e860b1093656 : SUCCESS in 1h 33m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f0118cafdb0d4bbd90ef0b918986969f : SUCCESS in 44m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae5aea3f72d4a9b8affd1efba0cbe2d : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac3ee51a153542279132c47fa7f770b2 : SUCCESS in 2m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/db1402c4a85841a593697c488b8b3ba9 : SUCCESS in 3m 51s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/80a01739fd9b4a7eaf90a1b649ccce9a : SUCCESS in 8m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/378c1701690e40a8afb0c30beb05fde5 : SUCCESS in 4m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cae1d5357a3486e941346e945a68e48 : SUCCESS in 5m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e768ac24597f4c0d9784d2629c5cfd29 : SUCCESS in 7m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/34b9f1fb3537400394fb0ea561072adb : SUCCESS in 10m 11s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4eecc4850c1f417fb941cc9777496dc1 : SUCCESS in 2h 07m 45s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/87e5c253f6cc48f98bafaa38efe883ac : SUCCESS in 1h 25m 33s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5c78c18abc544bde8c4a0591854c933d : SUCCESS in 1h 03m 35s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1bfdcb1dba1d4d7fa9ee5ffce0f46575 : SUCCESS in 2h 02m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/afead531770b49a9abb8e7469bedea5e : SUCCESS in 44m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/21d5e3dbd20e420d97ecebcb0b1f4e97 : SUCCESS in 1h 22m 13s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/12a3e05e0d014ffc88d64a9069171f84 : FAILURE in 32m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4c58c62e177041d08f39eb72891c8b06 : SUCCESS in 1h 19m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8efce06278ed4d468701c16e1be77b46 : SUCCESS in 15m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/8a51679575814428bfab640074aa6366 : SUCCESS in 1h 00m 46s (non-voting)","accounts_in_message":[],"_revision_number":17},{"id":"3aa861c5011ee87de7aaf2bb76482bbcbc913213","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-09-11 14:34:14.000000000","message":"Patch Set 17:\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"547e6cef15871aff04f39dc828d621e16290b7b1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-18 14:51:31.000000000","message":"Uploaded patch set 18.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":18},{"id":"a2c704e8672985b38b09d8bf12735dabc5e4773d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 15:26:36.000000000","message":"Patch Set 18:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4ee792136fa74922ae6d3b0de0a20f17\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b2d444709ee048849891beb9010b2441 : SUCCESS in 13m 00s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/3b4cddb959ad4e27a133b310eadf744c : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/2b9fb716d5984c939890ffc60668aec4 : SUCCESS in 19m 19s (non-voting)","accounts_in_message":[],"_revision_number":18},{"id":"0df28022b86898ad6c72e03e7ccf4225ae65eb9a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-18 15:56:48.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"5249d137733ad73e22e1fbc5dad68cdbe3bdfd9b","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-18 16:10:14.000000000","message":"Patch Set 19:\n\n(4 comments)","accounts_in_message":[],"_revision_number":19},{"id":"6fb747af4826e9b15d2fc88f8331baea93661adb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 16:22:36.000000000","message":"Patch Set 19:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8585cf5ab81f492e9df79ae3f483e395\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/6ab60a05663a40859b88034c2ee3c3f4 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/6742cbafb7cd48ed8bb015a1ef80bd13 : SUCCESS in 6m 04s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/fea6235ad9844c9f8f86e06a43ac1ede : SUCCESS in 8m 39s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"7b21aa72857e911e0e1009eb7462ce78515180a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-18 18:05:41.000000000","message":"Patch Set 19: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/354530e89f6e4d33b8e8a1582cfd95c4\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d2a4ebd04e964d27aa44f78f0c81f3ac : SUCCESS in 4m 49s\n- grenade https://zuul.opendev.org/t/openstack/build/6c51d4e7efec4cbdac85f54b84d96fbd : SUCCESS in 48m 27s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/2e1b74fd53304714bdef27cdb1c835ff : SUCCESS in 1h 10m 04s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e0342f413c064d989d9720242b05da7f : SUCCESS in 1h 46m 32s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/4f84f2e54530415882c382bbdb3787c7 : SUCCESS in 53m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90056dd4c2204fd797bcb5fd2f5aa43e : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6fb6379201b34b26aba667e70e9e2ed2 : SUCCESS in 6m 11s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/29f2ce1bc5cf46728e0bcb3c87780a9f : SUCCESS in 8m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/917ae9048b5942d69d924237ac251cd7 : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e7488c586ca54907bce9023c04717167 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12f6efd8432048f8a4ee47ddcc2bdd9d : FAILURE in 6m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/53ad5b6af3654274bd2bb36e907e4087 : FAILURE in 7m 30s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7374b11041d74755835a20cc2f3ba1c0 : SUCCESS in 1h 27m 04s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/9ee7907fad014504ac2e8b2c6291c338 : SUCCESS in 1h 55m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/cf5c81cb95de4a3bbda2b44be9c8775d : SUCCESS in 1h 13m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1a2b10ce3ee540edb149730639057e53 : SUCCESS in 2h 07m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a1029da616a04557a50bc1da46b54fe3 : SUCCESS in 55m 31s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ebf15b9ba5b245d39cb172f1ac7a0db2 : SUCCESS in 1h 53m 40s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/fb7a6f11049d4581950dc9a2d326b61b : FAILURE in 15m 51s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/b00c02af0b7f453eb5481e366a4974ec : FAILURE in 12m 38s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1537b415ab534f39bcbdc79ef5df3ec5 : SUCCESS in 33m 13s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/c7fc26d7c14745adb71b524d03733237 : SUCCESS in 1h 02m 59s (non-voting)","accounts_in_message":[],"_revision_number":19},{"id":"a7b9965cfc6c8eca43abd423f2204198797d775e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-11-24 16:36:46.000000000","message":"Uploaded patch set 20.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":20},{"id":"8a21e23856ab16fbcdb8fc446311d11b556e31f5","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 17:12:05.000000000","message":"Patch Set 20:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6cdd932e4fd5471492e476cbd14f0ac2\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e2ffd89dd0f742629bf991b6be39ebb3 : SUCCESS in 6m 25s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/fd336ef3787343b98de5ab3109141770 : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/4027ffcfd9e84416a36e63fc7e9b332e : NODE_FAILURE Node(set) request 9d98aaa0d2714ebc893a79d1a669cf20 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"455ccafbc677825c62a9dc350c08783784277fd9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-11-24 18:47:14.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/803261a05de9488ebb3173b29e688dce\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d18ae717206445509628a67b4e9f1cba : SUCCESS in 4m 58s\n- grenade https://zuul.opendev.org/t/openstack/build/8d3da07b480149c985eeafcfbc26fcb8 : SUCCESS in 51m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0a3a40258c844695a07f0dc239eec50f : SUCCESS in 1h 09m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/18b021078f5f4077b02f02bce92081d2 : SUCCESS in 1h 44m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/2ecded1d3fcb4311b5b50b9ebdf7a01a : SUCCESS in 45m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/820c3a5a32bc4855af3f0e8b8910f74d : SUCCESS in 3m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/24a9e31a451a4196809ea782b005aa6d : SUCCESS in 5m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/88e2b0a3f0ef4cc49aecaf1ec841201e : SUCCESS in 9m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25252a6efbaa42799f6a201613a4bee5 : SUCCESS in 7m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/01b48afdb9374587be21898127388b7d : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3605bf7df0c54437ba583f79681465e3 : SUCCESS in 6m 55s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/92f495df2529492baf43f3326dffa129 : SUCCESS in 6m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7cf96e86b3bd4e66b8ce3d8ab37d766e : SUCCESS in 58m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c50ee021526b488aa8d8fb8279b5f7ed : SUCCESS in 1h 57m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/698425e44be24a52b03adba8b53651eb : SUCCESS in 1h 44m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/990c47c9071a4beb997a34c591e21edc : SUCCESS in 2h 04m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/dc485474f59645fba5ab5b08bb7c2401 : SUCCESS in 57m 04s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c69f75d1f7024bb5830fbb2ab756a44f : SUCCESS in 1h 44m 23s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/3f752e69d2e044c9ac4ea8901c47bf8f : FAILURE in 14m 28s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/d0b96ee84eff48bd94ecc2d8e7013a8d : FAILURE in 14m 11s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/111e51d4337d4b2ab886a22cb94e9a55 : SUCCESS in 35m 49s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/0185558f2f24473c90adedbbaa5dd353 : SUCCESS in 1h 10m 19s (non-voting)","accounts_in_message":[],"_revision_number":20},{"id":"bce7f5749d91b47489856d153640c043801ee127","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-09 15:52:19.000000000","message":"Uploaded patch set 21.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":21},{"id":"7e362741f4cf4c16d05ec7dd44d67199d9f1347f","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-09 15:53:09.000000000","message":"Patch Set 21:\n\n(1 comment)","accounts_in_message":[],"_revision_number":21},{"id":"24e6688a92ebbd79f59fd6451cbc76b30dadbe75","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 16:15:08.000000000","message":"Patch Set 21:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47147312610457c8786ea75b3a97003\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/35e24796086f4e49b1b2bd4c124e1c6d : SUCCESS in 9m 30s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7eb29890480a4baf82f4274885d46a44 : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0f5b46cee4f343fc80ed66c4d784abc5 : SUCCESS in 14m 24s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"369574d60ed743d08c2cb8a29bca2fb97d3585d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-09 18:28:32.000000000","message":"Patch Set 21: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9f59462935a6479c8e9d9dd99eaadfb2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/56741dc79b1543f1b453729e4c48fd80 : SUCCESS in 2m 59s\n- grenade https://zuul.opendev.org/t/openstack/build/0d827679bf294d11b8bc6288a4022bb0 : SUCCESS in 50m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/9e55a40b10a34a96a85181a49b6bfbd2 : SUCCESS in 56m 56s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3e4949439c434ea5bae5dc420538f2c6 : SUCCESS in 1h 49m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1197fbfe143040fe86c833fe65b49bae : SUCCESS in 1h 06m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d886955cfb5e42c5b7c591c0cfbf06a6 : FAILURE in 4m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f2a191ea27c84898985bd55b28ac1d1d : SUCCESS in 5m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/effc6d2264dd40c3aa369e3d5c047a7b : SUCCESS in 8m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/151bf0b4d96c43708dddb6b66dfe427c : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f71c9aa2f1141f9ad0a680325fa8ec4 : SUCCESS in 3m 34s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a63ab232e78a4cb38bff52ce14cc95fa : SUCCESS in 4m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/45db6f0264004d6e8741a71b144b8ad4 : SUCCESS in 9m 52s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0754818dff8b46c0ab9e8a508858302e : SUCCESS in 1h 35m 01s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/7e89c43752c745c88354c052079b74e3 : SUCCESS in 1h 38m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6314c11526fb449ebf5309b80f1add20 : SUCCESS in 1h 43m 34s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/48a12a540a05401aa975d9f44041ecb2 : SUCCESS in 2h 30m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/482607fbf64f464598f16bf1c62e30de : SUCCESS in 33m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d8fda43892eb4bd7a48eb9caf9706f98 : SUCCESS in 1h 53m 51s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/78d22d1ad8c247579c563f418e0d791e : FAILURE in 11m 59s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1beb7282dc474ca8ba6f57f2412b034e : FAILURE in 35m 53s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/140b99abd4a04176889657afa64782ba : SUCCESS in 36m 48s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/602e29e892b24be7b4f6e032ea501098 : SUCCESS in 1h 14m 55s (non-voting)","accounts_in_message":[],"_revision_number":21},{"id":"72afb6840f7d02e0fca3def27779b864c53f5ece","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-10 15:40:35.000000000","message":"Uploaded patch set 22.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":22},{"id":"f2e8e048fe4e69c774f6681c6ca602dcb71be66c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 15:58:40.000000000","message":"Patch Set 22:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68ef77a6226e464ebf8137b17c58e654\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/239fd19058f4440ab08d161d7a23811d : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7e87c7138b1e4d8699aa1b36439a80c8 : SUCCESS in 7m 33s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/c822958bb9364ede9478fdc200ffd9e6 : SUCCESS in 15m 49s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"1a3bdcef028a31bbd2f41b742cd3f44755d515a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-10 18:04:47.000000000","message":"Patch Set 22: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0645bdc2d0c349a3abc822e2f3d43d55\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ff557af50c7046f280b3a63763694eef : SUCCESS in 3m 34s\n- grenade https://zuul.opendev.org/t/openstack/build/4917d4598069436a9eea20de0fca9aab : SUCCESS in 1h 12m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0b7b83e200514eff8aa36ea165695203 : SUCCESS in 29m 17s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/645c1b6e004d4372a6e85c532013240e : SUCCESS in 1h 21m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b69d2123945d4493b129a00fcf8babe7 : SUCCESS in 1h 06m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/202726332c2043f79e73567246df594f : FAILURE in 2m 40s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cbfedd04cfc842f7b05ad8cac8b43901 : SUCCESS in 5m 15s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9e0fc05e30e74e97a09e75a7f2f2615c : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/042eaff32e6845808577449ee55b1051 : SUCCESS in 3m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3dab3ade96f54012873583a9462bbb7f : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/14e0759f19cc4897b739c1e8ee91378d : SUCCESS in 6m 14s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/01797e9f057640919558a3a6e66fa935 : SUCCESS in 8m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cda110c467ac4336ae3e81a6854cb8cd : SUCCESS in 1h 53m 25s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e074fa5a02314c3e9c68b94ca47d1036 : SUCCESS in 1h 53m 21s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/97ccd1a6149146bbac441bf19b777468 : SUCCESS in 2h 04m 41s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c5bfc00729b64edda1e672a2f0696f8c : SUCCESS in 2h 18m 02s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/be64c0838fae4db88236c003c21f7fe9 : SUCCESS in 30m 17s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/69a38e9ea95b40468195f5edd44110d7 : FAILURE in 1h 03m 57s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/82605628ac7241b0ae8774d81d0ddaa8 : FAILURE in 12m 17s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/215942416081403fb8f2021010759cd2 : FAILURE in 26m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bb6bad94a49e45b4b45be061a0875174 : SUCCESS in 24m 39s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/2722bc0b5f23439aa1d5124f16d8f54f : SUCCESS in 1h 20m 11s (non-voting)","accounts_in_message":[],"_revision_number":22},{"id":"4b98b20349421741fb0e4e3897bd2533884226e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-11 10:16:43.000000000","message":"Uploaded patch set 23.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":23},{"id":"4c228c773ea56e4d080e9b6ef6c83502695b1563","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 10:43:11.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c35bb6c927584f9bb36c8521a022a15b\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/f27637163cdc448494e5beb94ad3e757 : SUCCESS in 6m 45s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/11d03d282a2b430c818afa5564b4112d : SUCCESS in 10m 51s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/dc04b52303b6410b9fc17c43f3add252 : SUCCESS in 10m 07s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"f78365cb47e56616db52a9cf79030bb732d895a9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 12:46:55.000000000","message":"Patch Set 23: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/be748a5a27664673ba75ff2026303570\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d92f9b7c47c245f3b1f03ddee08326f8 : SUCCESS in 4m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/20357f7b05174b609e1c2521f788afc3 : SUCCESS in 1h 02m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a9cc55d7f4994abba9644e321e6e668c : SUCCESS in 57m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b92897b3571a4b19930bfe872d3e65a4 : SUCCESS in 1h 52m 35s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6cff21b166764e4cbe0b0b3407b16e6a : SUCCESS in 1h 00m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c409da938634b128e7cc9ef792b1ca8 : SUCCESS in 5m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3c190d280f94d3798b29e204a8fda22 : SUCCESS in 4m 48s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c541d1ac46db4d23b49c99167baba7f7 : SUCCESS in 8m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d73fc86a0bad405cafb981115d32ecac : SUCCESS in 3m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9f973af54f44c03979d133ff993970e : SUCCESS in 7m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52d94598ce3b41c3bf66eab834fc58f8 : SUCCESS in 6m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9cb93ad2c36e4c068a25a23873092d74 : SUCCESS in 6m 46s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8b00de3ae2e5440299b58a2604e76c50 : SUCCESS in 1h 56m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1571cb840f2e49a5b391876f8fd34526 : SUCCESS in 2h 19m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/849d3dea27d849f5aa8ab12a05a194b6 : SUCCESS in 2h 27m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/57333583368946f986f949259175497b : SUCCESS in 2h 19m 15s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53b46de54be445a499d7c1e1b90c5316 : SUCCESS in 50m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8880465ded664e0b86a34c023a50af5c : FAILURE in 1h 01m 33s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/b2968437cc684e6983ba21bd9cbb1995 : FAILURE in 21m 01s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/f57f1a550aa24109972f2d009e9276d4 : FAILURE in 12m 50s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/08abbccd3d3846899e0c52c87ad44d88 : SUCCESS in 28m 19s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/1e12b5443e7b4e648e95f4f5c52ab2ba : SUCCESS in 1h 28m 39s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"3840e7fa44cf058f0c110eae450c76d49274f42a","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-11 13:03:09.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"856bbe0788f8c078ad714b249a249cc4d68a42b6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 13:41:57.000000000","message":"Patch Set 23:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/24ec84bcc0a447158b25cbaf1df1d3cf\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/7cfec884a3e74a44ae4b83cc64b5ceab : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/2840dd0e15124f3d96e92f614aacbc61 : SUCCESS in 9m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/ed392680fd56448fa80f2cdfbc5da507 : SUCCESS in 11m 16s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"426b976339e78b12032d73df470ab406c4fadabb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-11 14:54:54.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0153a4c011d4bc2ad47df19a7c2668e\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0f41a58b2b9f4f7d9d1e07b3c41d111f : SUCCESS in 4m 24s\n- grenade https://zuul.opendev.org/t/openstack/build/185155fb7ae34a49ab54da3246b3166f : SUCCESS in 29m 51s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/061db9f15125477eb11b3c26403fcfc6 : SUCCESS in 52m 29s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4daddde801744b6b99b76468117d5e2f : SUCCESS in 1h 35m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fd26b568023f45e1873f3056f605d830 : SUCCESS in 32m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31ee4e95a75547eaa34751cbc836dd9f : SUCCESS in 8m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0568e742279e42ae8449f1b11a7f97f4 : SUCCESS in 3m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/a01f847b984b4700882e63445173e51b : SUCCESS in 7m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8aac25f01be742a3b082c7c4d69200b3 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/56ff30e7a3554ff3a5e262ddab6615bd : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a0fd0b05f09a453994aad437073c948a : SUCCESS in 7m 53s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fa2a1b62d6af49b7b3899eeb0f444e80 : SUCCESS in 9m 44s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a7baefddd46a44729cea84229d8aca9c : SUCCESS in 1h 42m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6b4361dd39334f7493d377aa4b05bf37 : SUCCESS in 1h 09m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6d1b3a9ec1384f179fede83ec1a3f9bf : SUCCESS in 1h 29m 57s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d8982319841b440c9fdb5db5c3751fda : SUCCESS in 1h 19m 23s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e721d2e31bd64b9c874aefed989c327b : SUCCESS in 58m 52s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8c15cbfa2b764657a182d2e0ffacbdc3 : SUCCESS in 1h 49m 29s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/04c217fc8c454c4c87a5e86446a9a13e : FAILURE in 16m 22s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/25e835b5ff0d4be6b6e3aea91a636219 : FAILURE in 11m 24s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8107b5f802284eb2ac609b58830c21bc : SUCCESS in 33m 21s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cb7c6c6cff8e41e296f23e8d6936bf1a : SUCCESS in 1h 13m 46s (non-voting)","accounts_in_message":[],"_revision_number":23},{"id":"a853242af5cff5cbc8879f02c44b8066adba3835","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2025-12-12 18:22:51.000000000","message":"Patch Set 23:\n\n(11 comments)","accounts_in_message":[],"_revision_number":23},{"id":"376cbaffad803e7e6803c8023ece7ac92c8368d0","author":{"_account_id":4690,"name":"melanie witt","display_name":"melwitt","email":"melwittt@gmail.com","username":"melwitt"},"date":"2025-12-12 21:35:44.000000000","message":"Patch Set 23:\n\n(1 comment)","accounts_in_message":[],"_revision_number":23},{"id":"f726ba1a8ee103efb2d348c11f7a3acff9f196ee","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 09:52:30.000000000","message":"Uploaded patch set 24.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":24},{"id":"e374f8a3f8d01cf9ce0a33c2931771490fa48e0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:10:11.000000000","message":"Patch Set 24:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b5095ed7a424f0689243196a7201a5c\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/fa45861a1ce34f75a110ee2688adf2fb : FAILURE in 9m 32s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b5639e146b6141ec9c7c530a32e3fe17 : FAILURE in 10m 14s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/cb76a34b96974d979b7f9eb6f341d36b : FAILURE in 15m 13s (non-voting)","accounts_in_message":[],"_revision_number":24},{"id":"4df52e33650da1eff32a3439baa3577fccf5b1c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 10:25:51.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"8e67f9c21ed4f53a7cfbec870f196cd0cde704a6","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 10:30:37.000000000","message":"Patch Set 24:\n\n(9 comments)","accounts_in_message":[],"_revision_number":24},{"id":"bd26a7fcbe1c3e2408236b865d12f3f78fcfd49e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 10:45:46.000000000","message":"Patch Set 25:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ac2499ee878431881d69c437a595407\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/753c26ec8b2f4f58a9b7ebc4966eb20a : SUCCESS in 10m 11s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/54f35ee6c69b45f6a9ffc5173929d492 : SUCCESS in 11m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/9cf3d4e29c2a40b9bf5f509908dfd722 : SUCCESS in 17m 01s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"5848f6be25cde173f74f738d129fca6ec32c2d40","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 12:45:16.000000000","message":"Patch Set 25: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e30d6d4925604131866b37c7f18cc881\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b2952a2521c43b9a2969a0d32983431 : SUCCESS in 4m 28s\n- grenade https://zuul.opendev.org/t/openstack/build/f7d2e28d7cbc4b3d9748c02509cd8aa9 : SUCCESS in 53m 58s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/fb9925b384b74f088b98fbecd2dee307 : FAILURE in 26m 28s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d65e4003c571412a8575b9d700394d50 : SUCCESS in 1h 11m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5caaee04b62f463aa7e5b0b3d30672c0 : SUCCESS in 56m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ba7b6f0a8f7844928ddbe1ec144dfe65 : FAILURE in 4m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c33a400f2f64266844bf5d5d66a1d8f : SUCCESS in 4m 33s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/7c1e9523a4404126aa53562948966550 : SUCCESS in 6m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94313e993b474fe5b810ea9b47eaccfb : SUCCESS in 5m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/558a3292f66f41d09beb6dfff512c035 : SUCCESS in 7m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ec1ddbc1c4214db8b545d4f7491943f1 : FAILURE in 6m 15s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/9ea22ccf2fcb4e0ebe8d46da0ff637ad : FAILURE in 9m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e80dc05331834e7291e50b31540446c7 : SUCCESS in 1h 44m 51s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/207ab5e92f33477f9956cfb996f0b9e7 : SUCCESS in 1h 01m 26s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/93e9c00ad6cc4378812eb4102f2bfccb : SUCCESS in 1h 11m 31s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/85ddf31326844bd9956a8627e620f818 : SUCCESS in 2h 13m 59s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/627e11164860422f80a1386427f34fd6 : SUCCESS in 1h 00m 47s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7d04c164402945e6abfdfa2c91a2e409 : SUCCESS in 1h 45m 38s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/56fe5d8a74194b9ea51d64e9aadb2dff : FAILURE in 22m 06s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9cf0b05a57254cc196112d19f1b3ca47 : FAILURE in 12m 44s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/920d12804dc44f4389bfb87bf894107a : SUCCESS in 32m 11s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7266db77d7864df09d2f6b3a5e4204d5 : SUCCESS in 1h 14m 29s (non-voting)","accounts_in_message":[],"_revision_number":25},{"id":"cba61215094b44f1a0cf8f77d2aecc17639c6b5b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 13:55:01.000000000","message":"Uploaded patch set 26.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":26},{"id":"ec06989cf6d3bab4b47bbf8271acbefb94c41e6e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 14:12:34.000000000","message":"Patch Set 26:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b8761b50637a4df78bce1837359364db\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/c5ece350fdee4141b587dc9b9b6b72d9 : SUCCESS in 8m 44s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/69cd98b81d524590b66cbc389b33278f : SUCCESS in 7m 12s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/bf3d6347eed84049a43b78ba8aa669e6 : SUCCESS in 8m 34s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"18418a0e198db8065b1a25bbe5f0abcbc7b00adc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:23:27.000000000","message":"Patch Set 26: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4f75a4996afa434c9ed1c5e71ab2e3f6\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/33840e4ff3e247e79086a389c9dd8cdd : SUCCESS in 4m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/0481e47f441640bebb26de06f958e438 : SUCCESS in 31m 30s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/cbc7ac7ff9b14a44b7ae4408c668b741 : SUCCESS in 30m 05s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3f562d0281f14aa48d64ebc21650223a : SUCCESS in 1h 09m 21s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8cca90bd479b46e884fb1bf5a583742d : SUCCESS in 32m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/223d04b51bd248cdbcdfa8f7e63935b2 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a3980dbf718d4d38ae7e73deaf8c1ad3 : SUCCESS in 4m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e5049302f7854f0eb837f22e28768391 : SUCCESS in 8m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28ff3be529dc4130b808681db44defb6 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9c85e31a132646f1b6e799516584b199 : SUCCESS in 7m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e4e4613e44524a99a4d0c1007d887f88 : FAILURE in 4m 42s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/47f84a3e6dfc421599a1d02c5924b0b1 : FAILURE in 6m 02s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7a0eaa66ac744a9ba5f89fd6d158691c : SUCCESS in 1h 51m 56s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/21324afe40cf4d2faf076139ead2535f : SUCCESS in 1h 08m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/989fd5e493334373a2458adf94a86f7c : SUCCESS in 1h 38m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7303b320da7a437c9f780f76a46babce : SUCCESS in 2h 23m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d57bcb3aec844ac291d1fd7fcea288b7 : SUCCESS in 59m 36s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7c276d4d9d304543a88d45bf4e8b9cd4 : SUCCESS in 2h 02m 26s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/cfa4687642e0468cb0ac64f049180115 : FAILURE in 10m 08s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/6f7cad0a7ea9407c9dba872600895b9a : FAILURE in 12m 47s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b5793acbcee848cd95ebb540625a5d89 : SUCCESS in 21m 53s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/cfd0114d711f48788eb6eaf773035ec0 : SUCCESS in 1h 06m 30s (non-voting)","accounts_in_message":[],"_revision_number":26},{"id":"dbc7ff9a6a2d80094dada49ada336c95974f70d4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2025-12-16 16:47:31.000000000","message":"Uploaded patch set 27.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":27},{"id":"edc85ff4c7183280aa5ea5d0f25576cb4d95d3e9","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 16:58:16.000000000","message":"Patch Set 27:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/772dd6d141bb4fdcb2b0b2e982b33b89\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/e7f7d437595d406bb35644f4926369d2 : SUCCESS in 7m 57s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/f4c69119d14849c6a243ef71ede26c69 : SUCCESS in 6m 47s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/da52aa6f7c474025b2215046a7c9d14c : SUCCESS in 9m 08s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"4269feab7325c5e569e993f1faf3143a1b231f81","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-12-16 19:16:21.000000000","message":"Patch Set 27: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acc15f4757b6448482859e3fbbbd8980\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/77ce985f77bc4b0ea26acf7b2934f9ba : SUCCESS in 5m 01s\n- grenade https://zuul.opendev.org/t/openstack/build/2c8d3cfc8f69444e89c41360b2a97b24 : SUCCESS in 59m 42s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8c3fadbaed91432799f97c408f373c69 : SUCCESS in 1h 02m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8801c87d4dcf449ba9c5210d290f5e07 : SUCCESS in 1h 54m 12s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5c5c321e9a3a40d1b43fc6b3de6062ae : SUCCESS in 1h 03m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4311ac87f2e486fbb66a6b34013728b : SUCCESS in 4m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/050325ae26e4417489e6cce918d672db : SUCCESS in 4m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3c505e3897f14d44855eee13faee30d8 : SUCCESS in 10m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc2fb72bc9034224a4e8b18bbb2dd71b : SUCCESS in 3m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03aa645c1ff646c397ee38f1ebdfcae1 : SUCCESS in 7m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f77a8e71c68140a894a3678b70c17db7 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/7b9cd82a8b3449a985f806d70ab8a33d : SUCCESS in 11m 48s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f671c4f618084b7596dd880f4f8c51f7 : SUCCESS in 1h 52m 10s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8c6ce0f666a6432c9ab9d0b17a76748a : SUCCESS in 1h 26m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8d7339df6d8e44ab8b79d4854536a8c0 : SUCCESS in 1h 55m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df4995e0cd774c6586ee639a2e851545 : SUCCESS in 2h 23m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aba6a88beccb4a9582bd33693082e7e7 : SUCCESS in 38m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/74b0add1c7694b8796f12c5cc9794e9e : SUCCESS in 1h 00m 26s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/57bdc88696cc40e992306eec64557e1b : FAILURE in 22m 50s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1ec45379d8a94786ba83408e50a378b0 : FAILURE in 33m 48s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/82050cfbb2a44b88b0aaf5b076a6f5fe : SUCCESS in 23m 44s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/fda2fd897d314b19bf231e092247d609 : SUCCESS in 59m 14s (non-voting)","accounts_in_message":[],"_revision_number":27},{"id":"b2499bfec32fad399713725f9ece0d4e943166cc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-03-18 13:30:41.000000000","message":"Uploaded patch set 28.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":28},{"id":"a1603ccd0040cdfc7c8ca2ad40144c06430da568","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 14:52:58.000000000","message":"Patch Set 28:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca4953d9b4b949dca1113357783117e1\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/4c813b4729f64033bfe3ae44cd1381c6 : FAILURE in 8m 08s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/22ec8c13fc1d46f79e36559113aca382 : FAILURE in 7m 46s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/eb677e2f3342453baf94a4c825e93902 : FAILURE in 11m 33s (non-voting)","accounts_in_message":[],"_revision_number":28},{"id":"3f729c2a484986e9e048c4e9cb01d6d0854499e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 16:03:15.000000000","message":"Patch Set 28: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c6d63fdb75cc4527b9f75583b0ae4135\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1db456659707483fbe22d331762c078e : SUCCESS in 4m 27s\n- grenade https://zuul.opendev.org/t/openstack/build/b8e032f0da8e4c8cabd2bae526e3bbdd : SUCCESS in 1h 09m 18s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/845b1a99035647a9bc6eff4dd58ab00c : SUCCESS in 1h 02m 25s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb0651d48b424b979d832fae65a4a5b6 : SUCCESS in 1h 19m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/f3a62cf0434d4520a5c8b3ef50cfd0e1 : SUCCESS in 1h 04m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/728b51f6b9ba47df8e70ce79a717011d : FAILURE in 5m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2e7bb54dba24e1f8a2892e2428425d3 : FAILURE in 6m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/da2d8b5013f84faca808cf3e7c51f779 : FAILURE in 3m 00s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2a10a58c8eda41fd8f774d4e298e3e7b : FAILURE in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7c9d8871e717487d8884563fe0f94415 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38c29aa23c634f86b9a58f30c003a083 : SUCCESS in 7m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/119e483149be4a32893822eaeb7b0afd : SUCCESS in 6m 05s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/7c48268ba2a442efac61636373f08931 : SUCCESS in 9m 22s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6748ba9be0cc4245ac77a16af147e050 : SUCCESS in 1h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/14ea2756ee024afab78c958996c33b97 : SUCCESS in 1h 07m 36s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e721fd4cf8834d97a3821b381fa86f8e : SUCCESS in 2h 16m 08s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/95ca91e406e74d28a526a1f9b536789f : SUCCESS in 2h 21m 06s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6a3efc9775724419988708c1ec3f1a42 : SUCCESS in 57m 12s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/990e42c16b634f99883ec2500955a8d8 : SUCCESS in 1h 54m 14s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7bf8e4429a434934be7070476bcc4d6d : FAILURE in 13m 19s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/9e044ee549e942b18f9786020062cb3b : FAILURE in 59m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/662cc750587744398a7f159825e367f1 : SUCCESS in 18m 32s","accounts_in_message":[],"_revision_number":28},{"id":"30587cdcd0aef42f305e5ff65b47be886bc82fa7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-03-18 16:25:58.000000000","message":"Uploaded patch set 29.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":29},{"id":"3784533f7e175dbb0bdcf4d3ac79089cfeaf214f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 18:56:41.000000000","message":"Patch Set 29: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/eb0bce689896496fafcdf8d1d32b5b15\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/4fede65d3644431784010bb8c1c7039f : SUCCESS in 4m 19s\n- grenade https://zuul.opendev.org/t/openstack/build/30116a4d9bea4f9db92bd6591dbe117e : SUCCESS in 52m 24s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/55485be405be45a09509821b5d0789e4 : SUCCESS in 57m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f8c6f74dc13a432db7c101569bb0f6a8 : SUCCESS in 1h 33m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c118824cb08b4f43a8c8229302b8bc53 : SUCCESS in 48m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6b0a6bf8e2524af9bd2f6dca7718057d : FAILURE in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dc53e494235a4eb8a1f9f6235734ec4d : FAILURE in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/503438b530534670bbbeadc9cae7e0f5 : FAILURE in 5m 47s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b51037f043004458b1a98776f70a5d27 : FAILURE in 7m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc783efb861245bd8b144ecdc5aabf7a : SUCCESS in 7m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c951b06f5b904fac851985e67bfc43db : SUCCESS in 6m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/57c7542c17e64013bb2d105955f6fe4a : SUCCESS in 6m 35s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/6280435732f742df9785ee5bf751be9f : SUCCESS in 6m 56s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f2eeedc2ab584864bdffc8f51d2590ea : SUCCESS in 1h 21m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/f7f961320c0b469f830f1990f15c86c7 : SUCCESS in 1h 49m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8c90fc69305e4d5a8db110eaad3bfdd7 : SUCCESS in 1h 55m 50s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/180376d9a2f74d5382ae4d1a2c2fc7e9 : SUCCESS in 2h 21m 55s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ccca78c26e47406b88e07298cd5a5e17 : SUCCESS in 1h 02m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00de188b95284d44b27015e43ede2472 : SUCCESS in 1h 48m 29s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/bb0b40552ea840c19d497ba06efd05d5 : FAILURE in 18m 11s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/67260dc7daec4ab3874032eb5c907b49 : SUCCESS in 2h 13m 39s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db6ac6e590ba42559840cb93fc93de69 : SUCCESS in 25m 22s","accounts_in_message":[],"_revision_number":29},{"id":"1d561cdc4905e95afca7ec03c931304710f6eb92","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-18 19:57:46.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/89917d3705434e08b866a0346502db54\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/1c5a7dfb6dae4a7a917f57f5ffaa5037 : FAILURE in 9m 21s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/0fe2fcc9eb334e3791d271699b8627df : FAILURE in 6m 05s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/7608a5474a3c4c7fb6ab111ec276e411 : FAILURE in 12m 19s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"95c6111113908a013140b5ec8a26748e29f40ac3","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-05-13 14:37:09.000000000","message":"Patch Set 29:\n\n(1 comment)","accounts_in_message":[],"_revision_number":29},{"id":"59b40cdbe8832c47962edffd6694567da3e6c208","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 15:09:05.000000000","message":"Patch Set 29:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/517e40d391a54da7ab4e0601f4cbb819\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/b995605b03d04500a147adce7a6993b6 : FAILURE in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/3dbcdfaf84994886b228b3c7d7d55787 : FAILURE in 6m 34s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/fc28be66c5824a428ed6df67f34f7997 : FAILURE in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":29},{"id":"47ebd1c2bc6c180b0538d55a4a41a3894f53dc9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-05-13 17:18:37.000000000","message":"Patch Set 29:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/8b621b64af784c068e76878ac3b1a7f2\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/602ff37a26d8431e977596f5f538cda5 : SUCCESS in 4m 12s\n- grenade https://zuul.opendev.org/t/openstack/build/75db6aec3cfa4d939b082d4cb9d7eca8 : SUCCESS in 31m 13s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ac141ddb199c4f5e9907fdd50b23bec3 : SUCCESS in 31m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7f098ce992144169bbb2d78784da66f1 : SUCCESS in 1h 41m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/54e2dcfb7ea64c12a2874be42ba7ca5c : SUCCESS in 1h 09m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/466335cf072b49b8b89e95a119e87523 : FAILURE in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/67018062b2bd4c9da1ad5d4c192b3625 : FAILURE in 4m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/5c866857ca3143ceb09114d2091f71f6 : FAILURE in 4m 28s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/b163efe62469404ca8f2b7ce5394073d : FAILURE in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9ddd709c27fc462284b365d63c0b2b13 : SUCCESS in 5m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e80121a93ab4fcd8b3c4da8ece35cab : SUCCESS in 9m 07s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/07846c39f6ae4e8d8053f3f1cc27d538 : SUCCESS in 5m 52s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/86330dc635e24d678c6a15bae06be806 : SUCCESS in 6m 21s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9b732a848b0e4e3895f5756f56ffc382 : SUCCESS in 1h 50m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/69de9b9fe9b94e4c8355e74206fc11ec : SUCCESS in 2h 19m 58s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d8c5be1db87a4244a4a3bb39bdc76735 : SUCCESS in 2h 16m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a2df8a0929b44c31965ebf9405231437 : FAILURE in 1h 50m 14s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e241f313802e4fffaa3516dc97953590 : SUCCESS in 1h 22m 56s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/db73211679d447409aab119fff940f99 : TIMED_OUT in 2h 34m 12s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/7e6d6a2b3d1a456c8ee8c8a9cc75a0a1 : FAILURE in 14m 16s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/2049ae64d2824517a42a4236919d49e6 : SUCCESS in 1h 53m 55s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/295126d679814ec09667ebca9dc42124 : SUCCESS in 35m 02s","accounts_in_message":[],"_revision_number":29},{"id":"8ce3e87ed3008ce974244e8d315739f19fbf3633","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2026-05-18 09:29:10.000000000","message":"Patch Set 29:\n\n(7 comments)","accounts_in_message":[],"_revision_number":29},{"id":"86696d613b0558dc41ebe731f1657b19991aea93","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-06-10 15:11:50.000000000","message":"Uploaded patch set 30: Patch Set 29 was rebased.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":30},{"id":"1f0848e280a245889664fc8f5c347ae060d48055","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 16:25:09.000000000","message":"Patch Set 30:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdbdd57fde7940bc9e7301c8af0c230e\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/2340a0fddb3542b7940564d4b3e9b59d : FAILURE in 6m 40s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/03c378f6a5b14ce4ab5346fa4cf076ff : FAILURE in 8m 41s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/3bb44c68c88d4e4eb8855ce7e1ea149d : FAILURE in 16m 34s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"38c64e22efb5cd247bdc2f8795dedc0907edc1a2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-10 17:31:20.000000000","message":"Patch Set 30: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d20275dd1d5a44e4b1a8219f0e62a5da\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/69b3f890a49f4367a8050637201cc817 : SUCCESS in 5m 04s\n- grenade https://zuul.opendev.org/t/openstack/build/8c2935386a05434fb818e97b8b41849a : SUCCESS in 1h 03m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/c2e226b2ebf7401889b04030189551b2 : SUCCESS in 1h 17m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/786a67a47eaa4a5ead70a9d5c481c6e7 : SUCCESS in 1h 48m 01s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/ab9ba8d553344528801a9b08ce3304d7 : SUCCESS in 1h 08m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4e77f8d01cd54f189c12babd1ce9dd08 : FAILURE in 5m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f2a7f3a951a14001a4083952cab493eb : FAILURE in 6m 37s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/32e410ac35b74a02983cd9aa208b06f5 : FAILURE in 4m 56s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/38c84da74d004a4ebf97a2912edbe1cd : FAILURE in 9m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5ceb3f99f544628a04daace19409319 : SUCCESS in 5m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94b7ab369cb84adda0e40d185f1386a3 : SUCCESS in 3m 25s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/65f8b0d8863b492dbc9cb665e05f32ee : SUCCESS in 6m 21s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/37cabea55b8f44feb0a532a285a65669 : SUCCESS in 5m 16s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/7927ef7c7b4b45d2a294b2f4bceae946 : SUCCESS in 1h 51m 46s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/23e313662cd84444b39edc679c1cade1 : SUCCESS in 2h 11m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6baa23e80a904eb0a0d672e61e2d3119 : SUCCESS in 1h 46m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/9a25f27c78504dbea346b54a100540fd : SUCCESS in 1h 33m 49s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/80bb495db74149288aa41b07594b2a01 : SUCCESS in 1h 09m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4bea21474e184293b0e89a0b46f08ab8 : SUCCESS in 1h 08m 45s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/dc38f15c885147ea98e9de714b0799fe : FAILURE in 21m 33s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/060983e7502e46ad9b5c51bc9c72ed64 : SUCCESS in 1h 36m 33s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/ae0c509836d24c4aaf510f6bf0877fc1 : SUCCESS in 39m 21s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/e6a574bfb2ed49d5a8be3f9826ad4f39 : SUCCESS in 24m 46s (non-voting)","accounts_in_message":[],"_revision_number":30},{"id":"3936b79340932d782ddb5b9055a21f4fea27b725","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-06-18 13:59:58.000000000","message":"Uploaded patch set 31.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":31},{"id":"c74e4cf5976fb41cd3fcf43c22913d24bc0c50f4","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-06-18 14:01:30.000000000","message":"Patch Set 30:\n\n(8 comments)","accounts_in_message":[],"_revision_number":30},{"id":"bf685cccaf6a7e8ddfcc667895fde57f9710caa1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 16:36:25.000000000","message":"Patch Set 31: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/04334f7316f7440e9a4e8bf1fff713f9\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9b4b71995bd24cf8bd5cbbc704e63cf8 : SUCCESS in 5m 36s\n- grenade https://zuul.opendev.org/t/openstack/build/bff4b67b4f674751beabf86011a56287 : SUCCESS in 1h 05m 45s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ff893eb3090e484fb7d819980234fffa : SUCCESS in 1h 07m 21s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/df3a352685cf4a76ab44d78f5bdd7928 : SUCCESS in 1h 30m 42s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3acd3a6fa23f4d6d987377fc49d8a3c7 : SUCCESS in 1h 07m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af5f8d5acb7b4a918c0b4ef0f78d6024 : SUCCESS in 4m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2feaf4490103411e8bc64f569f9c37b3 : FAILURE in 6m 35s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/c64e70870b4b480fa80f100d42762423 : FAILURE in 3m 22s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/edb506a50d9945599b4084344b549e3b : FAILURE in 8m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbaab2c211b44e2c9ce07da9af07a53b : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f9121de1bc2847bd9762fe05f6934d4d : SUCCESS in 7m 57s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/96f6617fb30a4db38727de80d52fcbe2 : SUCCESS in 9m 36s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/a689e30358c34d50a3e67d3e1a2b0227 : SUCCESS in 8m 43s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/50259d793ba94f0eb708781df220d5e3 : SUCCESS in 1h 08m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f3d6c795bc54f29a924cf7e83340144 : SUCCESS in 2h 07m 55s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7231a728311e4385b79841600ad4ab46 : SUCCESS in 1h 10m 09s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6b14ed1b12c843979e61e08b0f4d79e9 : SUCCESS in 2h 29m 12s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6c9f97dd93824102883009b9b1ec2751 : SUCCESS in 56m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0566cb34083e4b329a5bea1ff592f71f : SUCCESS in 2h 01m 19s\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/d060e19f4ea4486f95b8a2e8ab458192 : FAILURE in 19m 10s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/3aa86bb09c474145bf934480e78b6642 : SUCCESS in 2h 00m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2310dba74f5846e98001e272924e1ce0 : SUCCESS in 16m 55s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/acfb031cfd374e018d900106c3edc9c7 : SUCCESS in 51m 15s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"ecb9c98d0eaf64ca9a9f78a63530d8efb2bd132a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-18 16:36:41.000000000","message":"Patch Set 31:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/34ed5f84859b4c28bbdb63511e7be87f\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bca804c0e4bf4c42b5c01aebd9fdc8aa : FAILURE in 10m 02s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/53bf842ea3384098b2a0cc51fb84632b : FAILURE in 10m 13s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/f0872ba964784443a3fdd7152b08a099 : FAILURE in 12m 49s (non-voting)","accounts_in_message":[],"_revision_number":31},{"id":"172721b72f165ad2cddef6b6ef4d7cc347176838","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-06-25 14:48:12.000000000","message":"Patch Set 31:\n\n(1 comment)","accounts_in_message":[],"_revision_number":31},{"id":"4798090e0a99a0edaea297bb414b7bb055602e20","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-25 16:45:42.000000000","message":"Patch Set 31:\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"5dccb633bc3c65f7411de03fbba8e94d9ca642d3","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2026-06-30 15:45:28.000000000","message":"Patch Set 31:\n\n(3 comments)","accounts_in_message":[],"_revision_number":31},{"id":"c6fa6661acb9c511a70797c5f2e24de7ad54687c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-07-15 13:39:18.000000000","message":"Uploaded patch set 32.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":32},{"id":"517bc2d71008e6b5e3b8b550ba975924b934e2b5","author":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"date":"2026-07-15 13:41:19.000000000","message":"Patch Set 32:\n\n(3 comments)","accounts_in_message":[],"_revision_number":32},{"id":"870db6afba4a379a10399c4f5ffd8ce964886015","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 14:25:35.000000000","message":"Patch Set 32:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ef69e0291cb464bada442c662f524cc\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7159a20175ec4e2b84978810bec9bdf4 : SUCCESS in 8m 07s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/5d8984d5d2904c14b1e39a6c56939033 : SUCCESS in 8m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/66669633b330424fb56934cf2060171b : SUCCESS in 15m 10s (non-voting)","accounts_in_message":[],"_revision_number":32},{"id":"07f5300c6c006b576debbf2f3d9da48f659ee2f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-15 15:56:40.000000000","message":"Patch Set 32: Verified-1\n\n(24 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/2d218dc1998e44428bfe0becae5a2172\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/9bb1f903f3cb476fa56eefc4d690822b : SUCCESS in 4m 21s\n- grenade https://zuul.opendev.org/t/openstack/build/2e648f5694dd4c3ca8641a23930637a6 : SUCCESS in 1h 09m 00s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/367db6a2e7b64f57a528827a501f19b5 : SUCCESS in 1h 07m 03s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/6e6c8667f4154c2895a916005cc179f5 : SUCCESS in 1h 11m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/034797b165274e7f95c7d822485a27e4 : SUCCESS in 1h 02m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/48bc3ecdcbf94532bed7a3c5e2dcf629 : FAILURE in 4m 12s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f68db417708c43138635777e8bb88e00 : SUCCESS in 5m 02s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/bf5ed44b75eb4a0088cea8b812c4ad73 : SUCCESS in 5m 11s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3841ab71b085400ca509dc65db7798cc : SUCCESS in 8m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ae3c3d6fdfc40549e1d2b88ebe2cd1e : SUCCESS in 4m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17c55dffed8048999c9f7203f498edf3 : SUCCESS in 8m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/af7289039f2a4c969811957306a7933f : SUCCESS in 5m 56s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/4e0e40d5e2164eda8f5e4b19293eb3bc : SUCCESS in 6m 39s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ea0f884fdd4f4da39fae5151bd63547a : SUCCESS in 1h 32m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e771d89b1f9c4a508f8216f45b3b7c31 : SUCCESS in 1h 21m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e1d10cc9b7994737a4336aed111e5fa8 : SUCCESS in 2h 11m 28s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/2af6054d3a64437e861f0b6e6491720a : SUCCESS in 1h 57m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0b3b916c718a45268f270333e0a358a3 : SUCCESS in 27m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/8ab6b8d5566341f2a9577529ab17ed45 : SUCCESS in 1h 41m 56s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/1d172b7d6988490883bd8d41e45a768c : SUCCESS in 1h 41m 06s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/acc76467553c4d3997367afa575a9e96 : SUCCESS in 35m 20s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/ba87d789f7cb4ea59f52e0828cd8c0d7 : SUCCESS in 23m 06s (non-voting)","accounts_in_message":[],"_revision_number":32}],"current_revision_number":32,"current_revision":"0ba40642d65c39e9e9537b69d2d0554c81c5dad1","revisions":{"20470a0fd70e1c8bd5b221332f9e200914b9f9b4":{"kind":"REWORK","_number":1,"created":"2024-08-14 15:57:10.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/1"}}},"commit":{"parents":[{"commit":"6f4d9c434679c7440b026e1ffef38bc7c73cc5f6","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f4d9c434679c7440b026e1ffef38bc7c73cc5f6"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:57:07.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 to known image metadata\nproperties and implements handling of these attributes related to\nencrypted images, replacing the \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties formerly exclusive to\nCinder-generated images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20470a0fd70e1c8bd5b221332f9e200914b9f9b4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/20470a0fd70e1c8bd5b221332f9e200914b9f9b4"}]},"branch":"refs/heads/master"},"5baedcf02f997d9ff80aca3def5aa836d82e3377":{"kind":"REWORK","_number":2,"created":"2024-08-20 13:22:12.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/2"}}},"commit":{"parents":[{"commit":"6f4d9c434679c7440b026e1ffef38bc7c73cc5f6","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f4d9c434679c7440b026e1ffef38bc7c73cc5f6"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-20 13:21:46.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 to known image metadata\nproperties and implements handling of these attributes related to\nencrypted images, deprecating the \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5baedcf02f997d9ff80aca3def5aa836d82e3377"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5baedcf02f997d9ff80aca3def5aa836d82e3377"}]},"branch":"refs/heads/master"},"7e1761e36a8fb0ecddb8709aefae3ebe321a0b48":{"kind":"REWORK","_number":3,"created":"2024-08-21 12:44:53.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/3"}}},"commit":{"parents":[{"commit":"6f4d9c434679c7440b026e1ffef38bc7c73cc5f6","subject":"Bump Images API version to 2.17","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6f4d9c434679c7440b026e1ffef38bc7c73cc5f6"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-21 12:44:33.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 to known image metadata\nproperties and implements handling of these attributes related to\nencrypted images, deprecating the \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e1761e36a8fb0ecddb8709aefae3ebe321a0b48"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7e1761e36a8fb0ecddb8709aefae3ebe321a0b48"}]},"branch":"refs/heads/master"},"198d21eac5a6f39561d6644bd005ea1dcb69b7a8":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2024-10-28 14:20:07.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/4"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-10-28 14:20:07.000000000","tz":0},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027,\n\u0027os_encrypt_format\u0027 and \u0027os_encrypt_cipher\u0027 to known image metadata\nproperties and implements handling of these attributes related to\nencrypted images, deprecating the \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/198d21eac5a6f39561d6644bd005ea1dcb69b7a8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/198d21eac5a6f39561d6644bd005ea1dcb69b7a8"}]},"branch":"refs/heads/master","description":"Rebase"},"28704dba6ca82c37e71025a6b9664f909497c037":{"kind":"REWORK","_number":5,"created":"2024-10-28 16:31:48.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/5"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-10-28 16:29:09.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28704dba6ca82c37e71025a6b9664f909497c037"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/28704dba6ca82c37e71025a6b9664f909497c037"}]},"branch":"refs/heads/master"},"ee925fdceafcd838abe12dcdd90340926d3eb64e":{"kind":"REWORK","_number":6,"created":"2024-11-11 13:32:42.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/6"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-11 13:32:29.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee925fdceafcd838abe12dcdd90340926d3eb64e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ee925fdceafcd838abe12dcdd90340926d3eb64e"}]},"branch":"refs/heads/master"},"269a79ab5498c54f22bd0ce1b1a37d6afb27202c":{"kind":"REWORK","_number":7,"created":"2024-11-11 14:51:59.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/7"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-11 14:51:47.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/269a79ab5498c54f22bd0ce1b1a37d6afb27202c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/269a79ab5498c54f22bd0ce1b1a37d6afb27202c"}]},"branch":"refs/heads/master"},"e5bda774f67c677ce380437891c1a6088c2996fa":{"kind":"REWORK","_number":8,"created":"2024-11-11 16:27:33.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/8"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-11-11 16:27:24.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e5bda774f67c677ce380437891c1a6088c2996fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e5bda774f67c677ce380437891c1a6088c2996fa"}]},"branch":"refs/heads/master"},"c0cbb87f742ee70db1d157b284c5d85bbf6b1f3e":{"kind":"REWORK","_number":9,"created":"2024-11-21 09:04:36.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/95/926295/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/9"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2024-11-21 09:04:12.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0cbb87f742ee70db1d157b284c5d85bbf6b1f3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c0cbb87f742ee70db1d157b284c5d85bbf6b1f3e"}]},"branch":"refs/heads/master"},"e9b24e8d4c898e6b4bce5b9701a851a343773023":{"kind":"REWORK","_number":10,"created":"2025-02-26 14:58:54.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/10"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-02-26 14:58:41.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9b24e8d4c898e6b4bce5b9701a851a343773023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/e9b24e8d4c898e6b4bce5b9701a851a343773023"}]},"branch":"refs/heads/master"},"9e42aa0acf9cafb480fe2690f2a31b1561d0a907":{"kind":"REWORK","_number":11,"created":"2025-02-28 21:46:50.000000000","uploader":{"_account_id":5263,"name":"Jeremy Stanley","display_name":"fungi","email":"fungi@yuggoth.org","username":"fungi","status":"missing, presumed fed"},"ref":"refs/changes/95/926295/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/11"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Jeremy Stanley","email":"fungi@yuggoth.org","date":"2025-02-28 21:46:49.000000000","tz":0},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e42aa0acf9cafb480fe2690f2a31b1561d0a907"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9e42aa0acf9cafb480fe2690f2a31b1561d0a907"}]},"branch":"refs/heads/master"},"224527ac7f9906af443dd130047efa598e660e9f":{"kind":"REWORK","_number":12,"created":"2025-04-09 09:01:12.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/12"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-04-09 09:00:59.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/224527ac7f9906af443dd130047efa598e660e9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/224527ac7f9906af443dd130047efa598e660e9f"}]},"branch":"refs/heads/master"},"a785c567b8019c0ff3e45dd72919e02321c69729":{"kind":"REWORK","_number":13,"created":"2025-05-23 15:29:02.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/13"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-05-23 15:28:50.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a785c567b8019c0ff3e45dd72919e02321c69729"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a785c567b8019c0ff3e45dd72919e02321c69729"}]},"branch":"refs/heads/master"},"509199ee53ba9896676316be0b32662df5e07699":{"kind":"REWORK","_number":14,"created":"2025-06-16 13:35:55.000000000","uploader":{"_account_id":28271,"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","username":"josei"},"ref":"refs/changes/95/926295/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/14"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Josephine Seifert","email":"josephine.seifert@cloudandheat.com","date":"2025-06-16 13:35:32.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/509199ee53ba9896676316be0b32662df5e07699"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/509199ee53ba9896676316be0b32662df5e07699"}]},"branch":"refs/heads/master"},"759bdc64390e22e2376832123791418ad9feae63":{"kind":"REWORK","_number":15,"created":"2025-09-03 09:56:16.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/15"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-09-03 09:56:01.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/759bdc64390e22e2376832123791418ad9feae63"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/759bdc64390e22e2376832123791418ad9feae63"}]},"branch":"refs/heads/master"},"574d81e8ee06d280d0ab84aa8e219d6e89e38d9e":{"kind":"REWORK","_number":16,"created":"2025-09-08 15:51:46.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/16"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-09-08 15:50:26.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/574d81e8ee06d280d0ab84aa8e219d6e89e38d9e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/574d81e8ee06d280d0ab84aa8e219d6e89e38d9e"}]},"branch":"refs/heads/master"},"bc17da6706f92f0608928c468034e5d08cfdd4bb":{"kind":"REWORK","_number":17,"created":"2025-09-09 07:34:44.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/17"}}},"commit":{"parents":[{"commit":"7a98b9f0f6c1df88ea776c31d8c7f8a746559308","subject":"Doc: Add info about multi-attach volume types","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/7a98b9f0f6c1df88ea776c31d8c7f8a746559308"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-09-09 07:34:27.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 disk_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc17da6706f92f0608928c468034e5d08cfdd4bb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/bc17da6706f92f0608928c468034e5d08cfdd4bb"}]},"branch":"refs/heads/master"},"b787fd6a8032658f1c8f43ae5aab3969a6474db3":{"kind":"REWORK","_number":18,"created":"2025-11-18 14:51:31.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/18"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-18 14:50:23.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b787fd6a8032658f1c8f43ae5aab3969a6474db3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b787fd6a8032658f1c8f43ae5aab3969a6474db3"}]},"branch":"refs/heads/master"},"6539876c8e94242af8bb4033fde39cf284c5bfc2":{"kind":"REWORK","_number":19,"created":"2025-11-18 15:56:48.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/19"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-18 15:56:29.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6539876c8e94242af8bb4033fde39cf284c5bfc2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6539876c8e94242af8bb4033fde39cf284c5bfc2"}]},"branch":"refs/heads/master"},"b4b32c848794169359e2b393c558948dd576cf75":{"kind":"REWORK","_number":20,"created":"2025-11-24 16:36:46.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/20"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-11-24 16:36:37.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4b32c848794169359e2b393c558948dd576cf75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b4b32c848794169359e2b393c558948dd576cf75"}]},"branch":"refs/heads/master"},"9f6d081765d934fe5a23492b00e69d299c01c031":{"kind":"REWORK","_number":21,"created":"2025-12-09 15:52:19.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/21"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-09 15:52:02.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f6d081765d934fe5a23492b00e69d299c01c031"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9f6d081765d934fe5a23492b00e69d299c01c031"}]},"branch":"refs/heads/master"},"a7ab30d06ebc770e61252bc86299e29a20a3b568":{"kind":"REWORK","_number":22,"created":"2025-12-10 15:40:35.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/22"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-10 15:40:23.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7ab30d06ebc770e61252bc86299e29a20a3b568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a7ab30d06ebc770e61252bc86299e29a20a3b568"}]},"branch":"refs/heads/master"},"1fc81b8118a7431e5827587cf10dc8614ea499eb":{"kind":"REWORK","_number":23,"created":"2025-12-11 10:16:43.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/23"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-11 10:16:34.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fc81b8118a7431e5827587cf10dc8614ea499eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1fc81b8118a7431e5827587cf10dc8614ea499eb"}]},"branch":"refs/heads/master"},"537d2058b0258509dd32094fc96d1fa2e8add2ae":{"kind":"REWORK","_number":24,"created":"2025-12-16 09:52:30.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/24"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 09:52:11.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/537d2058b0258509dd32094fc96d1fa2e8add2ae"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/537d2058b0258509dd32094fc96d1fa2e8add2ae"}]},"branch":"refs/heads/master"},"59595c8cb0b61a7ffd79f70db712065d067b7cb9":{"kind":"REWORK","_number":25,"created":"2025-12-16 10:25:51.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/25"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 10:25:38.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59595c8cb0b61a7ffd79f70db712065d067b7cb9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/59595c8cb0b61a7ffd79f70db712065d067b7cb9"}]},"branch":"refs/heads/master"},"b744215767e4e4f9f6ad83b37d7bac9d8032d568":{"kind":"REWORK","_number":26,"created":"2025-12-16 13:55:01.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/26","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/26","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/26 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/26 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/26 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/26"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 13:54:52.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b744215767e4e4f9f6ad83b37d7bac9d8032d568"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/b744215767e4e4f9f6ad83b37d7bac9d8032d568"}]},"branch":"refs/heads/master"},"11e7b12fbf1c0337504815719a4c49db73096ece":{"kind":"REWORK","_number":27,"created":"2025-12-16 16:47:31.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/27","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/27","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/27 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/27 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/27 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/27"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2025-12-16 16:47:23.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11e7b12fbf1c0337504815719a4c49db73096ece"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/11e7b12fbf1c0337504815719a4c49db73096ece"}]},"branch":"refs/heads/master"},"4168f413dd50a8ff6d4ce81b78cb359fb605ae84":{"kind":"REWORK","_number":28,"created":"2026-03-18 13:30:41.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/28","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/28","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/28 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/28 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/28 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/28"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-03-18 13:30:29.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4168f413dd50a8ff6d4ce81b78cb359fb605ae84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/4168f413dd50a8ff6d4ce81b78cb359fb605ae84"}]},"branch":"refs/heads/master"},"f460329a350dddb16a5006a5c2a2618c6098d99d":{"kind":"REWORK","_number":29,"created":"2026-03-18 16:25:58.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/29","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/29","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/29 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/29 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/29 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/29"}}},"commit":{"parents":[{"commit":"f418d3f00d7c1e2bd4d1458b0c500716eb0ba653","subject":"Merge \"[docs] Add s3 upgrade scenarios\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f418d3f00d7c1e2bd4d1458b0c500716eb0ba653"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-03-18 16:25:46.000000000","tz":60},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f460329a350dddb16a5006a5c2a2618c6098d99d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/f460329a350dddb16a5006a5c2a2618c6098d99d"}]},"branch":"refs/heads/master"},"145f90f55eae7719af4ab2a80b7530427d8179fa":{"kind":"TRIVIAL_REBASE","_number":30,"created":"2026-06-10 15:11:50.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/30","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/30","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/30 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/30 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/30 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/30"}}},"commit":{"parents":[{"commit":"c7a5f6e00037725846ab30df32339fe3fb68df4b","subject":"Merge \"Add non-voting Zuul Rally job for Glance on uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7a5f6e00037725846ab30df32339fe3fb68df4b"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-06-10 15:11:39.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2024.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/145f90f55eae7719af4ab2a80b7530427d8179fa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/145f90f55eae7719af4ab2a80b7530427d8179fa"}]},"branch":"refs/heads/master"},"d94cb0ad9057e78134e12668aa0d5d59ddc703cd":{"kind":"REWORK","_number":31,"created":"2026-06-18 13:59:58.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/31","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/31","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/31 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/31 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/31 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/31"}}},"commit":{"parents":[{"commit":"c7a5f6e00037725846ab30df32339fe3fb68df4b","subject":"Merge \"Add non-voting Zuul Rally job for Glance on uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7a5f6e00037725846ab30df32339fe3fb68df4b"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-06-18 13:59:12.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2026.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94cb0ad9057e78134e12668aa0d5d59ddc703cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d94cb0ad9057e78134e12668aa0d5d59ddc703cd"}]},"branch":"refs/heads/master"},"0ba40642d65c39e9e9537b69d2d0554c81c5dad1":{"kind":"REWORK","_number":32,"created":"2026-07-15 13:39:18.000000000","uploader":{"_account_id":27665,"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","username":"mhen"},"ref":"refs/changes/95/926295/32","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/95/926295/32","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/32 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/32 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/95/926295/32 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/95/926295/32"}}},"commit":{"parents":[{"commit":"c7a5f6e00037725846ab30df32339fe3fb68df4b","subject":"Merge \"Add non-voting Zuul Rally job for Glance on uWSGI\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c7a5f6e00037725846ab30df32339fe3fb68df4b"}]}],"author":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2024-08-14 15:54:08.000000000","tz":120},"committer":{"name":"Markus Hentsch","email":"markus.hentsch@cloudandheat.com","date":"2026-07-15 13:38:59.000000000","tz":120},"subject":"Standardization of encrypted images","message":"Standardization of encrypted images\n\nAdds \u0027os_encrypt_key_id\u0027, \u0027os_encrypt_key_deletion_policy\u0027 and\n\u0027os_encrypt_format\u0027 to known image metadata properties and implements\nhandling of these attributes related to encrypted images, deprecating\nthe \u0027cinder_encryption_key_id\u0027 and\n\u0027cinder_encryption_key_deletion_policy\u0027 properties exclusive to\nCinder-generated images.\nIntroduces the \u0027luks\u0027 container_format for raw LUKS-encrypted images.\n\nRegisters and unregisters encrypted images as secret consumers in the\nKey Manager upon image upload and image deletion accordingly.\n\nImplements https://specs.openstack.org/openstack/glance-specs/specs/2026.2/approved/glance/standardized_image_encryption.html\n\nCo-Authored-By: Josephine Seifert \u003cjosephine.seifert@cloudandheat.com\u003e\nChange-Id: I3a030b1e044653a6470377f58eeed5baf779b7c3\nSigned-off-by: Markus Hentsch \u003cmarkus.hentsch@cloudandheat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ba40642d65c39e9e9537b69d2d0554c81c5dad1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0ba40642d65c39e9e9537b69d2d0554c81c5dad1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
