)]}'
{"id":"openstack%2Fglance~926529","triplet_id":"openstack%2Fglance~master~I1c0b56afe2b2f04b49524980587cf54c7903ae0f","project":"openstack/glance","branch":"master","topic":"bug/1636243","attention_set":{},"removed_from_attention_set":{"11604":{"account":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"last_update":"2026-07-29 17:33:08.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2026-07-08 06:30:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-07-29 17:33:08.000000000","reason":"Change was submitted"},"19370":{"account":{"_account_id":19370,"name":"Prithiv","email":"prithiv.mohan@intel.com","username":"prithiv"},"last_update":"2026-07-29 17:33:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1c0b56afe2b2f04b49524980587cf54c7903ae0f","subject":"Revert \"Add CPU Mode Metadata Def\"","status":"MERGED","created":"2024-08-19 12:13:24.000000000","updated":"2026-07-29 17:41:13.000000000","submitted":"2026-07-29 17:33:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":10,"unresolved_comment_count":0,"has_review_started":true,"revert_of":392088,"submission_id":"926529-bug/1636243","meta_rev_id":"d3fcfa92907c775a1b24491832b0dc4cbf790ef9","_number":926529,"virtual_id_number":926529,"owner":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-07-29 17:33:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2026-07-29 15:49:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2026-07-29 15:49:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":19370,"name":"Prithiv","email":"prithiv.mohan@intel.com","username":"prithiv"}],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":19370,"name":"Prithiv","email":"prithiv.mohan@intel.com","username":"prithiv"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-19 12:13:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"CC"},{"updated":"2024-08-19 12:13:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-08-19 12:13:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":19370,"name":"Prithiv","email":"prithiv.mohan@intel.com","username":"prithiv"},"state":"REVIEWER"},{"updated":"2024-08-19 12:13:24.000000000","updated_by":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-01 14:34:42.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"}],"messages":[{"id":"c85e9deebe567f05641262ab880fed572366c140","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-08-19 12:13:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f328a4e5527da12924e878e9f78aaee2e1292e7b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 13:08:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7990e5d7dcf04f0cb64579878fdf63bb\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/cc89f1589f454599a2b4b72450678923 : SUCCESS in 10m 45s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/853fbc764c834c5dae238e5d337398fe : SUCCESS in 10m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"05e7e3120fffbf08e09c0ece6cd1dc00b800c7c2","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-19 14:10:03.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9c952a746854817dc4fe4a8668d2a727910dffa0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-19 14:52:22.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/ab1f2c3e75bc4d03a9706691a48e5eef\n\n- grenade https://zuul.opendev.org/t/openstack/build/9c3d75ea6967474f95ed339228ac3129 : SUCCESS in 1h 03m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/dbe5d20bde894eb09127a48d779cf049 : SUCCESS in 1h 33m 00s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/0dde9eb149eb48e6829448c9c4b662b5 : SUCCESS in 1h 06m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ac6e3561b72f43dab017a740318dcfdb : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b25dcb9e45414a2eb844f271d1cb0b05 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/73cebee4bfdc46398a1c71c4405be2ef : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/7e9804424ab444c3ab23b3be3a551f1f : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fa87e934a6794c76955239c555fe1c1c : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/76af336dd3e14e92a03c3dd16cb83290 : FAILURE in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/923105b5eb5a44aa8ceb8520c0195eee : FAILURE in 8m 59s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/73bdc3beb4d3426789786b096b2a65f8 : FAILURE in 14m 00s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/aa4cf31cae0f41e4a40a14b48194c8d3 : FAILURE in 9m 28s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/61e6f0393a3f4cb9bfc93922754cb9ca : FAILURE in 1h 58m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c1efa0c69efb48b7854fc81b052b0da8 : SUCCESS in 1h 49m 20s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/525f76da982d4e50b9c0ec6eefdeb9af : SUCCESS in 1h 52m 47s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/03efea1ebb9a4170bc7caf49a3bee0cf : SUCCESS in 1h 52m 45s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fcbde33b863943348dbb2a644b208d60 : SUCCESS in 2h 32m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8eb5f26a41f544c2aa2a64f555577a73 : SUCCESS in 48m 42s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/0894199758dd42c587b1b25e79d7b371 : SUCCESS in 1h 48m 06s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9678818e272d4afea7910bd86ee04888 : SUCCESS in 1h 51m 17s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a2583cb74b0742058ce68409446bef97 : SUCCESS in 1h 12m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d3e894e942d4f458a0405048db9d0f8 : SUCCESS in 23m 19s","accounts_in_message":[],"_revision_number":1},{"id":"9dd8c8138ee0d1ff28ac1706dc19bd0bcb669327","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2024-09-30 16:53:35.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"79b7c08be8532f1741104038e622d85f03591a5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 18:18:47.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2ca5b6dc5d70428db87d901748835b10\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/156a0f2c48374293a8856c782b90ea35 : SUCCESS in 19m 47s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/4d21251ef1ad4166a73d0d040d236d64 : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"627d6bc5c9d227de7e5237938a813b0bea206944","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-30 19:24:15.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63d2af6478f246bb95c80e419705476c\n\n- grenade https://zuul.opendev.org/t/openstack/build/00bf09007030487c835144daade98650 : SUCCESS in 1h 05m 05s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/a1c15e1035734ac995e22409b9c7620c : SUCCESS in 1h 08m 39s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/bbfedc05db39452eab0de9686c3c6cfc : SUCCESS in 1h 27m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/65b53c96b4ae43ceb6740af5896f2f4f : SUCCESS in 30m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e9eb49ecbcca4e47b05f663764a898b6 : SUCCESS in 4m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c97cd837e37a4a96a8c7922684f550a6 : SUCCESS in 5m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f0c89655bf81447787de1d92b4c9371e : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a576ba7a53ec49b0ab4fb1b3cbc461f5 : SUCCESS in 4m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c414424ccbac41bba2d36b0a4c9068d7 : FAILURE in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7523b7d7fab74f2bb7cbadfe3fc71fb4 : FAILURE in 8m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/b8f5c63d22ab43f0ba4aa9e02b6980c3 : FAILURE in 14m 08s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5611850dec4a47918bc48720a368c400 : SUCCESS in 2h 03m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/c6982c5f4eb4473e90ee345a362a3f29 : SUCCESS in 1h 49m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/5441ac6ffa494954bdcc80dd71fce0e1 : SUCCESS in 1h 58m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/333e16e565494d8b95ab357c262ccf45 : SUCCESS in 2h 03m 43s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fa2269f50e674092bfb280c47e9b0cca : SUCCESS in 2h 24m 42s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/5fecdf3353fc4cb184322ea289c6ab0d : SUCCESS in 1h 08m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a1e7e42f8aad4824aa3730dd6977175e : SUCCESS in 1h 10m 01s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/6560ed3f68b24082864c6b8148f7a323 : SUCCESS in 1h 27m 27s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/346925f7d57e4bbeb37bf3e6d5295136 : SUCCESS in 1h 09m 32s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3d5337d0117d4ba3a1679c4e1d7fb6c1 : SUCCESS in 38m 30s","accounts_in_message":[],"_revision_number":2},{"id":"badce8aaa9fb4649e723b69a97a353f8451c6010","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-01 14:34:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7f42a048a6efe0c6e9be14d66b7878be892310ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 16:55:39.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cc5251de6afa4149b55446540b78db14\n\n- grenade https://zuul.opendev.org/t/openstack/build/42e3d4e441e8456cbe5e5b57058c8c26 : SUCCESS in 55m 40s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/4974ca90499f45f2ab6da2785b0c88ad : SUCCESS in 52m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/428bcc40ef0d4f9ab0cbc6f82d6d8efb : SUCCESS in 1h 17m 51s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/18dad113fa9841e899106871d287666d : SUCCESS in 22m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73898204d35d4fb0a8e07d2130996e28 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b2a396e2279b4a36b7728d338cf16484 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ef3c6f4abbcf4aabb6b4030b2fe6b985 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1788a5310c8f44e1af94de909efcb8af : SUCCESS in 5m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3abd6f99f53643388f07b86ac3e57c1b : FAILURE in 8m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ce837c51ac524b958e4473af2d1ed822 : FAILURE in 7m 21s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/30cb234321d74278806eef60bfa37439 : FAILURE in 13m 17s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ba63900b6355499693df86b33ac6dfa0 : SUCCESS in 1h 59m 11s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8b0e1891e510489caf96754a5a5da4f7 : SUCCESS in 2h 08m 47s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/4df702acd81244e1906ad95b80c79550 : SUCCESS in 1h 43m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/00a24dd9473a402ca9d80a1303fbe91b : SUCCESS in 2h 02m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/f1ffcdfe868741d1b08e6273f64e31dd : SUCCESS in 1h 25m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eafa96e27b0e495593d5980f5e01916f : SUCCESS in 53m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/411106166219436d8b81be76d758e6ea : SUCCESS in 2h 14m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/131f304be58144d9b5c58f451cca0556 : SUCCESS in 1h 44m 37s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a6abb800f3324ace8fcb0ec2bdf37357 : SUCCESS in 1h 12m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/bd86d6895a6c4adb85f962ce2a00944b : SUCCESS in 20m 27s","accounts_in_message":[],"_revision_number":2},{"id":"278e7d8a457d9bbaaaa12d8fe1f48669eca48f15","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-10-04 15:21:13.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"e40cf81b9db8f09a59d76f29e0f8e6fe71aa0206","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 17:46:53.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/46092440984549bf8317aa954a6ca737\n\n- grenade https://zuul.opendev.org/t/openstack/build/254e8e43ff8748e899f9f871dd30c6e2 : SUCCESS in 44m 28s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f53b751327d84eae9f825aa66e79a000 : SUCCESS in 55m 10s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3bec56e2e7a241fd8df0bcf1bf580792 : SUCCESS in 1h 39m 36s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b18aff5f0cdf4d5ab1f88841ad37f190 : SUCCESS in 1h 08m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc4df8691d0648a08dbdda2686df8f9d : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8451616e9084a93a52e611bc9d563e6 : SUCCESS in 3m 35s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/89267c2828164666ba93b0d82fdeb452 : SUCCESS in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66c58551398f4ea69147e7e4f975d39e : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3f0c489485d84867bff60856a4088766 : FAILURE in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84dce2648af54cc3a43daa4f10c8343d : FAILURE in 8m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/dc9b2c8cc43f4726b0af416f9b3a4a43 : FAILURE in 12m 33s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/453c98c617de4f20aa84315c7ba87194 : SUCCESS in 1h 53m 59s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/081f2eecf6a44fe9b8e1d6213b24e59c : SUCCESS in 1h 58m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d34996bd8a604d91993ff851079f18fc : SUCCESS in 1h 20m 31s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4e61e77faae44c859ad2672b275b3a5d : FAILURE in 1h 37m 58s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/6fbb9077a9a14edfa876cc3d1fd3f9df : SUCCESS in 2h 23m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/fa1c750cd92a454cb28f6e05fbdb2873 : SUCCESS in 49m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a8de42ec7b1f4468b58ba6e95b268966 : SUCCESS in 1h 46m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ce8d2362edad484f87fa16f0ecbbff9b : SUCCESS in 2h 03m 34s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/756d5683bc994ee490d0417f9c464deb : SUCCESS in 1h 03m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/795b369989cd4d34a16348ad92590afe : SUCCESS in 20m 11s","accounts_in_message":[],"_revision_number":2},{"id":"28074ed2850dedb0a87f48fb78f3f6d32a10c05b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-04 18:25:25.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8631fdecbea456c839191c3d5ad15d7\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3ec614fac8640b1a9cc62eecc626744 : SUCCESS in 7m 39s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/5d21b1bf37b14f26a45dae066f8c21bf : SUCCESS in 14m 02s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"9872af1056b7096405041bc9097d4e2e6621ea7c","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-11-28 01:10:59.000000000","message":"Patch Set 2:\n\nrecheck","accounts_in_message":[],"_revision_number":2},{"id":"da3523dc3d99bbc9b8702714c6209ed933108147","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 02:17:59.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c2e60900d88444d99754d7d54cbecba\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/28a497c118d7452b804fc110167a98b7 : SUCCESS in 18m 14s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/a5c6873339424632be52430efdcd2d33 : SUCCESS in 22m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"2b76a8547a3567675b0910bc17c06d8c53988936","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 03:31:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/461c40f08d2e40e1a93af01512cd5d32\n\n- grenade https://zuul.opendev.org/t/openstack/build/ebd632a9fae048d88c1f54e5467827dd : SUCCESS in 50m 37s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/de3d02a0d6a7487b97e1d2551a239c19 : SUCCESS in 59m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ac800b93dc8b4043b1d32e7facf652bd : SUCCESS in 1h 31m 41s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b66f698186c345438b8370a5aa01b162 : SUCCESS in 1h 06m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/594811d629834d399b6c91097e7a11d1 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4f75f70fbc9e4f88ab79a933974dde39 : SUCCESS in 4m 37s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3d3245253cdf4216b0485461a48365e2 : SUCCESS in 5m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93e419d32a1f4210849c1aa21372caab : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0277821d4a0f40c4bc31d8887631afc3 : FAILURE in 8m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/de3bc9b4626d4c47b9cf864e4c23e62a : FAILURE in 7m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/c0c9903ba4d747e7b28f1c92cdb24e84 : FAILURE in 12m 26s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/4dde843776e44f82a17d55c511c010f9 : SUCCESS in 2h 02m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/0221deea02d94b8cb68dfd339167ee38 : SUCCESS in 2h 04m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/2b0633a535784482a3b9a97b02523f42 : SUCCESS in 1h 12m 44s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc34e31a283740dabb0c8ac16fb8f1af : SUCCESS in 1h 53m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/24219e1bf6a44e7aa0a4677481a209f3 : FAILURE in 1h 38m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bf6ce1c785324cc6a4a3a07097c1f0c3 : SUCCESS in 43m 46s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/5b208a0ecacc459480ec31b0d6bd2c4a : SUCCESS in 1h 49m 24s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/4a211811e03b402a8ab1033e1ecf74f7 : SUCCESS in 1h 16m 20s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/946f12ead0d4447b8ba4c30f039ef0ed : SUCCESS in 42m 40s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8d0b87a24ea74f918f3539e867750f81 : SUCCESS in 18m 39s","accounts_in_message":[],"_revision_number":2},{"id":"3680ff51f5c795e35dddd3c9d803beb6e6cb9627","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-11-28 06:58:40.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2a93036f71f045d36f13fcdf0b385f21f1a8b2a6","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 07:46:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0e0fcb8fd0340c7a6655f64a6a97e26\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/50878ecac1fe4fcc91ac32c6742283ab : SUCCESS in 11m 52s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/7c29d6deb9624965a67b3926ecc836e8 : SUCCESS in 12m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1f7c1b0eaef2f55e475ead1f5184d064d2db3ac2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-28 09:08:14.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e01426486d3243319f7e61398dfa7e57\n\n- grenade https://zuul.opendev.org/t/openstack/build/19d820eeaecd45028cd538992c125713 : SUCCESS in 1h 00m 06s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/e11000fa62964d73a259a7ee456facf3 : SUCCESS in 40m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6695be2d1c143d09073babb7467fe46 : SUCCESS in 1h 29m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/63b98e34a3e24dad88b82253d5c5d45b : SUCCESS in 1h 05m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/732d433ea5734818b14ce0daa6ce8c20 : SUCCESS in 4m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/193f58fd615042b7af31ff1286d4ab48 : SUCCESS in 4m 28s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1194012d16494ab687d1021ee84ce8e5 : SUCCESS in 5m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8928e2e76264fe2b5aed98899ba360b : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/540762c490e547538ff094b5566b66ad : FAILURE in 8m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8fee51a4bf944a3c9acf34e9ca8a9275 : FAILURE in 8m 40s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/139448a7f172407296d80080dff5062f : FAILURE in 12m 13s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9f573e99190d4b49aa744585db3452a7 : SUCCESS in 1h 11m 53s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/786fd7184b3041ee84ebf25561918bbe : SUCCESS in 1h 28m 46s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/609853d997dd43c1a2123760de0d93ca : SUCCESS in 1h 40m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/84395d23a29048e59c3eb1be9884f2a6 : SUCCESS in 1h 53m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/40bc5c8cab7f4470bcecb1ba9965384f : SUCCESS in 2h 07m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/c107b382ccbb4c1283349a0f04ad44db : SUCCESS in 46m 08s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4478f8d5096746ba9c5c58c2e927b23f : SUCCESS in 1h 58m 03s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/531b9cb6adf34f4abb82a83b7de91911 : SUCCESS in 1h 44m 36s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/9d49aed70d524e45b438ddd96c8b4dc5 : SUCCESS in 1h 04m 26s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/475408b215aa43999a52e9e8a4343c88 : SUCCESS in 1h 37m 21s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/7677f5f732a3401fb3f8af7a203fa125 : SUCCESS in 21m 41s","accounts_in_message":[],"_revision_number":2},{"id":"db2e5376e76a1684d4e60dc595f9249c6d8a968c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2025-05-21 10:27:51.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"38ed442b4719047ba8863bb4029e37dfacadde7e","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 10:44:07.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f948da7aed234c2da7f4215dc1d5c5d0\n\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/62d192cfa78a44aea7cffa579acb800c : SUCCESS in 6m 53s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/d9e78ab11b564490b13100444f6fd7e0 : SUCCESS in 5m 59s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/d51619a4d0e24c97ba6d07432ef2c774 : SUCCESS in 13m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8dc3b1c102a26bedb95aab23275ace8efcfa484a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-21 13:13:00.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/0338cc1f173543c89d948b6c28a07be7\n\n- grenade https://zuul.opendev.org/t/openstack/build/a41ee5aeb0324609806a50f522c50407 : SUCCESS in 1h 00m 16s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/ab2d4a74539b4f4586c7effe5ba96cbf : SUCCESS in 53m 24s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2196cd373bb940f1a094f50c073dc6bf : SUCCESS in 1h 24m 23s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9236279dd51446c581f0fc883fcc5c76 : SUCCESS in 57m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02840a02f6d043e89a53aec655b99144 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/de3af32753d442ecb87fc9f05e69986b : SUCCESS in 4m 18s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/420655dae9b74da5ac24e3a87a00629b : SUCCESS in 5m 28s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e8512c013e2e46d1aed9f1acce514861 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f25c60b2606461d95b14f4aa3cdd2c3 : SUCCESS in 5m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/88fe830dd9584ecba4369549a7661bbc : FAILURE in 7m 41s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/0f7473e7f41d4f1bb3858012d4cb3ae8 : FAILURE in 6m 32s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b286a3a25c2744bc9f6519cc4b2205eb : SUCCESS in 1h 57m 39s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6f5c102b54d1431e815a78f733e0796e : SUCCESS in 1h 40m 04s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f8a36c1c180a44f19663cb150880115f : SUCCESS in 1h 56m 46s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c95abdcc113540ff8e906c1f22ab7923 : FAILURE in 1h 34m 04s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/0958486bfab14ea4ae3ddcad8f5aad3c : SUCCESS in 2h 43m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/252c3ed7e7bf4be2ade8b7c4216931dd : SUCCESS in 56m 20s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/595e1f9597cc46e3ab823fa98cfb764e : SUCCESS in 1h 55m 46s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/04fdc50996ea48f4870c8753668be63b : SUCCESS in 1h 54m 21s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/1b28965e6d1b4ec884bdcae2aba1ad05 : FAILURE in 32m 02s (non-voting)\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/60331c311bd94ef8bd74bdcbcff674a1 : SUCCESS in 1h 05m 54s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8c539926046c4e8a9ca6d4bdf8ffc616 : SUCCESS in 34m 05s\n- ironic-tempest-ipa-wholedisk-bios-agent_ipmitool-tinyipa https://zuul.opendev.org/t/openstack/build/7e3285528f914a8586c2eeff3be40e79 : SUCCESS in 58m 32s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"134280cf6a93b8a6e5f022d6c338528bd2f85449","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2025-05-21 15:20:25.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"80df498efa7a66858b331e31cb7d3056b310fc4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 19:01:23.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1 (copy condition: \"NEVER\")\n* Workflow+1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":3},{"id":"9e50bd871b8885a1b18ee16edb8ab551afd73890","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-07 19:02:31.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"38128982da4bc365426f6992f5f8a288e4aae72c","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-07 19:23:13.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"95ca5915b2eb9c2c506cb05b3b8679773f0df0a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-07 21:33:02.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/a6a65e7d7d314c638f52300839391b19\n\n- grenade https://zuul.opendev.org/t/openstack/build/0ba087336bf046cabd678203b8dddc53 : SUCCESS in 1h 13m 01s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/8a1accadb33141d2b822e61d87eb3438 : SUCCESS in 1h 09m 59s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/8dbe4338800d4fd2a2a3bdb338624197 : SUCCESS in 1h 36m 03s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/aca0e37a5452434e98be1c9be55f86c0 : SUCCESS in 54m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fe734b0f36742debce6ebd863650af0 : SUCCESS in 3m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5cc965b7a6e940858cc31ed84214d968 : SUCCESS in 5m 03s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4ee6087768c9427985dd9ef6e65f7475 : SUCCESS in 5m 08s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/928d9c6a11e6495abbbf3e7ad59da5eb : SUCCESS in 9m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/16b1bdb9d4a045e3938a2850748dbeff : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84eb825341284a8390b2d2535b66e269 : SUCCESS in 8m 41s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8415d0c82314467d9dd293800bc2a1a1 : FAILURE in 7m 20s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/456230d502e3452f83553dd10adc25a8 : FAILURE in 6m 19s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0688f5e1fd4f4bbdb097442ffcf53124 : SUCCESS in 1h 42m 22s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b2825b43e7f2455ca85519278277ca57 : SUCCESS in 1h 44m 57s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3b65a56b84174ef690ca0e7a6cc2b93b : SUCCESS in 1h 31m 25s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/561b641f325a4aff83041201f125e61e : SUCCESS in 2h 24m 17s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/3bbe8e3664e24b24afcc2e35495b7e42 : SUCCESS in 1h 07m 21s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e0bac6a8fde04ec0b66cf67d092f765d : SUCCESS in 1h 32m 03s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/ac8103f3c5da4c188bc19e0662bf48c4 : SUCCESS in 1h 03m 19s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/c62b0d338e56484290b3a6fde941d294 : SUCCESS in 32m 42s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/6b49fad0fdbd4ff199270e54cf6887eb : SUCCESS in 27m 15s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"34ffd93416da09d79b811c75fe2a07b2d9bd80f3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-08 06:08:00.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9386cf1c6a604561aaeca745ab04e4aa\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9e36bc0c29d640808691249fe40f90e9 : SUCCESS in 7m 25s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/227e0645c39f4f388b6502e0965632dd : SUCCESS in 7m 49s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/837ba3ab3884428ebacdc28b3fb4dda3 : SUCCESS in 8m 42s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"88c82c05c54ddec5268468658e9af27bbfcaa7f8","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-08 06:30:48.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"489bec669e03587f67e8a554750345c9a3a7f390","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-10 15:23:07.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":4},{"id":"f3302b69b24c511b3a9af9dc6a881024dc7a3372","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 17:54:40.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c941ab5271474adcbe3992c40ef10846\n\n- grenade https://zuul.opendev.org/t/openstack/build/b2d54e5fe51e4bb299352cc64b8fc9f8 : SUCCESS in 32m 22s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/17fb941d659d4f17bdf6722c05df3894 : SUCCESS in 1h 06m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/90c2eb60756741c2ac32f31f3ca00287 : SUCCESS in 1h 24m 07s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dbf3e9554be44cfc9f787b028adfd479 : SUCCESS in 1h 06m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad15c8375c5473c8b7330d3d434b5d5 : SUCCESS in 2m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/51287fdf931c4a30b6e8f4c1274dedd4 : SUCCESS in 4m 44s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/71ab1613f9ce47ed86fa08d24159a28a : SUCCESS in 4m 43s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/2d2271dd48cc465e826df1bf24574c3b : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0e2ddb33b8c043588d3a853ab710434b : SUCCESS in 3m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/79a4466cc714454ba9b6adf5ad648263 : SUCCESS in 6m 28s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/bcbbceca809b4e09873f37997507ca8e : FAILURE in 7m 23s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/f91395b786084e7ca178ffc0d599fe4d : FAILURE in 4m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/9ec77ece58a54ac598b328dc1b9a1d77 : SUCCESS in 2h 12m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/6fdf6477654e410e981541a25362accb : SUCCESS in 2h 05m 13s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/566520c66aad4d09ae8b66e6bc88b528 : SUCCESS in 2h 00m 30s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/4bdac37b63b84646a528a1ffd04c4f0e : SUCCESS in 2h 25m 20s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/12bda9ad5d864ac880eaf358afa193c0 : SUCCESS in 1h 16m 54s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/465c7ed5ef7a4c3599c7c498f37f3377 : SUCCESS in 1h 03m 39s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/4ec72a5fedbf4f1794c00ce93d4ec552 : SUCCESS in 1h 55m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9c0dec2f74a2458a81624dddbdd16daf : SUCCESS in 18m 06s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/b13dc7f4bb914389be57efd50af9cdd0 : SUCCESS in 25m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"cd21c102b463e7ef9a85a2c16078719066aa19de","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-10 21:05:02.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/894cafe7836940d98e37de2f3fce6914\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/9bbe84d33d814428bb5063c8d8944843 : SUCCESS in 8m 35s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/102c27ebcca94d01ad742455aea72153 : SUCCESS in 9m 40s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/218fc88583d84da8a8890152bd158887 : SUCCESS in 11m 18s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9584c0e93662de98d1e9d7978ca4a5016cd651e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2026-07-28 16:19:52.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1 (copy condition: \"NEVER\")\n","accounts_in_message":[],"_revision_number":5},{"id":"23efe4dbcc697da2d17f5376a1346a68f6460263","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 17:22:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15f485ba7c57491182f90ac71066578d\n\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/eda785d4e4e74d88bf72b8c0acc247e1 : SUCCESS in 10m 31s (non-voting)\n- openstack-tox-py313-arm64 https://zuul.opendev.org/t/openstack/build/32b6a1b29999460992cf49e70a27ade0 : SUCCESS in 9m 33s (non-voting)\n- openstack-tox-py314-arm64 https://zuul.opendev.org/t/openstack/build/6ff80b9ce9d54b558bf965980f91c3f2 : SUCCESS in 17m 02s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"811e4a71017b8677dcbf15f3828eb7f2473679bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-28 18:03:19.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f10fbe40384485cbe1abfa10cd407b8\n\n- grenade https://zuul.opendev.org/t/openstack/build/7402e4e1f33e457587ebefef8f01f847 : SUCCESS in 30m 57s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/f4a688e74afa4e1c83ab677f2d6da572 : SUCCESS in 35m 31s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/07f415fce5bf4b03832dcad54f274ff4 : SUCCESS in 52m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/725dc93e1c484796bea2dbe28735e40a : SUCCESS in 27m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fad762740be54f4c8ecbbaac01480816 : SUCCESS in 3m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d90c7459fe034d4cb50fccd748bcdb7e : SUCCESS in 2m 42s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f54579e3178346d08eabab51aaa1e624 : SUCCESS in 3m 04s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/7ae5f9f31085485a8d67c5df418f6bf7 : SUCCESS in 6m 27s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2cbc07bb239b4c819f5e967fd2bca5cb : SUCCESS in 3m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03597a474ef642ddb94e020e03b474da : SUCCESS in 3m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/53c8fc08d0db48e6b1108aa683104762 : SUCCESS in 9m 41s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/59c77ef3c4844521b8a11c00b6119331 : SUCCESS in 5m 18s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/1a54863cb7d74537bf4bfa6f7e99beae : SUCCESS in 51m 14s (non-voting)\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/00baf6272d8a4c36b01cb0ff1cb7495a : SUCCESS in 57m 44s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/efba77c0f0314a709d2322ab16dff094 : SUCCESS in 1h 01m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2ca5845a72e04cef9c6111d00e0b4e15 : SUCCESS in 31m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/ae8cbf77ca9547acb4ba9872b44a4ab9 : SUCCESS in 1h 30m 53s\n- glance-s3-multistore https://zuul.opendev.org/t/openstack/build/0d830f3c8d844826aace1332608b19b2 : SUCCESS in 1h 37m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/921b5ce272f946dfa3c564cb7d341893 : SUCCESS in 26m 01s\n- glance-rally-task-uwsgi https://zuul.opendev.org/t/openstack/build/9cef9b947298430195a5ca56ad179e33 : SUCCESS in 53m 03s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"94bb2efb0aa87664538cb2bc61b8068677f80472","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2026-07-29 15:49:36.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"a6504d3fbe1aa031445e8bbe915e6e4263cfe2e3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 15:50:13.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"7bfbb6a2eb262acaba81c512b2a35bcbbe0d016e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:33:07.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fcd02929b34f41edafe01661dde6a227\n\n- grenade https://zuul.opendev.org/t/openstack/build/06b9294862a044319e52cea6304f934d : SUCCESS in 43m 23s\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/0db2f61198194bf6ae3bc44a74ce7955 : SUCCESS in 30m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0a09ddae43aa4f6dabb98bb91ae8198b : SUCCESS in 1h 31m 45s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b2b510f5ca174426a40ca224950718d5 : SUCCESS in 57m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4b1994119c64928beb5652b909002e1 : SUCCESS in 2m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dc50e25b494743d0bd59d4d6d0aac8ae : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/25a7548973a4466299dd5f5b90059580 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/57d9849316434beface172ba78090201 : SUCCESS in 3m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/05d5d05f08a440ceb41893d8feca4293 : SUCCESS in 3m 46s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/09a0664be7b8428f9ae6690bbb217484 : SUCCESS in 6m 45s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/670cccf81aac45b49e61be86bd5ae073 : SUCCESS in 4m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/db2d405c09a24c6cb4b142b57107761e : SUCCESS in 43m 48s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6762547dd8547fe9eb6a370fa25c488 : SUCCESS in 1h 01m 11s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/d42960b8799b4af392269c3d1b63d982 : SUCCESS in 1h 28m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/07dd56f8aa1f46efaa64dc531fe15281 : SUCCESS in 16m 01s","accounts_in_message":[],"_revision_number":5},{"id":"c921f5d8c6522657455c14baef51190c9ab7652a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:33:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"57da1cd16c66505d21ec5d03550a28b046a8ba0b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-07-29 17:34:10.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/10d9501a8dfa4234b98b55efafd284c0\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1513073eef1436f987e5d55430c5157 : SUCCESS in 51s","accounts_in_message":[],"_revision_number":5},{"id":"d3fcfa92907c775a1b24491832b0dc4cbf790ef9","author":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"date":"2026-07-29 17:41:13.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ba56e30b7a19558c9ee172492f18de60fd5bf96e","revisions":{"82937b15f5d6889a62c18ce86a55688032b8085c":{"kind":"REWORK","_number":1,"created":"2024-08-19 12:13:24.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/29/926529/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/926529/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/926529/1"}}},"commit":{"parents":[{"commit":"d8a108e7b3d1396f054199e0e56798cc281c98e7","subject":"Add CPU Mode Metadata Def","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d8a108e7b3d1396f054199e0e56798cc281c98e7"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-08-19 12:13:24.000000000","tz":0},"committer":{"name":"Gerrit Code Review","email":"review@openstack.org","date":"2024-08-19 12:13:24.000000000","tz":0},"subject":"Revert \"Add CPU Mode Metadata Def\"","message":"Revert \"Add CPU Mode Metadata Def\"\n\nThis reverts commit d8a108e7b3d1396f054199e0e56798cc281c98e7.\n\nReason for revert: nova does not allow you to override the cpu mode via flavor extra specs. this was proposed upstream and rejected so this is documenting an out of tree fork of nova and should not be encoded in glance metadefs by default.\n\nChange-Id: I1c0b56afe2b2f04b49524980587cf54c7903ae0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82937b15f5d6889a62c18ce86a55688032b8085c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/82937b15f5d6889a62c18ce86a55688032b8085c"}]},"branch":"refs/heads/master"},"2b6850c44744d9c7edcdf19984f014186ac1e435":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-09-30 16:53:35.000000000","uploader":{"_account_id":11604,"name":"sean mooney","email":"smooney@redhat.com","username":"sean-k-mooney"},"ref":"refs/changes/29/926529/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/926529/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/926529/2"}}},"commit":{"parents":[{"commit":"1c4aa9d31f93df912f0c7b3d49399b946bdd24d2","subject":"Merge \"Remove workaround for eventlet \u003c0.22.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c4aa9d31f93df912f0c7b3d49399b946bdd24d2"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-08-19 12:13:24.000000000","tz":0},"committer":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-09-30 16:53:35.000000000","tz":0},"subject":"Revert \"Add CPU Mode Metadata Def\"","message":"Revert \"Add CPU Mode Metadata Def\"\n\nThis reverts commit d8a108e7b3d1396f054199e0e56798cc281c98e7.\n\nReason for revert: nova does not allow you to override the cpu mode via flavor extra specs. this was proposed upstream and rejected so this is documenting an out of tree fork of nova and should not be encoded in glance metadefs by default.\n\nChange-Id: I1c0b56afe2b2f04b49524980587cf54c7903ae0f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6850c44744d9c7edcdf19984f014186ac1e435"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2b6850c44744d9c7edcdf19984f014186ac1e435"}]},"branch":"refs/heads/master","description":"Rebase"},"a57a8924d28849797866765392730daaab8be9e8":{"kind":"REWORK","_number":3,"created":"2026-07-07 19:01:23.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/29/926529/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/926529/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/926529/3"}}},"commit":{"parents":[{"commit":"9c79d46d1f58a541807d5a987c03ce6d296cf4fc","subject":"Merge \"Remove eventlet from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c79d46d1f58a541807d5a987c03ce6d296cf4fc"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-08-19 12:13:24.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-07 18:55:51.000000000","tz":120},"subject":"Revert \"Add CPU Mode Metadata Def\"","message":"Revert \"Add CPU Mode Metadata Def\"\n\nThis reverts commit d8a108e7b3d1396f054199e0e56798cc281c98e7.\n\nReason for revert: nova does not allow you to override the cpu mode via flavor extra specs. this was proposed upstream and rejected so this is documenting an out of tree fork of nova and should not be encoded in glance metadefs by default.\n\nChange-Id: I1c0b56afe2b2f04b49524980587cf54c7903ae0f\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a57a8924d28849797866765392730daaab8be9e8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a57a8924d28849797866765392730daaab8be9e8"}]},"branch":"refs/heads/master"},"8cc556871e2b4839aa7f026e5ecc01fa07fe8d46":{"kind":"REWORK","_number":4,"created":"2026-07-10 15:23:07.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/29/926529/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/926529/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/926529/4"}}},"commit":{"parents":[{"commit":"9c79d46d1f58a541807d5a987c03ce6d296cf4fc","subject":"Merge \"Remove eventlet from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c79d46d1f58a541807d5a987c03ce6d296cf4fc"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-08-19 12:13:24.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-10 15:22:59.000000000","tz":120},"subject":"Revert \"Add CPU Mode Metadata Def\"","message":"Revert \"Add CPU Mode Metadata Def\"\n\nThis reverts commit d8a108e7b3d1396f054199e0e56798cc281c98e7.\n\nReason for revert: nova does not allow you to override the cpu mode via flavor extra specs. this was proposed upstream and rejected so this is documenting an out of tree fork of nova and should not be encoded in glance metadefs by default.\n\nChange-Id: I1c0b56afe2b2f04b49524980587cf54c7903ae0f\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cc556871e2b4839aa7f026e5ecc01fa07fe8d46"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8cc556871e2b4839aa7f026e5ecc01fa07fe8d46"}]},"branch":"refs/heads/master"},"ba56e30b7a19558c9ee172492f18de60fd5bf96e":{"kind":"REWORK","_number":5,"created":"2026-07-28 16:19:52.000000000","uploader":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"ref":"refs/changes/29/926529/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/29/926529/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/29/926529/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/29/926529/5"}}},"commit":{"parents":[{"commit":"9c79d46d1f58a541807d5a987c03ce6d296cf4fc","subject":"Merge \"Remove eventlet from requirements\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9c79d46d1f58a541807d5a987c03ce6d296cf4fc"}]}],"author":{"name":"sean mooney","email":"smooney@redhat.com","date":"2024-08-19 12:13:24.000000000","tz":0},"committer":{"name":"Cyril Roelandt","email":"cyril@redhat.com","date":"2026-07-28 16:19:45.000000000","tz":120},"subject":"Revert \"Add CPU Mode Metadata Def\"","message":"Revert \"Add CPU Mode Metadata Def\"\n\nThis reverts commit d8a108e7b3d1396f054199e0e56798cc281c98e7.\n\nReason for revert: nova does not allow you to override the cpu mode via flavor extra specs. this was proposed upstream and rejected so this is documenting an out of tree fork of nova and should not be encoded in glance metadefs by default.\n\nChange-Id: I1c0b56afe2b2f04b49524980587cf54c7903ae0f\nSigned-off-by: Cyril Roelandt \u003ccyril@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba56e30b7a19558c9ee172492f18de60fd5bf96e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ba56e30b7a19558c9ee172492f18de60fd5bf96e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
