)]}'
{"id":"openstack%2Fglance~926645","triplet_id":"openstack%2Fglance~master~Idf71062a278657149ee0ee11b302d2a0fa6c3a4d","project":"openstack/glance","branch":"master","topic":"bp/configurable-hash-algorithms","attention_set":{},"removed_from_attention_set":{"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2024-08-27 14:51:55.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Idf71062a278657149ee0ee11b302d2a0fa6c3a4d","subject":"Refresh Glance example configs for Dalmation","status":"ABANDONED","created":"2024-08-20 13:46:16.000000000","updated":"2024-08-27 14:51:55.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"e50a31d3482d24bc376e0135145a6f1d3f9cce96","_number":926645,"virtual_id_number":926645,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2024-08-20 16:10:39.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":-1,"date":"2024-08-27 11:35:48.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 14:32:04.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-20 16:10:39.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-27 11:35:48.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"8cb29ed8d83c6f63c240f73a9c29871a970fadac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-20 13:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51c200fe958514728a16e94979b7269165ca46b8","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:32:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9fff1e9695bf4e188ac3fa3147049d5d\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/6530416e404b4d20a1db97374242c1d8 : SUCCESS in 15m 08s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/35a247c26abf459c9c0ab9ee68135340 : SUCCESS in 10m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e144a8a296c3b0f2b7ea8602dc1266907bd8c49a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 16:10:39.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/5c9b554eac1644769aada64940a236d2\n\n- grenade https://zuul.opendev.org/t/openstack/build/d27f943bce1146a5870624312de29330 : SUCCESS in 1h 06m 54s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42b5a0caedb54db7946cad599f108386 : SUCCESS in 1h 30m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/cdb0611e3027401daffd419a5c14e7a7 : SUCCESS in 1h 17m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce899508852e47e9a06f1645905c1589 : SUCCESS in 4m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0f00919d43a4224800d3ccae95ca67d : SUCCESS in 6m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d27f0ae842ed441797be0299e53f3931 : SUCCESS in 5m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8b795e1fd1aa4b76b3cd93340d58c57f : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4c56f9d713b416495d981f9e8bfdfb4 : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b6a5bfa596ab49489298c7a287e0c2bc : SUCCESS in 10m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2712d11deb4c4dab90ac21ac039ba91b : SUCCESS in 8m 32s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1ffce46827284b98acc9479ba0ba4c65 : SUCCESS in 12m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/43c7ac44e2c240a983d7127fb7bb8c8c : SUCCESS in 9m 20s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/d4ac16f595a7426dbe8262a9953cdd8d : FAILURE in 1h 43m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ee367d1ac5a424fa8b460058ada66d9 : FAILURE in 1h 45m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/540f069acf674419b261404fc7791200 : SUCCESS in 1h 55m 38s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/388c955fde314be38cb268026872fa58 : SUCCESS in 1h 59m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fbd3b34faa784e46ac11d958c6edd22b : SUCCESS in 2h 11m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ebc31a5fd0de479d87c3f28f2d816cdf : SUCCESS in 1h 05m 40s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/191ad8291a0346faa835c410b1b53e14 : SUCCESS in 2h 16m 16s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/ff9f6a8e0edf49cfb0323ad3b1cb8d60 : SUCCESS in 1h 47m 24s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/caa7e3414a6f4ccaaef2a47270dbd8b3 : SUCCESS in 1h 08m 01s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/765e3075f7ad48e7a9cd752d53db428c : SUCCESS in 32m 27s","accounts_in_message":[],"_revision_number":1},{"id":"ae32b742a6664178678614519bbf04a3774f7c52","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-08-27 11:35:48.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e50a31d3482d24bc376e0135145a6f1d3f9cce96","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-27 14:51:55.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9606641602c314013798c8b57d2686a85e861272","revisions":{"9606641602c314013798c8b57d2686a85e861272":{"kind":"REWORK","_number":1,"created":"2024-08-20 13:46:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/45/926645/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/45/926645/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/926645/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/926645/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/45/926645/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/45/926645/1"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 13:19:44.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 13:21:28.000000000","tz":60},"subject":"Refresh Glance example configs for Dalmation","message":"Refresh Glance example configs for Dalmation\n\nWe really shouldn\u0027t be storing these statically in our docs, but we will\nattempt to address that separately. For now, we simply update this. This\nis achieved by:\n\n  tox -e genconfig\n  mv etc/glance-api.conf.sample etc/glance-api.conf\n  mv etc/glance-cache.conf.sample etc/glance-cache.conf\n  mv etc/glance-manage.conf.sample etc/glance-manage.conf\n  mv etc/glance-manage.conf.sample etc/glance-manage.conf\n\nChange-Id: Idf71062a278657149ee0ee11b302d2a0fa6c3a4d\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9606641602c314013798c8b57d2686a85e861272"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9606641602c314013798c8b57d2686a85e861272"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
