)]}'
{"id":"openstack%2Fglance~926646","triplet_id":"openstack%2Fglance~master~I367dfee00f4518141a495bf39e9f33168b7fddda","project":"openstack/glance","branch":"master","topic":"bp/configurable-hash-algorithms","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"},"8122":{"account":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"last_update":"2026-05-21 13:53:13.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"I367dfee00f4518141a495bf39e9f33168b7fddda","subject":"tests: Simplify version test","status":"ABANDONED","created":"2024-08-20 13:46:16.000000000","updated":"2026-05-21 13:53:13.000000000","total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a35b11e4df4c89694664a47d84ae89a0e2d43aa8","_number":926646,"virtual_id_number":926646,"owner":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-01 10:20:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"permitted_voting_range":{"min":-1,"max":2},"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-20 14:35:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-20 14:48:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-03 14:00:09.000000000","updated_by":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"reviewer":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"state":"REVIEWER"},{"updated":"2024-09-09 09:16:55.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2024-09-09 12:33:23.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"CC"},{"updated":"2024-09-25 06:26:31.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"}],"messages":[{"id":"3a0dfc04c52e34b4d3218bb8f19384e4a4657b6c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-20 13:46:16.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"133acb2fdb5b7123bc1aeb10a3fe73ae33346246","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:35:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fa3b0f1f40f548ef8c9fb2169558aa8a\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/80e8ddf0308e40b2a58b95c88cf8f5ae : SUCCESS in 13m 14s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/af5e6db4c5d443099e85352fce33c739 : SUCCESS in 9m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d942a85f7dee54f7bd63466aee3b291935ad5cb1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-20 14:48:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/cda5931da33f49b7b8f698bba3bef0f6\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/202a0790bd874727bdbc1be0b053ccc8 : SUCCESS in 59m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d9b182aa003c45baafbd71fb58cf5145 : SUCCESS in 4m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/63e4b298f5e04a2aac198d9305253352 : SUCCESS in 5m 10s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8eafe9b8d99e4a4a9ab78b5c26b1f811 : SUCCESS in 4m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f250f9fde7ba4636a1c0240545e8334f : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5046a8f13db643568c57af1e4ced0b27 : SUCCESS in 5m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c50c53cf8aad4b689e3060e129843fa0 : FAILURE in 11m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a7e4cc2b49fa4c048ca299ab5e6d7b7e : FAILURE in 9m 39s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1d36ae7e74c64cddbe7da3f0e8db997c : FAILURE in 11m 11s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/b37f2af650574df1a72a8355b2022cc0 : FAILURE in 9m 22s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5e3057c724b445e3b93f25553562def0 : SUCCESS in 21m 06s","accounts_in_message":[],"_revision_number":1},{"id":"a87ab504b7962ceb4204a637e12115eb58566fdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-08-27 14:51:31.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"7728166f7f1ca392933b6a26a648de9fc142ffe0","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 15:31:39.000000000","message":"Patch Set 2:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3af51a5da1e4f8ca5602d58f7d98206\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/7fcd582c80514a50805983fd384a1552 : SUCCESS in 14m 57s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/f9f3b9fead574061a6c9c6960e5145d0 : SUCCESS in 12m 59s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8676172c4e03b12396f32e9439f313e880926614","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-27 16:35:01.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b597aa66070e4690b1fe221a0a6ef46c\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/14797f2182b04a19b5a864b7a3678339 : SUCCESS in 57m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6ab2ce4a3362498081e5b5053f85b03a : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52ba4fd1c47448299320392f08c70762 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddd7876cfb9c4f5a9426238bf1178f54 : SUCCESS in 5m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4a77c1276c25444e95837bab1b29874c : SUCCESS in 6m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/886f0f7abe374f2fa98e5dc0fc55a148 : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/665f9257bb1b41ebb60b7c725be1d9c8 : FAILURE in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b13e5eedc28443129f90b89ffd3dcdb1 : FAILURE in 9m 00s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4737991eb3ec4cbf9fb8d8edac5647c1 : FAILURE in 12m 55s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/1e245bb9b33b418fbaa93816e497b1ee : FAILURE in 9m 29s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/80ce46411c8e45dab54f453c9f2688be : SUCCESS in 32m 23s","accounts_in_message":[],"_revision_number":2},{"id":"2ff382aec2fbded979de1c7414ec2c91577402ab","author":{"_account_id":8122,"name":"Cyril Roelandt","email":"cyril@redhat.com","username":"cyril.roelandt.enovance"},"date":"2024-09-03 14:00:09.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dedb486521edb1e8da0d9aa2f809294552ea21ed","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-09 09:16:55.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"24bb1d78606c0634429a3ab45ed9ce26ad607a6e","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-09-09 12:33:23.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e1535700936b45f9c3b26ce04637ae61dc0f7288","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 11:19:21.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"4af5fd9e55e680e5b6d12fcb5fecce276f8e9578","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-09-17 11:21:53.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"bf131d1b0e2fc3b538ac56e8b31f0fe47bb8404e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 11:58:13.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf731c5ca4e44604b3b3c04933fa0608\n\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/40da381fbafb43459491b912a3434628 : SUCCESS in 30m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03deb249cede4136933db07ba51fcf2c : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/940d3bf8c72d431588c432784d2ffee7 : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d3648bf004f94bee86eb93b6eeb36934 : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a324cb64f4414e39b4872d705547a64e : SUCCESS in 5m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9864b2b3ad024045ae09e93c2d347a0d : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ea75475c5ef649aa9ae3dda5485c0302 : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e467c6fee7ba448eb90c19ae5be043cf : SUCCESS in 7m 26s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/14af16687c474a08bf0271b5126281d9 : SUCCESS in 11m 28s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/e6df32190881421da4e04c02759cf2f9 : SUCCESS in 19m 45s","accounts_in_message":[],"_revision_number":3},{"id":"4319fd6913521ccbc50fd30219fffc25dadfc53d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-17 12:19:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7cf4fe1765bd45968dd72dcbf6adacae\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ef188892e4dc41309aa1f5b56740c715 : SUCCESS in 10m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/73a1c87c958c4116a208df797096ae21 : SUCCESS in 8m 30s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"8d3eabac87942949709d3030894bd6bc40f09c35","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-09-25 06:26:31.000000000","message":"Patch Set 3: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4add80cb8d976bedb31086f528a92374c414b7d5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2024-10-01 09:27:15.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"fe910bfcd32ac079bc4fd7786ab5ceb16de0884b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 09:45:30.000000000","message":"Patch Set 4:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b9eef24ea13e455da18ebebf69b1b780\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/51eaaf2de52646e09f886b7ae1936396 : SUCCESS in 14m 17s (non-voting)\n- openstack-tox-py312-arm64 https://zuul.opendev.org/t/openstack/build/b99f22be34314f45b11f0373f76c23d4 : SUCCESS in 15m 21s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"05e2aac465e6adb3f42fe586da2c3064ba95ba19","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-01 10:20:21.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0cfba7918cda42878526050ed6bd01ff\n\n- grenade-skip-level-always https://zuul.opendev.org/t/openstack/build/1b60419e492d475583f8c11738830a98 : SUCCESS in 52m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9eb8783cc02c4c39b022621f7995ece8 : SUCCESS in 26m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea67bb5077a14f19acfb322ae82ec72e : SUCCESS in 5m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71e15bc35b8b42ceb382c83a74b60266 : SUCCESS in 6m 14s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3f8fb5b17130401984a7177b36a8a6c1 : SUCCESS in 6m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2d572da9f1b42efba74fbd38ba012ed : SUCCESS in 6m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/37d88f9be7b04f568bc7667066af832a : SUCCESS in 9m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0ea2423373c74df69ca73bf9f536e9f5 : SUCCESS in 8m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/f5722cee5e0c4c43963dd61fcff4d36e : SUCCESS in 14m 20s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/185ea3b41bf047bcbf076bf9bfa3a1d5 : SUCCESS in 24m 12s","accounts_in_message":[],"_revision_number":4},{"id":"a35b11e4df4c89694664a47d84ae89a0e2d43aa8","tag":"autogenerated:gerrit:abandon","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2026-05-21 13:53:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5d4973717f50184c1e7d8d694b09e86186f3f844","revisions":{"35477aff73398cd5fd742c3d1e32f30f6d59f59e":{"kind":"REWORK","_number":1,"created":"2024-08-20 13:46:16.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/926646/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/926646/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/926646/1"}}},"commit":{"parents":[{"commit":"9606641602c314013798c8b57d2686a85e861272","subject":"Refresh Glance example configs for Dalmation","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/9606641602c314013798c8b57d2686a85e861272"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 12:10:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 13:21:28.000000000","tz":60},"subject":"tests: Simplify version test","message":"tests: Simplify version test\n\nThis test has grown organically and is now rather difficult to follow.\nSimplify it so that the API version list is built linearly.\n\nChange-Id: I367dfee00f4518141a495bf39e9f33168b7fddda\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35477aff73398cd5fd742c3d1e32f30f6d59f59e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/35477aff73398cd5fd742c3d1e32f30f6d59f59e"}]},"branch":"refs/heads/master"},"ea5ae117f11f5960b694b411825cef5238455126":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-08-27 14:51:31.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/926646/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/926646/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/926646/2"}}},"commit":{"parents":[{"commit":"acb8ed5b3f5125b32569cc0a7a18c56da368c1fc","subject":"Refresh Glance example configs for dalmatian m3","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/acb8ed5b3f5125b32569cc0a7a18c56da368c1fc"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 12:10:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-27 14:51:25.000000000","tz":60},"subject":"tests: Simplify version test","message":"tests: Simplify version test\n\nThis test has grown organically and is now rather difficult to follow.\nSimplify it so that the API version list is built linearly.\n\nChange-Id: I367dfee00f4518141a495bf39e9f33168b7fddda\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea5ae117f11f5960b694b411825cef5238455126"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/ea5ae117f11f5960b694b411825cef5238455126"}]},"branch":"refs/heads/master"},"c3af590baa9ded93dacff7f8934e6c8f1b85bb68":{"kind":"REWORK","_number":3,"created":"2024-09-17 11:21:53.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/926646/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/926646/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/926646/3"}}},"commit":{"parents":[{"commit":"66cc74836af94f974d38dc4957d4a6dfd7f3b982","subject":"Merge \"Use format_inspector from oslo\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/66cc74836af94f974d38dc4957d4a6dfd7f3b982"}]}],"author":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-08-20 12:10:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"sfinucan@redhat.com","date":"2024-09-17 11:18:38.000000000","tz":60},"subject":"tests: Simplify version test","message":"tests: Simplify version test\n\nThis test has grown organically and is now rather difficult to follow.\nSimplify it so that the API version list is built linearly.\n\nChange-Id: I367dfee00f4518141a495bf39e9f33168b7fddda\nSigned-off-by: Stephen Finucane \u003csfinucan@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3af590baa9ded93dacff7f8934e6c8f1b85bb68"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c3af590baa9ded93dacff7f8934e6c8f1b85bb68"}]},"branch":"refs/heads/master"},"5d4973717f50184c1e7d8d694b09e86186f3f844":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2024-10-01 09:27:15.000000000","uploader":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"ref":"refs/changes/46/926646/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/46/926646/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/46/926646/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/46/926646/4"}}},"commit":{"parents":[{"commit":"1c4aa9d31f93df912f0c7b3d49399b946bdd24d2","subject":"Merge \"Remove workaround for eventlet \u003c0.22.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1c4aa9d31f93df912f0c7b3d49399b946bdd24d2"}]}],"author":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-08-20 12:10:27.000000000","tz":60},"committer":{"name":"Stephen Finucane","email":"stephenfin@redhat.com","date":"2024-10-01 08:47:07.000000000","tz":60},"subject":"tests: Simplify version test","message":"tests: Simplify version test\n\nThis test has grown organically and is now rather difficult to follow.\nSimplify it so that the API version list is built linearly.\n\nChange-Id: I367dfee00f4518141a495bf39e9f33168b7fddda\nSigned-off-by: Stephen Finucane \u003cstephenfin@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d4973717f50184c1e7d8d694b09e86186f3f844"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5d4973717f50184c1e7d8d694b09e86186f3f844"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
