)]}'
{"id":"openstack%2Fglance~927459","triplet_id":"openstack%2Fglance~stable%2F2024.1~Id706480e31687d8ade6f7199b600aff3ad7c68f7","project":"openstack/glance","branch":"stable/2024.1","topic":"format-inspector-stable/2024.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-29 21:33:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id706480e31687d8ade6f7199b600aff3ad7c68f7","subject":"Add iso file format inspector","status":"MERGED","created":"2024-08-29 04:40:30.000000000","updated":"2024-08-29 21:34:47.000000000","submitted":"2024-08-29 21:33:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927459-format-inspector-stable/2024.1","cherry_pick_of_change":923676,"cherry_pick_of_patch_set":11,"meta_rev_id":"9952a3784445d302e47856e0f7d6dc60e2a55305","_number":927459,"virtual_id_number":927459,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-29 21:33:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-29 18:47:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-29 18:47:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-29 04:40:30.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-29 04:51:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-29 06:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-29 18:47:01.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"fd5d2dcf3f218e66a1c02f00ae18adb91258672b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-08-29 04:40:30.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7109e6718e980a3ea808b17f5f15620f0b85a3c4","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 04:51:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28ef6d424995431f84a4b7a147f16119\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/f69d37d1fc7442bbb932906fed008bcd : SUCCESS in 8m 09s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/81ee0df0d05d4f2bbce0a35850abd9ea : SUCCESS in 9m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"feee369ec130a1c17e3c62802e79023d1ab3f398","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 06:50:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db7a8e4839c347a88c6097b96bb1a065\n\n- grenade https://zuul.opendev.org/t/openstack/build/c8a825ee0ddf43d0a64b07de9ba57001 : SUCCESS in 52m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/e9259b9c268046b98c1841e29557c390 : SUCCESS in 53m 08s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/f6b5013f88f04580ad440afde743c21a : SUCCESS in 1h 15m 24s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1611536163fe466380ac294a300352ba : SUCCESS in 54m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0462afdcdf424bed8287edc9bf60b314 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/65ef71a629d541399581546465fa4b60 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9347dac80c054b9c95267f84da59eb23 : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/09fd329ec7a9446fb462b7033b77a4c4 : SUCCESS in 4m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a360d7a423224e6188d4812489fd2a62 : SUCCESS in 10m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f6713123fc5846cca0a7c398c5f404e5 : SUCCESS in 10m 17s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/dbcf8b059beb443d88543b537297ba7f : SUCCESS in 8m 03s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/2c9a07c32ff746ae806fd79495934968 : SUCCESS in 1h 50m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/89d13fd261504042958066ef54529c81 : SUCCESS in 1h 54m 22s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bfc4dbf2221344a5b84f587000650db1 : SUCCESS in 1h 49m 16s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/e9b2937d6d2f47d2bb2a7aa13dad16d6 : SUCCESS in 1h 44m 10s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/52efa81345324a49bdb76beb1e4fbfac : SUCCESS in 1h 55m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6afb62ec2f74e51a9eaf1464a7eecab : SUCCESS in 1h 03m 06s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bff106d2212e4c4280fcf6753d7c7681 : SUCCESS in 2h 08m 10s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/875fe9c6de544deb89586807a8b79d7f : SUCCESS in 1h 47m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/5cf02d057ffb440483803cb705e64255 : SUCCESS in 1h 09m 20s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/351136ced8904fd394ddd0eed7040cde : SUCCESS in 39m 13s","accounts_in_message":[],"_revision_number":1},{"id":"c902d4cfa01d8fae779d602c099c881a08ce01f0","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-29 18:47:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bfdf6df5c3f17ce1f5353d368d88e1ff6342fce9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 18:47:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f7b7c9bc478107b70f50152062da51044bde66bc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 21:33:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2783d4578241412283b4f2ea7d83cfb6\n\n- grenade https://zuul.opendev.org/t/openstack/build/f814393846404e9783e0ba4317c553c9 : SUCCESS in 58m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/80fc0bed6b274b60b32924f827d415f4 : SUCCESS in 1h 02m 26s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ceab1e7172ab4e3c9904940fcdb833e1 : SUCCESS in 1h 26m 16s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/3d8a1ef260864fcdad8d984a4a39a14e : SUCCESS in 1h 13m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/df208424690a44bb8cd6a15542729c97 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f0b308e3183d406b8a8b7c97056d89f3 : SUCCESS in 5m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffcb26e1dc6f45199484c217dacfffdb : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6f5f2e024df41d5b6a32dbdca9a3aba : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f661a57881724b7e9653feaa4c9f98e9 : SUCCESS in 9m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36b03cdc24c54a4c9b615a1c9e0d843f : SUCCESS in 8m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/faeb5488978640e093cf352723c57b0d : SUCCESS in 1h 54m 54s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ba4f8218aacb4671a8a17ed1c8df8ae6 : SUCCESS in 1h 15m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c46b1303ba954386bd967830ba150165 : SUCCESS in 1h 13m 32s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/bd338a085639421eb06efa600b2a928b : SUCCESS in 1h 09m 53s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7a3435091c494c1e9bf6f390f8ab9092 : SUCCESS in 1h 44m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8672832ff5674c8d98fd23e92b998082 : SUCCESS in 41m 48s","accounts_in_message":[],"_revision_number":1},{"id":"2b2c8fbd71a489a20b99f73b7fe4a8d9971dc3b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 21:33:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9952a3784445d302e47856e0f7d6dc60e2a55305","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-29 21:34:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae3306c4008b4accb88f4cce42d2a6a2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31380eaa32c04bdab2629dfe23c16a96 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff","revisions":{"eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff":{"kind":"REWORK","_number":1,"created":"2024-08-29 04:40:30.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/59/927459/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/927459/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/927459/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/927459/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/927459/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/927459/1"}}},"commit":{"parents":[{"commit":"a775ab9384ee2fa2c5614b501dbf96df75c16656","subject":"Merge \"Fix LP#2054575\" into stable/2024.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/a775ab9384ee2fa2c5614b501dbf96df75c16656"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-08-29 04:40:30.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n(cherry picked from commit d8de63a5008b811f2e22cc5cdda59bd2cc05b207)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
