)]}'
{"id":"openstack%2Fglance~927568","triplet_id":"openstack%2Fglance~stable%2F2023.2~Id706480e31687d8ade6f7199b600aff3ad7c68f7","project":"openstack/glance","branch":"stable/2023.2","topic":"format-inspector-stable/2024.1-stable/2023.2","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-30 10:25:00.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id706480e31687d8ade6f7199b600aff3ad7c68f7","subject":"Add iso file format inspector","status":"MERGED","created":"2024-08-30 06:00:02.000000000","updated":"2024-08-30 10:26:05.000000000","submitted":"2024-08-30 10:25:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927568-format-inspector-stable/2024.1-stable/2023.2","cherry_pick_of_change":927459,"cherry_pick_of_patch_set":1,"meta_rev_id":"56d46065a04a0a691b7b942c4ff585f8690f2227","_number":927568,"virtual_id_number":927568,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 10:24:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-30 08:22:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-30 08:22:37.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-30 06:00:02.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-30 06:12:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-30 08:07:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 08:22:37.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"e7c11efb807889defd6d09fe679ed28ea07df799","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-08-30 06:00:02.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"0f585f9c1836780cda28d8f6c136736b56487dcb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 06:12:47.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/623128dfb27e4babb219fd358c557efc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/5125b1500f6f42619fde0758fb2858d2 : SUCCESS in 9m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/e453c4ae95ab45f09ee6d69a65064cf9 : SUCCESS in 9m 22s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/03732c4b8c33472e843299f96e7073bf : SUCCESS in 9m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/cd5448dfef6749c6b6ac8da46d1e933d : SUCCESS in 10m 35s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c228270d05581a71ef2328a4f89afca13f6c26ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 08:07:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c471ca409ab4a61828da1f758a88628\n\n- grenade https://zuul.opendev.org/t/openstack/build/fef192c4e678401d97c6a7046e629e31 : SUCCESS in 52m 36s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7fb7ff8752594d908dfefa0940f3c460 : SUCCESS in 1h 21m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/9ff587f269ec4c6a88a1ad5106aa484e : SUCCESS in 57m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a6fb4a768554f44bbc20420fb18da82 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2bbafbaac0d248e8bb74d0b5c2522a5b : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998729f5fca146868ac97dd69a3a88f2 : SUCCESS in 5m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/91b7634349b049318cd01a62f4668036 : SUCCESS in 5m 06s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0279fa8c7b6f45a38049dcb8cd31aca0 : SUCCESS in 6m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e37da9f0e55f4fff89d9d0f8c0f73e6b : SUCCESS in 5m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22b6fad9d19f4249ace726a497be48bd : SUCCESS in 8m 39s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/efdcef06b23045a59f357048cc1e10cf : SUCCESS in 8m 26s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/cfa9894e857b4030b764ca2951417b55 : SUCCESS in 8m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6c6d2fe67522420eb5c2eff8c80a1214 : TIMED_OUT in 2h 03m 21s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/13f5d51d008740e28ce7bbe7c73f1f52 : SUCCESS in 1h 36m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/0a886d0903f541aa94b8a4892904302a : SUCCESS in 1h 43m 28s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bbdd243a345d4ab0a2b59cc173698a69 : SUCCESS in 1h 24m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/53ed0119286f4630af07c47976a3a748 : SUCCESS in 1h 34m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/476f8ea99eae4732be12d94877cddc26 : SUCCESS in 58m 55s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/dddbcf0779c34765b3c624c44ae798b9 : SUCCESS in 2h 00m 42s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/124763830e844f9fa2a9e62938d52034 : SUCCESS in 21m 52s","accounts_in_message":[],"_revision_number":1},{"id":"a41c7f6914a7978e73ce62be29f66189ed334d0d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-30 08:22:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"230fcc01b35f9b552f37d822ce71ca100df0fd21","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 08:23:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"32a5186f172fe60bd8d456c8c94d1ac409b05812","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:24:59.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d7f30cc9d3e3415aa806ca55e6c97d87\n\n- grenade https://zuul.opendev.org/t/openstack/build/e1f505aabd8a4e298f08d401fd228096 : SUCCESS in 51m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/9f6866722d7f4ed488c9245b5aadf9da : SUCCESS in 1h 24m 46s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6313b64ba62d4fdca37eb93e15df8de2 : SUCCESS in 1h 00m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a19c03bc1ca94ed198677e0f4f0cf40f : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/056293d281f7435685cc4692dbed5b85 : SUCCESS in 5m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fae526f71a994315bc669fee706a7b00 : SUCCESS in 5m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8d57ba2ae394996acabca9f83679f24 : SUCCESS in 5m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/feed15ab0eff4334b55167f6b425213f : SUCCESS in 4m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9b75c258662841f89787ccdf304de304 : SUCCESS in 7m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c8c85f008f354d258516f850258c1bb8 : SUCCESS in 7m 22s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/79b133df07a74f628d2a3c6d3dd23aeb : SUCCESS in 1h 56m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b29698c11e04431d916b5fb5be08cbe9 : SUCCESS in 1h 42m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d49f2995179044e3824a7701a61a55f5 : SUCCESS in 1h 40m 01s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/6e8121f54a124f539d394a9b04f69ddf : SUCCESS in 52m 29s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/bb380648b27f4440b883208162394227 : SUCCESS in 1h 44m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/88320817b3ed4b61a24ec6bd51283efb : SUCCESS in 32m 50s","accounts_in_message":[],"_revision_number":1},{"id":"09e55510c27ca8de5048df428e7fb4e0afa4f00d","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:25:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"56d46065a04a0a691b7b942c4ff585f8690f2227","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 10:26:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a4c168de3b4d0390226ee8a5b326d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c469d9e06d9b49a7bcf70ea490805664 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0","revisions":{"5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0":{"kind":"REWORK","_number":1,"created":"2024-08-30 06:00:02.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/68/927568/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/68/927568/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/927568/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/927568/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/68/927568/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/68/927568/1"}}},"commit":{"parents":[{"commit":"8540ffc6ee34976285209e9f3c7c728bf649e0ff","subject":"Fix LP#2054575","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/8540ffc6ee34976285209e9f3c7c728bf649e0ff"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-08-30 06:00:02.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n(cherry picked from commit d8de63a5008b811f2e22cc5cdda59bd2cc05b207)\n(cherry picked from commit eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
