)]}'
{"id":"openstack%2Fglance~927580","triplet_id":"openstack%2Fglance~stable%2F2023.1~Id706480e31687d8ade6f7199b600aff3ad7c68f7","project":"openstack/glance","branch":"stable/2023.1","topic":"format-inspector-stable/2024.1-stable/2023.2-stable/2023.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-08-30 16:45:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id706480e31687d8ade6f7199b600aff3ad7c68f7","subject":"Add iso file format inspector","status":"MERGED","created":"2024-08-30 11:10:39.000000000","updated":"2024-08-30 16:48:10.000000000","submitted":"2024-08-30 16:45:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"927580-format-inspector-stable/2024.1-stable/2023.2-stable/2023.1","cherry_pick_of_change":927568,"cherry_pick_of_patch_set":1,"meta_rev_id":"8f791c9780d914f720b00daa7b779af2ae66673a","_number":927580,"virtual_id_number":927580,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-30 16:45:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-08-30 13:35:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-08-30 13:35:13.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-30 11:10:39.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"},{"updated":"2024-08-30 11:23:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-08-30 13:12:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-08-30 13:35:13.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"89d61d110317cca0455dc79ac5e0c414dcb91b9e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-08-30 11:10:39.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"521aef7a6fd875599727a353ee0817bec92edd2b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 11:23:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92dfe9bac9db4192be0ad8a2e61dd200\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9895ca05a3ad4cd58fff7f76ad5ae53b : SUCCESS in 10m 50s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fa079453443bcdd29753c1c4d82e61e74ca333bc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 13:12:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/788a44b94e714c3bb98445284dcf8bf6\n\n- grenade https://zuul.opendev.org/t/openstack/build/5b5d4e2d396446dd8593c2b9cf55a9cb : SUCCESS in 56m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/169a45c6676c451f9e1377f4035feaea : SUCCESS in 1h 02m 36s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/3dc196618c7442aaa1f6e499e3a8f9ba : SUCCESS in 1h 19m 10s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a0274fd668644c40a775c920897cf6da : SUCCESS in 52m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aadd2d4a72e6442eb33262a2122b8b20 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfbe9c648c3c4ff2b47d1d9fa40cdc6e : SUCCESS in 5m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b84743e8986646789754fe0a9e331b9f : SUCCESS in 5m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7eaad94445f847d8ad9eaba2f4699bb1 : SUCCESS in 8m 45s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0dbc3921d7434fb2a93d63237b1c5cbe : SUCCESS in 8m 53s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/faae9d0bbe074bc6ba8ee2533fafe173 : SUCCESS in 1h 38m 33s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/8282ecd575484f3aa50d5250b015c0a8 : SUCCESS in 1h 20m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7a9130061d1c458ea20b1ff5fdf17530 : SUCCESS in 1h 19m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d1c2c049cd004863a772d5ac19f5ebc9 : SUCCESS in 1h 42m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/83a105f4fa4a4b68b842fd8be0739874 : SUCCESS in 1h 37m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/69c7263e37ba48fa90cf6166f037594a : SUCCESS in 54m 05s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/08ffd7d794d04dfabfc36b7ad8f84507 : SUCCESS in 1h 55m 44s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/3429385e3c43486eb2dd64caf6932b98 : SUCCESS in 26m 31s","accounts_in_message":[],"_revision_number":1},{"id":"fe59eaeacf103e45d7ed6c9b7ce323fe74114886","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-08-30 13:35:13.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d4dbc272cc1003a47ffa3236c3bbf35c1c674662","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 13:35:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0506d962e220e16f150a10ccbc0f70f92830b8aa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 16:45:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/28212797dd0046bcbee8bac64e16ce07\n\n- grenade https://zuul.opendev.org/t/openstack/build/47710ac1f82b4ac79367da53117c5d67 : SUCCESS in 53m 49s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f16b541ad1dd48ba817048801580acbc : SUCCESS in 56m 32s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/fb787317c746482b83c5289a05b33cc2 : SUCCESS in 1h 20m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8b0bde2604274faba390399f281f0d8a : SUCCESS in 1h 00m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83beeb1976a641fdb2db25756d3d06c8 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8428d5c867547f28b2066d8909c6129 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/55acd438e1ed46808fca496604f8d6a8 : SUCCESS in 7m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/22089281d5554e1d97dab60fe39b55b8 : SUCCESS in 8m 17s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/631bef825a9d4c30b3de5a1bd93afd42 : SUCCESS in 9m 42s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/00c1cbbe96df4c5092af54122c84c226 : SUCCESS in 1h 23m 52s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/a0b5e581cf4e488b874c29b10a11a438 : SUCCESS in 1h 36m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/f6fb26d50a2245ca852d7b4ff29953c1 : SUCCESS in 1h 22m 46s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/53a67a2bdd7e40d3955b0057971b2658 : SUCCESS in 58m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/379fc6f0bf144a6c890ff2f564c844c7 : SUCCESS in 1h 38m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/8bafd50c1e0643f19908d40cc1954108 : SUCCESS in 29m 38s","accounts_in_message":[],"_revision_number":1},{"id":"572095d0390f63fc54993f055058a83ee1b68e32","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 16:45:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f791c9780d914f720b00daa7b779af2ae66673a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-30 16:48:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a0d9fb2efe1426e8e0fa69d6c414729\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/74910d47b3ab4f8a8ee2dce7b9cb1917 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1427857a7d2a823bb754cf2f12f2b10e7b3df31d","revisions":{"1427857a7d2a823bb754cf2f12f2b10e7b3df31d":{"kind":"REWORK","_number":1,"created":"2024-08-30 11:10:39.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/80/927580/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/80/927580/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/927580/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/927580/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/80/927580/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/80/927580/1"}}},"commit":{"parents":[{"commit":"6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d","subject":"Fix LP#2054575","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/6a17e4e2bae0f3661ffed08b2d4360dfc8b3841d"}]}],"author":{"name":"Abhishek Kekane","email":"akekane@redhat.com","date":"2024-07-08 19:28:38.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-08-30 11:10:39.000000000","tz":0},"subject":"Add iso file format inspector","message":"Add iso file format inspector\n\nThis change excludes image conversion if source image\nformat is ISO.\n\nThis change includes unit tests for the ISO\nformat inspector using mkisofs to generate\nthe iso files.\n\nA test for stashing qcow content in the system_area\nof an iso file is also included.\n\nThis change modifies format_inspector.detect_file_format\nto evaluate all inspectors until they are complete and\nraise an InvalidDiskInfo exception if multiple formats\nmatch.\n\nRelated-Bug: #2059809\nChange-Id: Id706480e31687d8ade6f7199b600aff3ad7c68f7\n(cherry picked from commit d8de63a5008b811f2e22cc5cdda59bd2cc05b207)\n(cherry picked from commit eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff)\n(cherry picked from commit 5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1427857a7d2a823bb754cf2f12f2b10e7b3df31d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1427857a7d2a823bb754cf2f12f2b10e7b3df31d"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
