)]}'
{"id":"openstack%2Fglance~928559","triplet_id":"openstack%2Fglance~master~I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","project":"openstack/glance","branch":"master","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-09-09 13:15:29.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-09-09 08:50:33.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","subject":"Fixed typo in discovery.py","status":"MERGED","created":"2024-09-09 08:39:23.000000000","updated":"2024-09-09 13:16:45.000000000","submitted":"2024-09-09 13:15:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928559","meta_rev_id":"c3b23a49798902b0951cebef2fc37948d0c95ec2","_number":928559,"virtual_id_number":928559,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-09 13:15:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"recommended":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"all":[{"value":1,"date":"2024-09-09 09:18:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":2,"date":"2024-09-09 08:50:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":1,"date":"2024-09-09 08:46:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":1,"date":"2024-09-09 09:21:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-09 08:40:46.000000000","updated_by":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-09-09 08:46:39.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2024-09-09 09:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-09 09:18:50.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"state":"REVIEWER"},{"updated":"2024-09-09 10:49:47.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"67eb2d11d32fc2c86fe6ef5c36a56a8db74c7b00","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-09-09 08:39:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0d0ce613f226945258807089405f930a127307c4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-09 08:46:39.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"5e767d72a5fb136254b76f7a7c4740ffb3e4a7ec","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-09 08:50:33.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"27e3915664f627f102fae4d15dae02f521753a5b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 09:09:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f09ac85e4ba3432aa9567eb7a96faca0\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/206161eb244746d3b5c89ccc453b8d10 : SUCCESS in 11m 05s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/195b245912a04d5986e76b4fe600eb59 : SUCCESS in 10m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ece337aa161808c9f36f6a41097fa8f3ab227186","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-09-09 09:18:50.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"78d4c6dbb28e2aea993acc7fa174e799d122ab61","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-09 09:21:50.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"345509afcf2fd0a68b0f203f201cae9e13f6ad62","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 10:49:47.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acd741647f4e4e67ac69c9f1d33f9a21\n\n- grenade https://zuul.opendev.org/t/openstack/build/c09fa3aa46884bf7bb5a951c3d0d6a8e : SUCCESS in 58m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/0ddc58c0841d465baec23e7e011cf454 : SUCCESS in 1h 28m 39s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/fef605b2bdf54240b1d3c27ee6f8a920 : SUCCESS in 57m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6711c105de4ff4bbb439e24e6468d9 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a70838d65962425cb98774f732141820 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/db6088cd3ce541ba98bb383a9d5a82db : SUCCESS in 5m 09s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a9fa4867a78945419f8a89130ec3d18a : SUCCESS in 7m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78ea75fbc29741d7b8150ef6c8ed7668 : SUCCESS in 5m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e5c4a869c3eb4caea46291e5530db268 : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4826d9f276b747839874d77d493f2440 : SUCCESS in 7m 49s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fda5ccdbca674f43814e9868971093e7 : SUCCESS in 10m 03s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/cc440c4a43a349dbafcd459eaf98a4d3 : FAILURE in 1h 13m 52s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/3b95d22c40484ed69dac6074e06c1d31 : SUCCESS in 1h 43m 32s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/816c2bebde0047a6923d5ca45398fae8 : SUCCESS in 1h 58m 14s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7fc8f3156e7743c4944566269754028c : SUCCESS in 2h 03m 59s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46ed4106b6404d759129d942364e99c6 : SUCCESS in 2h 02m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/2859d716aec84745bc54398982137051 : SUCCESS in 55m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/f174e934953b4810a295d4e1a6c2466d : SUCCESS in 1h 39m 02s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/8624e41bc04048cdba15caa941ed07b2 : SUCCESS in 1h 48m 52s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e5bab354fcd445f59caad2e2788df8f9 : SUCCESS in 59m 52s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/9190c1f03cc645c0b88afa8bc271d3b1 : SUCCESS in 30m 18s","accounts_in_message":[],"_revision_number":1},{"id":"9bba823a77821ad16a1201453769bacd8b7b5211","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 10:50:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cac27a3d32a9bd47df9a099639ae21ae035b70a1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:15:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0938f91a55bf4ceb8667ec43f334a192\n\n- grenade https://zuul.opendev.org/t/openstack/build/089b558cb41a4581ad3d12d6edabbc60 : SUCCESS in 51m 41s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c1b5b32a7d8045dd98a193f9e9bfef07 : SUCCESS in 1h 32m 15s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/a8f5be5b099c4a68a472c646da3f836e : SUCCESS in 1h 04m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0370c77f582b4a6a8ecbb76645c45077 : SUCCESS in 4m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/62ef434ee93f46e697d90fb987163eb5 : SUCCESS in 5m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c626f20122e44c9887ff1aee67ec1d2 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d67e1626bb4e639b0c526546ba621a : SUCCESS in 5m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b94c3e7bf8134422ab67cd5b967fa914 : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f94e3a167e214c46b372932f355ae6ff : SUCCESS in 8m 19s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/82503877083c4c11a5ae26403fdd4bec : SUCCESS in 13m 14s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2ae193be6cdd4e11a086a390f2d6a480 : SUCCESS in 1h 51m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/b255916cd98643b8bdb876c06cf24dcd : SUCCESS in 2h 13m 20s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1e949fb3db7e4e839750a8ea0f97304b : SUCCESS in 1h 53m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e6b9738159544ccc9c4aea46668ccf34 : SUCCESS in 43m 51s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c59dda1208564dbe95ca1fad483ecf29 : SUCCESS in 1h 40m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/740ff598cc2d4bec8cee33d90f8414db : SUCCESS in 23m 51s","accounts_in_message":[],"_revision_number":1},{"id":"8a3b9c2b918e6216c5403f4e3688ed56b428daae","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:15:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c3b23a49798902b0951cebef2fc37948d0c95ec2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-09 13:16:45.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e00529f6f54e4bae917820a3ff073d5e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75c2e60706794da7af0852f2bc8b7e79 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ae69b51fa20810b86d32bc4d4f7b6a0db649364","revisions":{"2ae69b51fa20810b86d32bc4d4f7b6a0db649364":{"kind":"REWORK","_number":1,"created":"2024-09-09 08:39:23.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/59/928559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/59/928559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/928559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/928559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/59/928559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/59/928559/1"}}},"commit":{"parents":[{"commit":"695fcb67c9cd558db7c96d51290e69259a268ecb","subject":"Merge \"Revert image state to queued if conversion fails\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/695fcb67c9cd558db7c96d51290e69259a268ecb"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-09 08:36:27.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-09 08:36:27.000000000","tz":0},"subject":"Fixed typo in discovery.py","message":"Fixed typo in discovery.py\n\nReplaced \"additonalProperties\" with \"additionalProperties\"\n\nChange-Id: I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ae69b51fa20810b86d32bc4d4f7b6a0db649364"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/2ae69b51fa20810b86d32bc4d4f7b6a0db649364"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
