)]}'
{"id":"openstack%2Fglance~928757","triplet_id":"openstack%2Fglance~stable%2F2024.1~I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","project":"openstack/glance","branch":"stable/2024.1","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-09-10 11:34:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","subject":"Fixed typo in discovery.py","status":"MERGED","created":"2024-09-10 05:36:12.000000000","updated":"2024-09-10 11:35:41.000000000","submitted":"2024-09-10 11:34:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928757","cherry_pick_of_change":928559,"cherry_pick_of_patch_set":1,"meta_rev_id":"696c39a841bae2b84a7b280dbf99c90cec7a3cb0","_number":928757,"virtual_id_number":928757,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-10 11:34:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-09-10 05:59:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-09-10 05:59:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 05:50:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-10 05:59:07.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-09-10 07:36:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"13dbbac90572a8f75a7ec422abec14ff79453720","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-09-10 05:36:12.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"f9c6d8a36b779a96336e6a2977d62d91c814ea0f","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 05:50:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/562894a77ddc4a6385d5aeee88332019\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/95f9df26c2a64c4e85e2f39e37aeff88 : SUCCESS in 12m 02s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/006e2909424a4158a99eee44a2df2777 : SUCCESS in 8m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"264fa9b20ee65b1e52061bd254bb4970ee8b5b5d","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-10 05:59:07.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bffb0f9bbae4788406b5da981b569ca0225d8676","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 07:36:51.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/22f77f2371a141c9b4445310bc2c4ce9\n\n- grenade https://zuul.opendev.org/t/openstack/build/e02d05a06af240a59c4ca3387eb72675 : SUCCESS in 40m 03s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d29a9b8ec24d4587841bf0cfc995eb9e : SUCCESS in 59m 44s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b9aaf8bc242a4c0d999b28f866608167 : SUCCESS in 1h 18m 26s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d20c18d4a83b472888d0e85b30a057ed : SUCCESS in 47m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/002f905706244ef5913481a66376c561 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e2d98043d1541a0a47aad6b30a92caf : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dacb99d4785b4a0494f3876e316dc498 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3defdd2c982143d78d90345fe0f7d1e8 : SUCCESS in 5m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7475495e6ada444aa1797560a2bfa31f : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/659b72abed6c4933bb83c60627a04a6d : SUCCESS in 8m 19s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/10dfcd50333745cd9a2d07b4e03aab6b : SUCCESS in 8m 32s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/f21f5c0c528e4719afc6b9d1f9cbf54e : SUCCESS in 1h 15m 49s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/2475f1b165ed4bb4a5f5884abdc77a6d : FAILURE in 1h 22m 49s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/8973aebf90f64e9fb104c087f69c1a56 : SUCCESS in 1h 18m 27s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/fb80d9f4b083434d95830eb98db81520 : SUCCESS in 1h 22m 12s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/e6afc7b8480b4bbe93e9f3603c11126e : SUCCESS in 1h 26m 43s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/90442f9d53ac41659adda818f9f4acfb : SUCCESS in 40m 41s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/eeba6643917b4bf7be385428560c4bff : SUCCESS in 1h 05m 54s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9f7aa51fa73e414eb91c79f32b0d7107 : SUCCESS in 1h 49m 51s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/03d5d7ae313a432393388be9f5c3cab8 : SUCCESS in 41m 14s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1792fbd91bbe44d5b493337849eb8924 : SUCCESS in 37m 39s","accounts_in_message":[],"_revision_number":1},{"id":"d2dc49889841fca2652a8c79faf05ce8e96c05e1","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-10 07:40:29.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b0de83bfb684cef90dab599489edc3b92fa5d212","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 07:54:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/81ddd41946034feaab9b0f25b25875fb\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/e44e3d947e334ca08e447a7b71692d29 : SUCCESS in 12m 12s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/328bde5b25294617ad18d168cc69b6c1 : SUCCESS in 9m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"aaac7ae9779b68a3fb29027484d9fef4e584b5ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 09:33:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/243504ea0ccd4260b2c429f45fd9fea7\n\n- grenade https://zuul.opendev.org/t/openstack/build/7055488e0f1c4069bda7bf20fd3110c7 : SUCCESS in 1h 15m 57s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/34475b8c7e524dff8e577a5d5a16fb56 : SUCCESS in 37m 23s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ec26b39474024988a85d742d7e931a91 : SUCCESS in 1h 20m 33s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/56b137d0251947f5ba6384d6523313e6 : SUCCESS in 1h 00m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/220b75a8347848dc9d4a0bbff1dc35c3 : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fbd2261494d4a728b6a52af5a7402d7 : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22617e854abf4ccaa06ba67c80312203 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e18b8dea41fa4173b4e9404799fa137b : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c90274f4f48c4a53b07a6ce610cf2f41 : SUCCESS in 9m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/13c47dad23bb491e862bcf57405fa227 : SUCCESS in 8m 04s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/42db37a6f8214414bffee44ff331ef1e : SUCCESS in 7m 12s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/c8c7adc6ff4442b6a32399e71a316d5d : SUCCESS in 1h 35m 40s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ab2944a70ffb47d89f126716fa7df8bf : SUCCESS in 1h 46m 41s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/7c6466b8aaf4429a8dc828666dfbdfd6 : SUCCESS in 1h 37m 17s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7bca4faef3f243eb807b76d5529928a9 : SUCCESS in 1h 40m 19s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/7b39f4f411fb4f32848afe2b37276ad7 : SUCCESS in 1h 12m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/ac15f490cb7d4109b80ed7935054fd40 : SUCCESS in 57m 32s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/b33416a095b84e55bcb4e449e1260f23 : SUCCESS in 1h 34m 27s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/3fc5a6c37dcf4693aeafd753142ff7bf : SUCCESS in 1h 43m 33s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/a18344924718498aa430d5a97aadb576 : SUCCESS in 41m 12s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/93308ca52f07472cbdd5725756bc72a6 : SUCCESS in 28m 14s","accounts_in_message":[],"_revision_number":1},{"id":"b266fed9a30d5b96270b40d377f9987d3d13b8ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 09:34:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a8c58162312bfc7696b4651fd4ebd48cb2c54890","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:34:28.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3b52282bdd44687895ceb0e305bc9de\n\n- grenade https://zuul.opendev.org/t/openstack/build/2969f52e73e94416885ba6665d9918cb : SUCCESS in 56m 20s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/d69eef86edde48068f1371f3c05077aa : SUCCESS in 1h 08m 20s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/2166c1903d724d71a1407643d2285720 : SUCCESS in 1h 32m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7b1e3a0cf3444b5995285f954f4df87f : SUCCESS in 59m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ec045ee114d549739625a4e12d75cdbe : SUCCESS in 4m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f3003ddf48cb46cb90f42843aeb347cd : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af2331c522c9492eba3550c29b100627 : SUCCESS in 6m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bfe04d6762ad483d87bf6bc2e82c897c : SUCCESS in 6m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a06242b702224b44b9040e71f6b52aec : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/03f1bca574a14070ac15d37b0fe8e44f : SUCCESS in 8m 35s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5f918d79731c46449d1f9690f774611b : SUCCESS in 1h 42m 56s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9531f339b72d48b5bb617488fa249119 : SUCCESS in 1h 53m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/d7e80a1a35f04d23b3702a713dc0df88 : SUCCESS in 1h 34m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d8e08133566471ebf1f96956fbed8d1 : SUCCESS in 1h 00m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ea2859161994308b032e3d980bdc968 : SUCCESS in 1h 26m 15s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/72dc48fd5aa74924995c0e141d4ec3db : SUCCESS in 32m 46s","accounts_in_message":[],"_revision_number":1},{"id":"8259ae3bfce5a3fccd8c96195122df08be737022","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:34:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"696c39a841bae2b84a7b280dbf99c90cec7a3cb0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 11:35:41.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3bc4fda03c6b4fa7b5ee33b0b2892bcc\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a013ac7c00a6460d97ff7e3ef084dd6f : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0be1bededcc482946970312d92832f085d35e17","revisions":{"d0be1bededcc482946970312d92832f085d35e17":{"kind":"REWORK","_number":1,"created":"2024-09-10 05:36:12.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/57/928757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/57/928757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/928757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/928757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/57/928757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/57/928757/1"}}},"commit":{"parents":[{"commit":"eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff","subject":"Add iso file format inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/eed2a8aeb7a2e1ba47ff5642a346a97e85b6b0ff"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-09 08:36:27.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-10 05:36:12.000000000","tz":0},"subject":"Fixed typo in discovery.py","message":"Fixed typo in discovery.py\n\nReplaced \"additonalProperties\" with \"additionalProperties\"\n\nChange-Id: I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3\n(cherry picked from commit 2ae69b51fa20810b86d32bc4d4f7b6a0db649364)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/d0be1bededcc482946970312d92832f085d35e17"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
