)]}'
{"id":"openstack%2Fglance~928786","triplet_id":"openstack%2Fglance~stable%2F2023.2~I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","project":"openstack/glance","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"33765":{"account":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"last_update":"2024-09-11 09:56:25.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","subject":"Fixed typo in discovery.py","status":"MERGED","created":"2024-09-10 11:42:45.000000000","updated":"2024-09-11 09:57:27.000000000","submitted":"2024-09-11 09:56:25.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928786","cherry_pick_of_change":928757,"cherry_pick_of_patch_set":1,"meta_rev_id":"13091ffd4fccc81f0f71e155a797d107d3c949fb","_number":928786,"virtual_id_number":928786,"owner":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-11 09:56:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-09-11 07:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-09-11 07:50:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 13:45:46.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 07:50:41.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"}],"messages":[{"id":"a0d6dd8790d0cfe0f75eb1309cc3db49c4a72983","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"date":"2024-09-10 11:42:45.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2024.1.","accounts_in_message":[],"_revision_number":1},{"id":"91a64abb83989e6b722d281410951d464a64d14b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 13:45:46.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f3597dfcb75a4e44b300dc11061782c6\n\n- grenade https://zuul.opendev.org/t/openstack/build/f392a2f92e58403aaa7785d34ef2bdd1 : SUCCESS in 59m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a34a48124b7948c29220adc61cdb42ab : SUCCESS in 1h 21m 20s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/34890b7827f847aaa21a6b8c14417c2b : SUCCESS in 1h 03m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ff185a7960648c58be8e474cc0c2bd7 : SUCCESS in 3m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e70e61085942483c8a8fb91246d13aa5 : SUCCESS in 6m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4b6b78d9a824483f805ac4882ffdd427 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f45007b2afcb4331bc3ce47592f79bcc : SUCCESS in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6f964b65faf24c009a968ca877a6bea0 : SUCCESS in 5m 37s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41f61aa7cef84835a09fced7d7c5bcfb : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7263a1f9b8af44579e2d4ea4809f28e5 : SUCCESS in 7m 56s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4d6f263690ee46a2a0ba2fabb2178392 : SUCCESS in 7m 44s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/8a9fad425146494ebd434b4bdcf3259b : SUCCESS in 8m 40s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/94cefc14fc494f52a30192944a318274 : SUCCESS in 1h 37m 37s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5105531b3d9d4b5f81db904eef6382fb : SUCCESS in 1h 57m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d3f4acf7eea94075a60a65dda951f6e1 : SUCCESS in 1h 55m 57s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0ffe83a41a5f4aeab2540aa39a852dfc : SUCCESS in 1h 40m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/73418d0160384257be94cfd6ccd3260e : SUCCESS in 1h 48m 34s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/02ffb3aadee647e5afb4d8ee150a85a3 : SUCCESS in 40m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/97350b8b40024e1695ba51252bfeee22 : SUCCESS in 1h 11m 58s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f8f52637e598413b9f5325c6b9e65346 : SUCCESS in 27m 54s","accounts_in_message":[],"_revision_number":1},{"id":"0c0371fa31d25048c074543888b79f5656426a80","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 14:21:25.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ef150b7a4e54efeb038cb094e250920\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/1dd0357da074468a83e6f4139f186fb9 : SUCCESS in 16m 57s (non-voting)\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/c3109d033120401584475afaf56ed2ff : SUCCESS in 16m 10s (non-voting)\n- openstack-tox-py310-arm64 https://zuul.opendev.org/t/openstack/build/942f812259d2464e8ea71e62edf9db1c : SUCCESS in 25m 55s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/bf1480154d73479cbba58550dba28e39 : SUCCESS in 17m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f7b94774c18ae0cd69a60509e41abb570d2310c4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-11 07:50:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3dde8fafb6e90e062d83427fee24924802c7b6fa","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 07:51:18.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"70a6604bc652596b511ab187d569544ea2cc66d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 09:56:25.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2dbc65b1aaca429bbc51023113fbef1a\n\n- grenade https://zuul.opendev.org/t/openstack/build/46039ca5712f4a33aa93fe545fa59b83 : SUCCESS in 46m 55s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ae7af48e7f2d4d06ab95804105bfa7b0 : SUCCESS in 1h 50m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/58db80b591424c17b1e886b665e86825 : SUCCESS in 1h 00m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8f702beb23641a49f473fb0b1277550 : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8a4b1d9ea61c4860ae980495656df7ed : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/87a8b0fc4e464482b08bcadb57af9c6d : SUCCESS in 5m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e2ac6ea4a0284407bc8abdf4b3531eb6 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ab64070f63a4530a635a22f1d39c9d3 : SUCCESS in 4m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3a1f9babb5ab47e692fb8f2df4de2c5b : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f1ec30249e784295bce2c32526eecd33 : SUCCESS in 7m 20s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/15f41c843d514851a95be1fbb07154f3 : SUCCESS in 1h 09m 44s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/767ae6ba84004673b8762ba87fd965e3 : SUCCESS in 1h 53m 06s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4b66ac8d69c646268a6c12e881270d5d : SUCCESS in 1h 54m 25s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/66ca002b022548b49f2fa6e7d95f6300 : SUCCESS in 40m 07s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4b427d76e63f411b8b119ea9f466894a : SUCCESS in 1h 45m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/280aafe65e724ee3aed4f52d749b9935 : SUCCESS in 38m 09s","accounts_in_message":[],"_revision_number":1},{"id":"ffdc3c9a1d3ffe84dab936f8e10453cc59dacf83","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 09:56:25.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"13091ffd4fccc81f0f71e155a797d107d3c949fb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 09:57:27.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f5c677e896842c8ba7d8a568233ac66\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb2f0cf534524be4b109f45e9e4613c7 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"18453f86021947168d799b3d48e5b3131c987137","revisions":{"18453f86021947168d799b3d48e5b3131c987137":{"kind":"REWORK","_number":1,"created":"2024-09-10 11:42:45.000000000","uploader":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"ref":"refs/changes/86/928786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/86/928786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/928786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/928786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/86/928786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/86/928786/1"}}},"commit":{"parents":[{"commit":"5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0","subject":"Add iso file format inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/5b4be5b6f2f67ec8e61bceff8c2ed79b870aace0"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-09 08:36:27.000000000","tz":0},"committer":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-10 11:42:45.000000000","tz":0},"subject":"Fixed typo in discovery.py","message":"Fixed typo in discovery.py\n\nReplaced \"additonalProperties\" with \"additionalProperties\"\n\nChange-Id: I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3\n(cherry picked from commit 2ae69b51fa20810b86d32bc4d4f7b6a0db649364)\n(cherry picked from commit d0be1bededcc482946970312d92832f085d35e17)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18453f86021947168d799b3d48e5b3131c987137"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/18453f86021947168d799b3d48e5b3131c987137"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
