)]}'
{"id":"openstack%2Fglance~928820","triplet_id":"openstack%2Fglance~master~Ifdc92d4cc25d4683538c727e2ed24df6e0b6030e","project":"openstack/glance","branch":"master","topic":"use-oslo-inspector","attention_set":{},"removed_from_attention_set":{"4393":{"account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"last_update":"2024-09-12 18:42:02.000000000","reason":"\u003cGERRIT_ACCOUNT_4393\u003e replied on the change","reason_account":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"}}},"hashtags":[],"change_id":"Ifdc92d4cc25d4683538c727e2ed24df6e0b6030e","subject":"WIP: Retry image update on closed transaction","status":"NEW","created":"2024-09-10 15:09:02.000000000","updated":"2024-09-12 21:00:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":6,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"b1f9515f617d74d5b8beead70b9d10fdfd8456a7","_number":928820,"virtual_id_number":928820,"owner":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-09-12 21:00:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-10 16:20:40.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-10 17:51:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-09-11 07:42:23.000000000","updated_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"CC"}],"messages":[{"id":"a2c189419b653b685df4fcf8dc24d72cab114f7a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-10 15:09:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f38416c45f897ca359ab76a785ea22dedb03faef","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 16:20:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/024627f24d0f41239040b9c4b637003b\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/ca55747e5f124f8c890adf4040a08b6d : SUCCESS in 14m 10s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/1c026ec0925846adbca2880a88c01f62 : SUCCESS in 13m 27s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"80508d40d8cc2bd437a3677edddb67d178b9cdc4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 17:51:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/63d63f1dfbf14547affd5298628112fa\n\n- grenade https://zuul.opendev.org/t/openstack/build/688cb35889b3434680523813cc2fab87 : FAILURE in 23m 34s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/84072998e92b4fc7958f560585e57295 : SUCCESS in 1h 05m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6e1584f138e645d88938f11fdcbccbb0 : SUCCESS in 38m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1c6b20f3cd3b4a4db8ec08f64377d556 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e657e949b00845b294510a809e09084b : SUCCESS in 5m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d8375843de58453b97d15247d67bbcb0 : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e90e02957ead412da103159eb3617870 : SUCCESS in 6m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6d95c80b75824fa1968c5dd6b2a85adb : SUCCESS in 5m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/71e58b3c670b41019f2a827b43884d2a : SUCCESS in 9m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5c32f84aa2cc4df083e11bbba074837e : SUCCESS in 7m 37s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/a3bb8b2ddee44fcea034fb6bd95b9e3a : SUCCESS in 12m 24s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/895d35fc5fce4d858dc0ed4bd6313521 : FAILURE in 1h 25m 41s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/d1d16ec526954e5da9b6c271b9d76f90 : SUCCESS in 1h 53m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6b9def3a9ec24e9f9cbc430b9d8fa5a2 : SUCCESS in 1h 42m 18s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/bccff12b38ba424ea8249507e5a346e9 : SUCCESS in 1h 45m 47s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/a805fdd1acd640138b2ece94b5d39e6c : SUCCESS in 2h 05m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d53b095c2d944a7babbb0b75aaf97393 : SUCCESS in 38m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/7ab6cf4fc4be49ac887a9c6d5e96540f : SUCCESS in 1h 20m 42s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/02a76accac05430ca086062a2dc5bab0 : SUCCESS in 2h 41m 18s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/27fd8ede52684b11a438a780ad8053d4 : SUCCESS in 47m 59s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/15bb4b00cb38442e8db20d7b9aa077ad : SUCCESS in 34m 25s","accounts_in_message":[],"_revision_number":1},{"id":"fbd580c9770cfe5e7ffa874f405d6c2a412ae5fb","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-10 17:52:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8541073406289b5139d59689ff445a249e03192d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 20:05:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f214967fa8f434fb2c63291b8bfe4ef\n\n- grenade https://zuul.opendev.org/t/openstack/build/af022b61383342e98b3fa5c49cbbae61 : SUCCESS in 1h 12m 49s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/7e971ca0fb48484b84764e56deb42ec2 : SUCCESS in 1h 26m 37s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7864d848606643b7b199603f27f377b1 : SUCCESS in 1h 24m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5bd7805c3de4882b21d26a713e17675 : SUCCESS in 6m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/36d54b229cb84e7a92d575d7a711bd1b : SUCCESS in 3m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8b7c4d43095b44b8a1490cd60404b8e1 : SUCCESS in 4m 47s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c60434bc2d664c7da3258b13f865763e : SUCCESS in 7m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/94010a91ac03473e90f73cc7926f6e83 : SUCCESS in 5m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/21e5a5f632ee46bebd12c15d2db9cbe3 : SUCCESS in 7m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e78945c75eef4daeb9ea7ade7f41ac5a : SUCCESS in 8m 31s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/1c5c9807402e498cb3c4f184d1c2162d : SUCCESS in 10m 53s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/49b2df2d2149498b92c54658bf194c05 : FAILURE in 1h 56m 14s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/ea9d0117c1b2451f9842cbe3d7c2d825 : SUCCESS in 2h 07m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/bd489387cda044a18d02f4df41e3b9b7 : SUCCESS in 1h 44m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dd166ad2b1b6405a90f52d70bede19bc : SUCCESS in 1h 53m 55s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/df400044b7cd4aeeb78853ad92a77a98 : SUCCESS in 2h 04m 39s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/77a01a54d08d41dda89f90376f8e2a36 : SUCCESS in 56m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4e039bec7e3b4e5aaded6738aac7645e : SUCCESS in 1h 33m 52s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/9956e8e9b468415caeeb05addc4b0f3b : SUCCESS in 1h 33m 39s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/ccbf69b26b6b419caf242c8b37fe7b8a : SUCCESS in 59m 56s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/53da77c749cc4c63886964dbdf3dde1f : SUCCESS in 30m 08s","accounts_in_message":[],"_revision_number":1},{"id":"c3417be2ec7311dcff0a734f57feb004adefc786","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-10 20:50:14.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7e965e7a05e811483e6eea5ef725c5261c33ee07","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 20:56:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a75ddec14f3a4ae4a759a73b3c25a6c4\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/a3e01cddcd6f419a97a1c9eee58b094f : SUCCESS in 15m 03s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/57e3fbe5adf74af08c6c0fd9bee70e79 : SUCCESS in 13m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3570afef4dd430f8434ed8aac04715ba909bc64b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-10 23:12:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cff50117ebb04a258a1961f85a944b7d\n\n- grenade https://zuul.opendev.org/t/openstack/build/8fb7eaf8efad4905bb997a903b9febcf : SUCCESS in 1h 06m 46s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4759dc39949e4477b1529889eabc1ed9 : SUCCESS in 1h 17m 19s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8adb415637a944a28a83b9f60438f09c : SUCCESS in 1h 01m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f24cde5d0384386bd17277330a2d81e : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da490d63c8234556a78e92f29eb5582e : SUCCESS in 5m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2ee3370be4b64dc9801610c45f0ea6e0 : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3688bc59beab4de38430fbf133e1e11e : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c10c69737d14411eb4835509bc9deab2 : SUCCESS in 5m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25da8269a88243f58fdea1aaa508d84a : SUCCESS in 7m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/69c93f63b80c465291023273f98661ca : SUCCESS in 7m 46s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/59e1f5cff30c4dcf9aeb48882480ea05 : SUCCESS in 13m 59s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ee4c0cbc0db44565b2aff2835a2c0570 : FAILURE in 1h 36m 38s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/5d2b8db6339e4616b2fae4bf59eed5eb : SUCCESS in 1h 48m 37s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/e7162d4a6fc246cc9461249a9b608ac3 : SUCCESS in 1h 55m 25s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/49bcafe2a48a462da9ca897f67103f7b : SUCCESS in 2h 04m 23s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/58fd015ef0084649819e572319d0e2f1 : SUCCESS in 2h 16m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/8e473d1016244c439819e8d095ab4338 : SUCCESS in 58m 16s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/348a82c570f548fe9eee762ddef99aeb : SUCCESS in 1h 49m 09s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/230f6bbfd4ad47089a2255ad403d4088 : SUCCESS in 1h 41m 05s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/0e40edd2892f4f53923965207e3ad7c4 : SUCCESS in 1h 17m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/696b2dd6828f49f1bb0580b63e77de41 : SUCCESS in 33m 58s","accounts_in_message":[],"_revision_number":1},{"id":"87747669c0d1477d07b8e8afef33a18b66b83031","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-11 07:42:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a88649253b808619f65652d36e57db707eefbd8a","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 08:00:09.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/94750d4a86e24a689c82312490f0d0fe\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0bf94812635d4dc2854d8b8733437c83 : SUCCESS in 15m 23s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/3844f78c3402420f99f5664e002a9089 : SUCCESS in 9m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1c68aea69e6b6c7f5e4bc1966ce6c5051b9cd70","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 10:20:54.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5c9f6bddd39342d7bf7df51bbc15a6a4\n\n- grenade https://zuul.opendev.org/t/openstack/build/c5f3fd64bc6f4a25b485d8594c689a92 : SUCCESS in 1h 07m 53s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e5423f2ae73343ea96d08f2352e445c6 : SUCCESS in 1h 33m 52s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/04dc3454f8814d879c97aae3b3ef8acf : SUCCESS in 53m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a7ed5793886406a91b18c62b6d93f47 : SUCCESS in 5m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/af06fc4cabf1496abce1c1ee9fb1e761 : SUCCESS in 5m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/63f944a2345c4df4a65e5a1045e864ca : SUCCESS in 3m 49s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/deb588b1ba764038ab680d0352f7668f : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f62aa3afcfe747a9b89b19fb7f6d9676 : SUCCESS in 5m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e5b1d69e950943d1b097db0f3eeb8bdb : SUCCESS in 8m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/111fd2267e174adda0a4b3a5d83f0adb : SUCCESS in 7m 12s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/4ba2659370184707857e2c78e59c108d : SUCCESS in 12m 29s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/bbc5aab307a6418eb286f75d3538f59f : TIMED_OUT in 2h 03m 09s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/44c7dc96a38645f89196e964fa378e2b : SUCCESS in 1h 44m 43s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/163593f4b4eb4bee9dfa2ef50319e047 : SUCCESS in 1h 47m 42s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0d573464f8a34b868c5a4660ec9a4800 : SUCCESS in 1h 38m 54s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d6cc231da9e04da09a9f0de702a1e6c7 : SUCCESS in 2h 16m 37s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/60026b1c32124f199160968d8ca234f9 : SUCCESS in 53m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cf4d755263db456d90745530061d341c : SUCCESS in 1h 27m 04s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/cd4d32e0c3cc4fd28afc542a3052042b : SUCCESS in 1h 36m 48s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/4197943bc296406f9ef5c967480525aa : SUCCESS in 54m 42s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/db8428dee3fc4a4caa2eb560d7c2f440 : SUCCESS in 36m 20s","accounts_in_message":[],"_revision_number":1},{"id":"fe1001bb42e721056ed9f14375c8e34c787e59b8","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-11 20:30:26.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"659d1bb2c4dceb92d0a0ba91de059c08f7101bfb","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 21:02:48.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a6c6180add0545398bd9ff28a67033ab\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/0a105d8ae5d442babc6ea34a040f47b2 : SUCCESS in 10m 20s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/7daeed2421d243f1a17d918b8518fd3f : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fe543b451c971755db31db9e015010bc09b22bcc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 22:47:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e9459501c1524e1586439455c0d806d3\n\n- grenade https://zuul.opendev.org/t/openstack/build/f1459ac4b73d4c58968f4ad78135d8cd : SUCCESS in 1h 06m 42s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/c6e9a534f1144dc198f068bd4b39b40b : SUCCESS in 1h 31m 50s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/86daf0e91c51493390370bcb559c3694 : FAILURE in 36m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea6f20123d884ff1b257e6bd79e7c2a8 : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6235f1a1472a45e88c170cab5174e411 : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b22ad9f89ce4469dbdb3bbe283d90f13 : SUCCESS in 3m 36s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/e62625b7ff724da9b392c95b3087c411 : SUCCESS in 4m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25f5725bf7724842bd99f15a617726cc : SUCCESS in 5m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e17578a862034392af2f7b71226fdb0a : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2c084e188a7b4432880bdf15456e36c8 : SUCCESS in 8m 01s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/079600198aa14f46bdba0673e40bf9f4 : SUCCESS in 14m 27s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/e75f04d2bd00464d849f53ac7baac32e : FAILURE in 1h 11m 44s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/a412d56ffd64467590cd199a579a3a70 : SUCCESS in 1h 51m 10s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fc2636946b364b0c86423ad353ce3a22 : SUCCESS in 1h 56m 58s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/4bc564620ba6458585877eb763b9f5cb : SUCCESS in 1h 54m 16s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/c17a6f344f1c465bab457d4dd07e846f : SUCCESS in 2h 14m 45s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/94a00291eabf4c9eba56a1ffdcbcb570 : SUCCESS in 46m 01s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/9c69118160044575b268a5db66848f02 : SUCCESS in 1h 27m 32s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/99215e0b43cd4b759b5aa6b814b86d96 : SUCCESS in 1h 34m 07s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/e7564d1cc90a4f98a68dd38b8ca6ce66 : SUCCESS in 52m 27s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d39fcc5e9a2e4ed8bca76c66fe7f3777 : SUCCESS in 24m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fc236b0f37ba1a2008636794575f0ad7854277e0","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-12 14:21:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e58b330d3e206a93932e58b8bb01fd7f665d4c14","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 16:20:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4891a18a115948c0a9a9729dc0cb862f\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/3059e84232834d6a8c899f7fa3b3f93b : SUCCESS in 18m 26s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/e0aff83a278f44659562055a48a22709 : SUCCESS in 12m 08s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4e324b46143cd29c9a628705abea5254d2adb6c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 16:56:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/59becdf0602a421eb573fe07bc1a29f0\n\n- grenade https://zuul.opendev.org/t/openstack/build/87a62c5129364d0b8149b74d41db42b2 : SUCCESS in 1h 01m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a35e17c2451946e7bcae7f2b25409238 : SUCCESS in 1h 17m 58s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/15b1dba404264e4bbb48024bdddba143 : SUCCESS in 1h 00m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a56fb8d9d64d45ca8f69a340390dda53 : SUCCESS in 8m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d93c129fb6e04cfbaa5afa7105ba893b : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dadd4cf06fa9486b982ffea21213c7c8 : SUCCESS in 8m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2f5c480206204ba09fa93a5f5137fff5 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8f3445b342d466aaafb96c6784488bf : SUCCESS in 9m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dd99c8c9fb4345c48a39c4b42373eb35 : SUCCESS in 10m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d652fbd6176c4ad5a045a8f0f649d24f : SUCCESS in 9m 23s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/8b8fadf1dd3d4583948f3129ca5599a5 : SUCCESS in 15m 09s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/35d32125f40144a3b3c695b0201fe1c5 : FAILURE in 1h 43m 31s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e699d32e39594ed49863c7463964a53f : SUCCESS in 2h 09m 24s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/3acc83b4240e4906920d60b6e6ad289d : SUCCESS in 1h 52m 37s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/88e8621ecd874bee8187252169ca82d7 : SUCCESS in 2h 13m 53s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/ac0781f9973c40d1a9ba1a36f99247cd : SUCCESS in 2h 27m 57s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/aeafda20baa9466081ac77a92790eb69 : SUCCESS in 45m 13s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/14ca86a022004a34a390e81208ef07b5 : FAILURE in 2h 03m 41s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/1efef0db833547d3b42a8ff427a9b03a : SUCCESS in 2h 01m 23s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/17363aa704074cfeb62a4646bbe2a20a : SUCCESS in 57m 57s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/eee79563ac8e4922a47e3b20863bf482 : SUCCESS in 31m 55s","accounts_in_message":[],"_revision_number":1},{"id":"3bc79a5bac7e34b96fc74c393c401db1731bc737","author":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"date":"2024-09-12 18:42:02.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"97e26be1cb6a81deef5ae6414f3bb8a066ddb011","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 18:53:36.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec5b15f6cd0e42ca918a88d77eb2a7c3\n\n- openstack-tox-py39-arm64 https://zuul.opendev.org/t/openstack/build/52b0e323e35b4f0fb14015af4e2d1bd9 : SUCCESS in 8m 29s (non-voting)\n- openstack-tox-py311-arm64 https://zuul.opendev.org/t/openstack/build/4e702615c20f49a0ada248ee8b4f195a : SUCCESS in 8m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"b1f9515f617d74d5b8beead70b9d10fdfd8456a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 21:00:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b07b2300e0dd4a3394db9ffe4c75d9eb\n\n- grenade https://zuul.opendev.org/t/openstack/build/9b17eba33b9345799ac4f965f638e124 : SUCCESS in 1h 02m 57s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48634748c8c340858d2e6237db2b1a8a : SUCCESS in 1h 36m 30s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/153d186c841a4b8a8a79aa4cc2241cca : SUCCESS in 1h 02m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8c9674ba7f1f49e5a17b1f7cff76fa71 : SUCCESS in 3m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd8e2343f35f43ea862ef52ea7c47282 : SUCCESS in 4m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d51be60e618f4051b4792d8082a7283c : SUCCESS in 5m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3496f2693a454e739759d62f32f38c69 : SUCCESS in 6m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cc422826600a4a4ea3ad1e80a98988ac : SUCCESS in 6m 04s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b349656b42cf4fb1b5d02476918e7c57 : SUCCESS in 9m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ae4f69ea4ffb47e089633aefe41b49ee : SUCCESS in 6m 58s\n- openstack-tox-functional-py312 https://zuul.opendev.org/t/openstack/build/fcff328d28714817b3e0dfe0c684f02c : SUCCESS in 12m 37s\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/b88f8f5917ab486398be4b00a867659b : FAILURE in 2h 01m 17s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e678cdb42f754486b9f9fc02b6cf5da1 : SUCCESS in 1h 48m 25s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ca0decb6a43c4118a46e6b20b023b508 : SUCCESS in 1h 55m 45s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/5bde1c8c82884a50940e1fa75ca7e436 : SUCCESS in 1h 47m 39s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3ae53d65ef9a47a1ba022c68a235ad25 : SUCCESS in 2h 17m 51s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0fd893549a4f4da1880f9847a75fda33 : SUCCESS in 58m 00s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/27f1c2eec6914603bc63545674074d1e : SUCCESS in 1h 41m 23s\n- glance-centralized-cache https://zuul.opendev.org/t/openstack/build/b208a24657fd4a719fefaef95b4d9172 : SUCCESS in 1h 53m 32s (non-voting)\n- glance-grenade-centralized-cache https://zuul.opendev.org/t/openstack/build/8134d0395ce44c229149fd93864d32cd : SUCCESS in 1h 00m 00s (non-voting)\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1b0149fcc41a4d389878cfdb5001570f : SUCCESS in 27m 10s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"431f9f92573f132630e74d8cad986aec1588a7c6","revisions":{"431f9f92573f132630e74d8cad986aec1588a7c6":{"kind":"REWORK","_number":1,"created":"2024-09-10 15:09:02.000000000","uploader":{"_account_id":4393,"name":"Dan Smith","email":"dms@danplanet.com","username":"danms"},"ref":"refs/changes/20/928820/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/20/928820/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/928820/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/928820/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/20/928820/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/20/928820/1"}}},"commit":{"parents":[{"commit":"0be2737d66ea90b64270e09402ea5b088a28f415","subject":"Use format_inspector from oslo","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/0be2737d66ea90b64270e09402ea5b088a28f415"}]}],"author":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-09-10 15:08:37.000000000","tz":-420},"committer":{"name":"Dan Smith","email":"dansmith@redhat.com","date":"2024-09-10 15:08:37.000000000","tz":-420},"subject":"WIP: Retry image update on closed transaction","message":"WIP: Retry image update on closed transaction\n\nChange-Id: Ifdc92d4cc25d4683538c727e2ed24df6e0b6030e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/431f9f92573f132630e74d8cad986aec1588a7c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/431f9f92573f132630e74d8cad986aec1588a7c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{"label:Review-Priority\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
