)]}'
{"id":"openstack%2Fglance~928894","triplet_id":"openstack%2Fglance~stable%2F2023.1~I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","project":"openstack/glance","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"19138":{"account":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"last_update":"2024-09-15 05:58:35.000000000","reason":"Change was submitted"},"9303":{"account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"last_update":"2024-09-11 10:40:48.000000000","reason":"\u003cGERRIT_ACCOUNT_9303\u003e replied on the change","reason_account":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}}},"hashtags":[],"change_id":"I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3","subject":"Fixed typo in discovery.py","status":"MERGED","created":"2024-09-11 10:37:52.000000000","updated":"2024-09-15 05:59:35.000000000","submitted":"2024-09-15 05:58:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"928894","cherry_pick_of_change":928786,"cherry_pick_of_patch_set":1,"meta_rev_id":"55d80336ec6a3f75495e4bb3911b7e49ded4f6cd","_number":928894,"virtual_id_number":928894,"owner":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-09-15 05:58:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":2,"date":"2024-09-11 10:40:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"all":[{"value":1,"date":"2024-09-11 10:40:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Review-Priority":{"all":[{"value":0,"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Branch Freeze"," 0":"No Priority","+1":"Important Change","+2":"Top Priority / Holds Gate"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-09-11 10:37:52.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":33765,"name":"Mridula Joshi","email":"mrjoshi@redhat.com","username":"mrjoshi"},"state":"CC"},{"updated":"2024-09-11 10:38:19.000000000","updated_by":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"reviewer":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"state":"REVIEWER"},{"updated":"2024-09-11 12:29:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"CC"},{"updated":"2024-09-11 12:45:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"d5e6af26646ca8902f739ff6cc06b23fa37e6246","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"date":"2024-09-11 10:37:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/2023.2.","accounts_in_message":[],"_revision_number":1},{"id":"591747d1d07ccbd835c1f18751d9850488f203f6","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-11 10:40:48.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e16ce5c32bd8072729216cbed131094bb64f755b","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 12:29:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9002602d9e2c4d9988940e7498ca8f58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/251f513cf6db46d3aefaa7292d36aa07 : SUCCESS in 14m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"571aedf21334d1197b376f800ca1c3c012b5bb9d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 12:45:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dda42b50267542839c3fecb73dc9f683\n\n- grenade https://zuul.opendev.org/t/openstack/build/8b5cd9612cfe456ea51013e6203b1194 : SUCCESS in 49m 59s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c6247cd846f4445ba423960bb39a2708 : SUCCESS in 1h 02m 04s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ca66811da166457790b5c5cf8cf6b592 : SUCCESS in 1h 13m 13s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1144b4f509ac431eb2bd3aad5b526ef8 : SUCCESS in 58m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/17c5942a604245799985e296681dc7d7 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/76fb24b7202e40af89e6937652001656 : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d65544c888904a1d9516814fc12c35fe : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/857d4aa2ffce4663a6f60abfacdd42b6 : SUCCESS in 9m 46s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/91d6d2180741480a8c3269fd21202090 : SUCCESS in 8m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/527a9cfa8bdd43fbb9bd405753ddf581 : SUCCESS in 2h 00m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e3746e3e90064b579131391f04f366fb : SUCCESS in 1h 19m 15s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/92977c16a7bb44639959ee9b87f3f148 : SUCCESS in 1h 23m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/168b30c3631c45a48fbe3b4e907c2c5f : SUCCESS in 1h 22m 40s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/3802ba2d74d7428d9bd504ceccccfd59 : SUCCESS in 1h 42m 58s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/255d3ecc567241f891f96e121f781006 : SUCCESS in 55m 34s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/a61e2bd65694490788be27a4569fe9c1 : SUCCESS in 1h 41m 47s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d4f58921c1b94ccf8e6c81f1330e7fc2 : SUCCESS in 29m 45s","accounts_in_message":[],"_revision_number":1},{"id":"c0da57fd0c8b5eb332fccfd602bb11a95dee34d8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 12:46:00.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"303f2e7938752c00b498dc6723dab9fca6f07682","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-11 15:19:28.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3ce7d9e59fae4bccaa700ca986cc5b86\n\n- grenade https://zuul.opendev.org/t/openstack/build/d0dd7e94422d4dc2b42207d1c97a64ed : SUCCESS in 50m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/894bf1e15a094874a40507d9b53f9cb8 : SUCCESS in 34m 30s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/42da69e8cd124b1585dfcd076e999eb5 : SUCCESS in 1h 12m 43s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/b6e61fe6e31c430c897ea16944337624 : SUCCESS in 54m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89086541e82842fa98748f9726cc6261 : SUCCESS in 3m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/38e5d193d36e4f1c86daac2a1ed366dd : SUCCESS in 5m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6487330c39e349f582893920fba6bb36 : SUCCESS in 5m 19s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/68a43d50ed954b7c997e13f5ae1e3feb : SUCCESS in 7m 30s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c857ffc6c58b40a697fe6e48eaaac120 : SUCCESS in 7m 32s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/e086f8b3da054f6bb5eadc22bd41cdad : SUCCESS in 1h 26m 45s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/79854754e7704ec996e121820c250ef1 : FAILURE in 1h 19m 23s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/12249c228b944f369ba86f1c1d7ba655 : SUCCESS in 1h 41m 03s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/883c16b1eecf4daf8d6e4636c5d45e3d : SUCCESS in 57m 03s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/00babbc5cb4b4bffa2b6410fc852f232 : SUCCESS in 1h 51m 55s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b3483428f32d42e1bbe937ab262ce640 : SUCCESS in 37m 51s","accounts_in_message":[],"_revision_number":1},{"id":"273686bd244e6e274810bfb4bda008256bc14742","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-12 05:39:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b2175ac3e8ec7fce47dd06fe30c227252b115ec3","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 05:53:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/42f07871abe94dac933358c95f985f2c\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/3447e4d7af574b8f91a085f301cf78d7 : SUCCESS in 12m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0de6bce8ed96631d432d2927cd8453c134b1984e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 07:21:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/471215da381a4a97ad594d4f374c1fa1\n\n- grenade https://zuul.opendev.org/t/openstack/build/b354edca6d774cd0bf088438dd803b17 : SUCCESS in 54m 07s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/6fc57c36d4754bae96683a6faa16a0af : SUCCESS in 49m 58s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/e302e1501f0b4b249998a49517e347d8 : SUCCESS in 57m 49s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/dd6e98f5d1e74fc5b0c328461942bef1 : SUCCESS in 40m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fdfe17711772431baff63159a6676448 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f372c986eb354345aba5580e1a47b2c8 : SUCCESS in 4m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f28eb546647649639e738ddec3011a24 : SUCCESS in 3m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ac6a139b718f47f997bba95794a632b8 : SUCCESS in 8m 56s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/9e0e25c27c394b16a8475b3370473f4f : SUCCESS in 8m 33s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/8bdfe470d065427ba0281f10db9113b0 : SUCCESS in 1h 15m 57s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/28f02b69e7fc4c7186d4f425251b754f : SUCCESS in 1h 08m 12s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f409426b55504ee2ac56bb000c78fce3 : SUCCESS in 1h 30m 51s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/0fb7971428764818b2ed2bfe5c4e532d : SUCCESS in 1h 18m 05s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/d7c2c1632ad545e3b18ba5a1f5b638bb : SUCCESS in 1h 40m 16s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/eadcdd587b804b118d9e94057ea48f8e : SUCCESS in 40m 35s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/cd87c05263cd4d5c8f73ad0b0b37cb35 : SUCCESS in 1h 21m 46s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/b5a01740e0c24700aacc8832e3a10824 : SUCCESS in 29m 07s","accounts_in_message":[],"_revision_number":1},{"id":"c6b6273558bc0d0e81c7b370075fa5f9910f69e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 07:22:15.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b1a928c5c6d4e4cf49f672005385481a6b28639a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-12 10:09:09.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/4b5b48134e434e9fb29c747736dfe65c\n\n- grenade https://zuul.opendev.org/t/openstack/build/2cb00800132e401fa6e9fc6ac7458799 : SUCCESS in 57m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/14a19d583d514cc39573f835af3cdb5b : SUCCESS in 1h 00m 12s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/5647cdac6ad647faaf8f16f27f5834ac : SUCCESS in 1h 06m 11s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/6a3c1f91c816412c8a79f539f03453ac : SUCCESS in 52m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0d10701e13e24f678e677bf049af9a97 : SUCCESS in 3m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/91c83185baa041ec89010050bc68129f : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5dfdcbca90344084a89056f41b5a8918 : SUCCESS in 5m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/061ce805f0754beea5dd8e89ae98a605 : SUCCESS in 8m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6a9ffe9ad03b47058de2e08e9fc23549 : SUCCESS in 10m 01s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/88570daa728a4ab198d6355d0db098cf : SUCCESS in 1h 22m 05s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/9c6a7be0df124805ba23d5b66202f158 : SUCCESS in 1h 41m 01s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/dc60c0aac60d46c49bd5cec0aa1bedcb : SUCCESS in 1h 35m 41s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d6205774684041ffba8e57726d2182eb : SUCCESS in 48m 19s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/589163cf22794ff4b32b1af8410bb514 : FAILURE in 2h 04m 56s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/24bb097774584b558cfceab23959730d : SUCCESS in 34m 03s","accounts_in_message":[],"_revision_number":1},{"id":"33ab828dbd528817d4b4229d490847d57b0c43ff","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-13 06:09:16.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"cf2a280d4b12a78163c4fa123e69ea71315ffd73","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 06:19:34.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e2a4fe42c4b4c458015486e8f968ff8\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/7872db94057449dfbc95fa6545bb709e : SUCCESS in 8m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"753ffc7af7f2607f4201dbe44501c48511d60dd7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 08:06:58.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9148247dd9a646b5bd2db3769f4a42b9\n\n- grenade https://zuul.opendev.org/t/openstack/build/e8c2b83258694636bbcc6deda5e354b3 : SUCCESS in 41m 40s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0e8fe75078bb4366a0ad3faee2a70dc9 : SUCCESS in 40m 28s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/a1367f06d5e947ae8f933719179fbfee : SUCCESS in 1h 29m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/98e3708318f249ea957532cf20871542 : SUCCESS in 33m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8109d02ef74748ffba60d197298dd9a2 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93e61cf77d3945e0b2bd2f250c013697 : SUCCESS in 4m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4c85952621fd490fb57c578a598bc95e : SUCCESS in 4m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/11c6b85145bb4e999c68090407f44e36 : SUCCESS in 7m 58s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/fea7e7a68ecf4e6787a8cffa0369854c : SUCCESS in 7m 52s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/5eef774a21f44165a8e903b57aba251c : SUCCESS in 1h 27m 03s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/cc4b8df561e3419590028aa6e3364f9f : SUCCESS in 1h 19m 23s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/ee4541a919d74b919b9cf1fcfb5f89d4 : SUCCESS in 1h 35m 41s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c6fc45168b6649d293cbfa19390269dd : SUCCESS in 1h 47m 03s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/022a2ded3aa74b2891cfc8654abb0cd4 : SUCCESS in 1h 33m 10s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/4d3426a4fe8f41f2b79f7fce9e25944b : SUCCESS in 57m 22s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/2ad9ca4601be44aaae75f6f0d3288293 : SUCCESS in 1h 51m 59s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/1a26629ef54b41399ef9599d28f298a4 : SUCCESS in 38m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0e263b8843829260113167671baba3b77d8bb43d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 08:07:34.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b2c039b2e6d4b5daf9c06ef94875739ed3804f8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 09:59:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/aadc2a8040114223893f8c0f87e653aa\n\n- grenade https://zuul.opendev.org/t/openstack/build/268798ec44a64daba8f1a38c39158a33 : SUCCESS in 58m 10s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/bc6caa4256dc4c6998830d06182ee6ae : SUCCESS in 32m 19s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/48fcae54e4fa405db78c62d86dc0849e : SUCCESS in 1h 16m 28s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/8175e8e34d834214ac8eb308ff16088d : SUCCESS in 40m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9509df0e943341d7a8cd2e0ca2bbf379 : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4f151528404546338b08b784a05562bf : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ef6231ed95f409ba89ec4eeaa44f679 : SUCCESS in 5m 22s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/be7f4c20f64f473292d7633a449ba6dd : SUCCESS in 11m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4f68353519154bf191248c1dbe965f64 : SUCCESS in 9m 28s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/44afa7d0693148268ce1301467fa27f2 : FAILURE in 1h 13m 01s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/080e97a266924fef92589fdbea8b9e2a : SUCCESS in 1h 16m 24s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/1607a426b93c4f2da702a099ae9caa73 : SUCCESS in 1h 38m 21s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/617f299f63ed4594854e85e84e44285d : SUCCESS in 41m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e7b0ce42190844d696d27631f5f3764e : SUCCESS in 1h 47m 31s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/5866bbde838043fca6d8b5d57bdfa3ae : SUCCESS in 28m 56s","accounts_in_message":[],"_revision_number":1},{"id":"794d8f2758b8b30613a03a4f041fc8dff79da734","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-13 15:05:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e9997e23a357b0949963f6b9db0fbf17b5cfd14d","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 15:20:42.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc0a707ceec74c44acd49464933f3ab0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/87b5c67aaca542a1a26abdd29acc32c4 : SUCCESS in 13m 31s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c20635e0cab04744fa6813a26018e0098e69c65f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 17:09:41.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/9aab874716d14b0f91ddaee2903a290e\n\n- grenade https://zuul.opendev.org/t/openstack/build/060ddb339ebf48dd97df847f81ced290 : SUCCESS in 1h 07m 26s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/f0db62514c49423c8fc7acbfd8286c85 : SUCCESS in 1h 02m 55s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/d809504cf9e24ddfb30b14e20604e843 : FAILURE in 1h 18m 59s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/d2452f79422e43cfa196cc94f0cb411f : SUCCESS in 1h 01m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32ff20d670294820a2b6daffd7775497 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eecf44e7a15c49baa3ade362c46e5ea4 : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a48ba84aef834563952c83ea6d065339 : SUCCESS in 5m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1ea40d9436a04745992a56c808866b86 : SUCCESS in 7m 22s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/0073be22737a4ce4b4db2a441ef61370 : SUCCESS in 8m 30s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/ff9b5e5e45c6457d91fbda74508ede50 : TIMED_OUT in 2h 03m 19s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/de8b75da905b466b9210340c53a3b938 : SUCCESS in 1h 21m 50s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fdd9da970e8a46658228d1a269fd8ee9 : SUCCESS in 1h 12m 05s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/048575234cf24bbaa9d9c62c357db27e : SUCCESS in 1h 31m 26s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/287c44e54134468981676d380fb9a77d : SUCCESS in 1h 24m 29s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/7d5eea6cfca6470499d205eae41e1453 : SUCCESS in 53m 30s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/334d6f238af74aa9a83514040374e1c2 : SUCCESS in 1h 23m 34s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/029e117a28e74b67ac7103fe83e8694f : SUCCESS in 36m 10s","accounts_in_message":[],"_revision_number":1},{"id":"5a0d4f68b7751de3ea73587d78911b313c9076f4","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-13 18:35:11.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"63c3ba043e67d2664f9e1e60cf4a1f30f31ca34c","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 19:22:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9ec6c00b36df4fb7974cc55e3cd4bebc\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/9857e077ec7d4c9697987160b4d03ab0 : SUCCESS in 17m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"42236976cfa305b345a2db3abd8d49abe21e8c67","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-13 20:26:37.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a9b63e780e548dba498b568165e3d5c\n\n- grenade https://zuul.opendev.org/t/openstack/build/7bd9a51062e44888a75d5b712690a881 : SUCCESS in 41m 23s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/5ae6b4cef7db4133b0a5eeb16345038c : FAILURE in 24m 53s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/4957ce1d94cf44e2892347d981eb6f1b : SUCCESS in 1h 14m 29s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/7137cb31eb564ce6963c551193dd57e1 : FAILURE in 22m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3fdf8df735d04b3fabb6b20962166692 : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3dabfe22c525466a9ac7571ee7711c5b : SUCCESS in 6m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49ae6905075a40549c947e8f65507235 : SUCCESS in 5m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9e790f21acd34742a4b486e86ab74123 : SUCCESS in 8m 07s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/4a6687cc6e714b788bc2433f5c60bb55 : SUCCESS in 7m 07s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/3ad84289c6e44ce78ca820f53a617295 : SUCCESS in 1h 45m 12s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/98471a65d8254010a54c0ffa8381058d : SUCCESS in 55m 30s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/fb2c41d1a77c440c9080988f49ceb256 : SUCCESS in 1h 33m 07s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/2c37fedf1eca4ad2b08efcd23cce4eb4 : FAILURE in 25m 24s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/46bccb52a9a04e6cb1ac4fb25204bb80 : SUCCESS in 1h 26m 56s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/d22c90242f0d422da07621d0a4913e29 : SUCCESS in 35m 49s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/138b2c0c46f34f51a85fa90b53988431 : SUCCESS in 1h 06m 45s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/6309ce7a589140c9a82a42efd6d9de55 : SUCCESS in 35m 33s","accounts_in_message":[],"_revision_number":1},{"id":"08f6158cd6b5508e326a764ef77776de4c4b9f39","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-14 05:06:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"09429f1648a4257b1fe1ad59b1988797d166f490","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 05:19:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c8a2a5630da482da7666cd224072e58\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b28379218d3b4ccdb5e863fa4c857b72 : SUCCESS in 11m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e66c801ee2320068acef638acbc09a8f6b1f0249","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 07:16:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6059ab0da2f54c3eb3df6151d322b807\n\n- grenade https://zuul.opendev.org/t/openstack/build/ac6e88a5c3d440778fafe6172dbcb734 : SUCCESS in 39m 33s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/3ee8371731204698bb109416f252c144 : SUCCESS in 40m 36s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/b0186857e93f469f89a12704a3e82c63 : SUCCESS in 1h 00m 40s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/670ecb46889c474aa9635cc6c575b503 : SUCCESS in 57m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cb6fc7928a1a45c9b04c25c974268c71 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/18f0ca35c5864caab52301332e0a947e : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/15c2d37c45ab4c278ff9baf1dc942b4a : SUCCESS in 4m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/41914ae9e13a493bb95df34e521385e0 : SUCCESS in 8m 27s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a941aba0a83949b38a600b9bbf04ff5e : SUCCESS in 7m 22s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/a22662601e584474ae451016964db1f8 : TIMED_OUT in 2h 03m 07s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/60fff0ba39434946974d805960897552 : SUCCESS in 1h 01m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/442f1c2f35fa4f47a8762b12081735ef : SUCCESS in 1h 06m 59s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/7bb7dd69bceb4984ae7f97563f535193 : SUCCESS in 1h 52m 56s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/fa7ed20c1edd45c0b8602f7bd2ad109a : SUCCESS in 1h 41m 35s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/0686b93355604215b55eff77e7d47cfe : SUCCESS in 32m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/de8dc5ddb3d64dd097d68163efd7115b : SUCCESS in 1h 11m 27s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/79c41743bba04fada7f851eff8e379e9 : SUCCESS in 22m 54s","accounts_in_message":[],"_revision_number":1},{"id":"68cf7ccd962486d344f3649e6a1d39281497ea5e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 07:17:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7a350bd1205ad5e5f2be99cfa23f5590a43b8c00","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 09:28:45.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/baf6f62555f14452bd6c3fc7718289b2\n\n- grenade https://zuul.opendev.org/t/openstack/build/b256da285f0a460ab64597dd89d5896b : SUCCESS in 39m 24s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/c5ce529c515444bcbbce4b08f23e02ba : SUCCESS in 39m 47s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/97d0a3d4f89a4308b213c69d366765ec : SUCCESS in 57m 04s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/837d5a6e057f46ccb1db60d049992153 : SUCCESS in 42m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6af56d39864c7b9c641347e6f0d442 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63f28e7cf6124f1597ba43580d1d3e8f : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a78189624c854af1bc5757396b40b370 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/edfd33df627441f28e7244f579254309 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f156570747844c679dd4d46f30e968d3 : SUCCESS in 7m 51s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/312457263dea4cbd90d1512b9f1c6624 : FAILURE in 1h 11m 38s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f9a7e146567f43e594af68cd7408c764 : SUCCESS in 56m 19s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/c5e9149ed1de456d9857b4d5109dce97 : SUCCESS in 1h 32m 40s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/a747c2efa27943188516c3b87fb5b85b : SUCCESS in 32m 50s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/c4d7c794d9374dcdb64993847219ffe3 : SUCCESS in 2h 06m 33s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/41b43fe4c1af4ec7aa5b7d8aac830574 : SUCCESS in 36m 23s","accounts_in_message":[],"_revision_number":1},{"id":"8f18df2b83038bc8b80c0fac219d6bd19d73bba5","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-14 21:36:18.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7a2ec114df9f09194fd0fc38f752f7891fcd4b01","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 21:46:31.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7811b5f3b4fe47d6a7755b7c365a1337\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/b87fc1f16978467ca74894df6067cfe3 : SUCCESS in 8m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"66e7fbd4076bcd1de53dce7243ae36a33d338fe7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-14 23:02:06.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f43e4912f2144932aabb825aaf105aa6\n\n- grenade https://zuul.opendev.org/t/openstack/build/0fcca14db4fb474bb6b636ab92c42c9e : SUCCESS in 33m 43s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/0bd37be258c4486c8ee53cb363bdd554 : SUCCESS in 59m 45s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/22ef4187c95e47c1be74a4799c8667c1 : SUCCESS in 1h 06m 53s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/1ca7ae135789492e94524af6cb619eb0 : SUCCESS in 57m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8b649f64be2447a893d73d0922293c9e : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/635b426123384476ab6f2376662baaca : SUCCESS in 5m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/89268a426d96444289b762939a1eaf79 : SUCCESS in 6m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da37d4b84309446b836b26e3af5d0557 : SUCCESS in 8m 33s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/a74fc368a6be4866b95f2f42d96573c7 : SUCCESS in 8m 04s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/6399c6fa0b7342a98cbc1a437c52f094 : SUCCESS in 1h 19m 50s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/1826e5ddea7649ba99e84ab7881e6be5 : FAILURE in 47m 19s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/d28fea57ccfc4f6b93fc5fcd17dbe2bf : SUCCESS in 1h 02m 55s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/6a5e5de179804cf4a97e75e1b8a46f0a : SUCCESS in 1h 09m 38s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/84d47349e3aa47559d32502fe5bcf3dc : SUCCESS in 1h 18m 24s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/e7d129ac72434785a5702881fce2eb31 : SUCCESS in 31m 44s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/e3a4e03cdcf64f308a72e3749601fa6d : SUCCESS in 1h 24m 39s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/2aa766f256cf4c95a1ba77b1bf9c4d40 : SUCCESS in 22m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0aaa5211b5ccbdc74c7da654010a02c4d5b6f90a","author":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"},"date":"2024-09-15 01:55:51.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ed1e09d8fa9e408d1d704a426af2d37f017f8ad1","tag":"autogenerated:zuul:check-arm64","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 02:09:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (ARM64 pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a75ddff2d6e4db0b6b3120fb37943b0\n\n- openstack-tox-py38-arm64 https://zuul.opendev.org/t/openstack/build/2ed9f4b7aa9e4c31a31183048bc9006a : SUCCESS in 13m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f9d3079715bbda60f14ac093ac6cc64ef63b291c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 03:50:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4dc36403803d4e728b0b226bd8142f64\n\n- grenade https://zuul.opendev.org/t/openstack/build/8daba440576f476e86cd829dc475c600 : SUCCESS in 43m 50s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/396bd6a7128a4b7fa3af7130363ea3ae : SUCCESS in 45m 27s (non-voting)\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/176e43cefa8a4a36bcd11e37f9e03c87 : SUCCESS in 1h 03m 06s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/c3633637a56b4352a384ebc504b3c1a0 : SUCCESS in 47m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a7c43213eac487da4b631db79b9cee7 : SUCCESS in 3m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd1a6f15f70f4ca983dc85ad1acc5989 : SUCCESS in 5m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/35a6fb9d82e046bbab49cc704df95a05 : SUCCESS in 7m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7675bbf8219b42d2ab9f24c268a1f4ae : SUCCESS in 11m 50s\n- glance-tox-functional-py39-rbac-defaults https://zuul.opendev.org/t/openstack/build/2378658d27674da58f57edb696ba4fef : SUCCESS in 7m 25s (non-voting)\n- glance-ceph-thin-provisioning https://zuul.opendev.org/t/openstack/build/0638d639799f4df28e8db2d8bd9997b5 : SUCCESS in 1h 49m 27s (non-voting)\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/dafa72de3d6c41fcba432cd829db0798 : SUCCESS in 1h 21m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/f60f92bd64a04787a3e45808d4f56415 : SUCCESS in 1h 34m 39s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/482cd3f3d20541a68a85e24c44f31dd9 : SUCCESS in 1h 14m 46s\n- glance-multistore-cinder-import https://zuul.opendev.org/t/openstack/build/1d9be8827a1b48da8f1d5ebcb89c0cd7 : SUCCESS in 1h 34m 33s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/665b921e87694a1d835dd9f6b006c50b : SUCCESS in 53m 10s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/4df2468853f244deb76441bf4f7c7652 : SUCCESS in 1h 33m 22s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/d811618ae2a24b0a9c1a1de20db20611 : SUCCESS in 32m 19s","accounts_in_message":[],"_revision_number":1},{"id":"8950b208b2e5d8921113a291a455e6aa193e0fb0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 03:51:33.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b974d2d84b6f0a0d27a518d956bbd6a6acbb49e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 05:58:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0092a9620c1648e6a2e1411044c00ef6\n\n- grenade https://zuul.opendev.org/t/openstack/build/e3139730b305419abb4ed9fa59380369 : SUCCESS in 46m 16s\n- grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ce1619119aa46069db6e5fdde1e22b7 : SUCCESS in 41m 22s\n- tempest-integrated-storage https://zuul.opendev.org/t/openstack/build/ce12e3649697472e8ef2a628a2a6efbb : SUCCESS in 1h 01m 38s\n- openstacksdk-functional-devstack https://zuul.opendev.org/t/openstack/build/5ce47b92aa94488e88bbe8a7126b922f : SUCCESS in 40m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/45bed505e8cb4cef940dfd527447210c : SUCCESS in 4m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/90a010ae3925458f9e2d04aa4db751ae : SUCCESS in 4m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c04c30498934399ba7c2863e83934d5 : SUCCESS in 5m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/198c6a781c844f399743fd6b7939a73e : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4da5755d18a4341bcd5915a691882c5 : SUCCESS in 7m 33s\n- tempest-integrated-storage-enforce-scope-new-defaults https://zuul.opendev.org/t/openstack/build/b354cc2bda274e3fa3601cd254a2769e : SUCCESS in 1h 01m 14s\n- tempest-integrated-storage-import https://zuul.opendev.org/t/openstack/build/6a985d8436c548ddac1d1b9bcf74c8cf : SUCCESS in 55m 21s\n- tempest-integrated-storage-import-standalone https://zuul.opendev.org/t/openstack/build/14ec6ec15c834b65a27ced3e196ba813 : SUCCESS in 1h 34m 47s\n- tempest-ipv6-only https://zuul.opendev.org/t/openstack/build/f7a1777664ed493397d448f731531fd9 : SUCCESS in 31m 48s\n- nova-ceph-multistore https://zuul.opendev.org/t/openstack/build/850acaa4cfbd4a9d81121de711b51d16 : SUCCESS in 2h 02m 07s\n- glance-secure-rbac-protection-functional https://zuul.opendev.org/t/openstack/build/f33858ec1a1f4819a2c886a641386a9f : SUCCESS in 20m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8e829bb7f5642fcf54c98442a7abc2a32fa54e43","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 05:58:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"55d80336ec6a3f75495e4bb3911b7e49ded4f6cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-09-15 05:59:35.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1da06df30a1c4340b7f947319c217c20\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f6eef8c7d83f4aa28ac6795049c252e6 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c2ac316e7cc4df9e40888870adf6decd46f2b71a","revisions":{"c2ac316e7cc4df9e40888870adf6decd46f2b71a":{"kind":"REWORK","_number":1,"created":"2024-09-11 10:37:52.000000000","uploader":{"_account_id":19138,"name":"Pranali Deore","email":"pdeore@redhat.com","username":"PranaliD"},"ref":"refs/changes/94/928894/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/glance","ref":"refs/changes/94/928894/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/928894/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/928894/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/glance refs/changes/94/928894/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/glance refs/changes/94/928894/1"}}},"commit":{"parents":[{"commit":"1427857a7d2a823bb754cf2f12f2b10e7b3df31d","subject":"Add iso file format inspector","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/1427857a7d2a823bb754cf2f12f2b10e7b3df31d"}]}],"author":{"name":"Mridula Joshi","email":"mrjoshi@redhat.com","date":"2024-09-09 08:36:27.000000000","tz":0},"committer":{"name":"Pranali Deore","email":"pdeore@redhat.com","date":"2024-09-11 10:37:52.000000000","tz":0},"subject":"Fixed typo in discovery.py","message":"Fixed typo in discovery.py\n\nReplaced \"additonalProperties\" with \"additionalProperties\"\n\nChange-Id: I550f0d49bf1a55b6b4b02a0c3a449725c25f63e3\n(cherry picked from commit 2ae69b51fa20810b86d32bc4d4f7b6a0db649364)\n(cherry picked from commit d0be1bededcc482946970312d92832f085d35e17)\n(cherry picked from commit 18453f86021947168d799b3d48e5b3131c987137)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c2ac316e7cc4df9e40888870adf6decd46f2b71a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/glance/commit/c2ac316e7cc4df9e40888870adf6decd46f2b71a"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":9303,"name":"Abhishek Kekane","email":"akekane@redhat.com","username":"abhishekkekane"}},{"label":"Review-Priority","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Review-Priority","description":"Review priority","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"-label:Review-Priority\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":[],"failing_atoms":["label:Review-Priority\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
